)]}'
{"id":"openstack%2Fironic~930288","triplet_id":"openstack%2Fironic~master~If8d833bc5b012a17588aa7e3ebcb143bba12de12","project":"openstack/ironic","branch":"master","topic":"fix-doc-inspect-interface-redfish","attention_set":{},"removed_from_attention_set":{"35929":{"account":{"_account_id":35929,"name":"Mahnoor Asghar","display_name":"Mahnoor Asghar","email":"masghar@redhat.com","username":"mahnoorasghar"},"last_update":"2024-09-24 12:49:08.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"If8d833bc5b012a17588aa7e3ebcb143bba12de12","subject":"Fix inspect interface for redfish driver in the docs","status":"MERGED","created":"2024-09-24 10:57:09.000000000","updated":"2024-09-24 12:50:52.000000000","submitted":"2024-09-24 12:49:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"930288-fix-doc-inspect-interface-redfish","meta_rev_id":"f4d277abc304d4e8d515850e5a84dec6c944be48","_number":930288,"virtual_id_number":930288,"owner":{"_account_id":35929,"name":"Mahnoor Asghar","display_name":"Mahnoor Asghar","email":"masghar@redhat.com","username":"mahnoorasghar"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-24 12:49:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2024-09-24 12:34:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-09-24 11:22:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":1,"date":"2024-09-24 12:34:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-24 11:21:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-24 11:22:30.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2024-09-24 12:34:14.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"}],"messages":[{"id":"1470bb9f660396537b17ef91b37e3d8472522471","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35929,"name":"Mahnoor Asghar","display_name":"Mahnoor Asghar","email":"masghar@redhat.com","username":"mahnoorasghar"},"date":"2024-09-24 10:57:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"951cbfade69b0c02e99b453b9b621c69443b11a4","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":35929,"name":"Mahnoor Asghar","display_name":"Mahnoor Asghar","email":"masghar@redhat.com","username":"mahnoorasghar"},"date":"2024-09-24 10:59:02.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":1},{"id":"82d307c5f3b1be12c8695b9c49fba6611026b528","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35929,"name":"Mahnoor Asghar","display_name":"Mahnoor Asghar","email":"masghar@redhat.com","username":"mahnoorasghar"},"date":"2024-09-24 11:08:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c437ab7c36ad4d16b192117e6c3d90a97c889c73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 11:21:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45683c109e334a8e9bc4321807ed3ac4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf3b57f0b1934313ae49d0f8ee4dc39b : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8787799f4ff6408c93602703aec23510 : SUCCESS in 10m 53s\n- ironic-tox-codespell https://zuul.opendev.org/t/openstack/build/1ea4f746e2aa4c16ad8fb00f1cdd8683 : SUCCESS in 3m 31s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/fb2d6f5d28fe4da6925aab375bb03343 : SUCCESS in 3m 26s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/173a86083092469aa19a92a2787f46a0 : SUCCESS in 7m 09s","accounts_in_message":[],"_revision_number":2},{"id":"0408076b7ddab967a279112591821c667da44900","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2024-09-24 11:22:30.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ee2f0f966263e9b33752dca8e1c4f0c2ea9b9ce1","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2024-09-24 12:34:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7a588cbf0837016964f2929c67e41360dd044fed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 12:34:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c2d6084c7e9fead69769e7904775a2f84da9d1e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 12:49:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08152c2f2438485394c012919fc962bb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b91f25dd22f41e08cc8de3fbd1d35a9 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29c179111399435db9ef23cf72abfcd7 : SUCCESS in 13m 10s\n- ironic-tox-codespell https://zuul.opendev.org/t/openstack/build/c1fd86815db546ee9b2c54b08658173e : SUCCESS in 4m 26s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/27d5e88e76834739802083c6e44b98b8 : SUCCESS in 5m 52s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/a011c3923da34f3788e2a61f2a6d4c20 : SUCCESS in 8m 56s","accounts_in_message":[],"_revision_number":2},{"id":"4995c4e4f24dbcbd03b7048b8ca4f8db1e81af37","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 12:49:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f4d277abc304d4e8d515850e5a84dec6c944be48","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 12:50:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb90f8e908f74354b85c430e48d254c3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f93475b2f364e8d94db1201e47fdbac : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"adfb091695a734b5e18ba52662a14e0ddd497827","revisions":{"b68b62ff17dcf6fe6032c05906a8c6c16a7352ff":{"kind":"REWORK","_number":1,"created":"2024-09-24 10:57:09.000000000","uploader":{"_account_id":35929,"name":"Mahnoor Asghar","display_name":"Mahnoor Asghar","email":"masghar@redhat.com","username":"mahnoorasghar"},"ref":"refs/changes/88/930288/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/88/930288/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/88/930288/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/88/930288/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/88/930288/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/88/930288/1"}}},"commit":{"parents":[{"commit":"5857ca0cd5bc4949f77b9f3b6c3e0d7ab97f6b88","subject":"Update release mappings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/5857ca0cd5bc4949f77b9f3b6c3e0d7ab97f6b88"}]}],"author":{"name":"Mahnoor Asghar","email":"masghar@redhat.com","date":"2024-09-24 10:44:54.000000000","tz":-240},"committer":{"name":"Mahnoor Asghar","email":"masghar@redhat.com","date":"2024-09-24 10:56:48.000000000","tz":-240},"subject":"Fix inspect interface for redfish driver in the docs","message":"Fix inspect interface for redfish driver in the docs\n\nChange-Id: If8d833bc5b012a17588aa7e3ebcb143bba12de12\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/b68b62ff17dcf6fe6032c05906a8c6c16a7352ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/b68b62ff17dcf6fe6032c05906a8c6c16a7352ff"}]},"branch":"refs/heads/master"},"adfb091695a734b5e18ba52662a14e0ddd497827":{"kind":"REWORK","_number":2,"created":"2024-09-24 11:08:31.000000000","uploader":{"_account_id":35929,"name":"Mahnoor Asghar","display_name":"Mahnoor Asghar","email":"masghar@redhat.com","username":"mahnoorasghar"},"ref":"refs/changes/88/930288/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/88/930288/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/88/930288/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/88/930288/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/88/930288/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/88/930288/2"}}},"commit":{"parents":[{"commit":"5857ca0cd5bc4949f77b9f3b6c3e0d7ab97f6b88","subject":"Update release mappings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/5857ca0cd5bc4949f77b9f3b6c3e0d7ab97f6b88"}]}],"author":{"name":"Mahnoor Asghar","email":"masghar@redhat.com","date":"2024-09-24 10:44:54.000000000","tz":-240},"committer":{"name":"Mahnoor Asghar","email":"masghar@redhat.com","date":"2024-09-24 11:08:18.000000000","tz":-240},"subject":"Fix inspect interface for redfish driver in the docs","message":"Fix inspect interface for redfish driver in the docs\n\nChange-Id: If8d833bc5b012a17588aa7e3ebcb143bba12de12\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/adfb091695a734b5e18ba52662a14e0ddd497827"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/adfb091695a734b5e18ba52662a14e0ddd497827"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
