)]}'
{"id":"openstack%2Fironic~930477","triplet_id":"openstack%2Fironic~bugfix%2F26.0~Ie1c64b4807ecf29fa0da54501798d363675977c8","project":"openstack/ironic","branch":"bugfix/26.0","topic":"inspect-abort-bugfix/26.0","attention_set":{},"removed_from_attention_set":{"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2024-09-25 15:18:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie1c64b4807ecf29fa0da54501798d363675977c8","subject":"Fix double transition to INSPECTFAIL on aborting in-band inspection","status":"MERGED","created":"2024-09-25 14:29:32.000000000","updated":"2024-09-25 15:19:41.000000000","submitted":"2024-09-25 15:18:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"930477-inspect-abort-bugfix/26.0","cherry_pick_of_change":930279,"cherry_pick_of_patch_set":1,"meta_rev_id":"4980ed4a257d1ed23f2a0c82c227313c9bfcf4f4","_number":930477,"virtual_id_number":930477,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-25 15:18:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2024-09-25 14:58:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":1,"date":"2024-09-25 14:58:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-25 14:57:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-25 14:58:17.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"}],"messages":[{"id":"6a4a84d564458ef09493f60ded7d8b68cf557ba0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2024-09-25 14:29:32.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"21989797e1f8e8884c4ffc6652525ff5ccb0a3ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 14:57:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c2fd74053a0467e940f833cf3086ffb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b8d73f418634b8d9c829cd717d0e313 : SUCCESS in 10m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79fe50c9eef44751927bcca18816313c : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9defa58332284bd398cece365dc16fcd : SUCCESS in 13m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99750b65375c49a9bb4630f6f236e1bb : SUCCESS in 20m 48s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/3b78c09ba8f14af9a4bb255d91e18d5e : SUCCESS in 5m 25s\n- ironic-tox-codespell https://zuul.opendev.org/t/openstack/build/465a7b6f243c4f5f91bdc25cba61f00c : SUCCESS in 4m 07s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/31422a3b8a454cb39f9c79d95d07b78a : SUCCESS in 5m 24s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/7ca52d76c8d14842b7b7977dccd307f5 : SUCCESS in 7m 51s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/77204570c490446fa75ec3b838fc44eb : SUCCESS in 8m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b138aacf771631728c4581e1fc349369f24d3b21","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2024-09-25 14:58:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c67c46cf9eb398628717354a73b65e99b9afe814","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 14:58:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6673e2020abc22dbe6410c3a5ffc79b3920db99c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 15:18:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9da0931d639443b7a88662ccae1c1859\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0584db3b944a4ac7b2827d3f2a5e9d71 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f39d01032a849e7a7aa801d53468019 : SUCCESS in 13m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4902a9c75314c21978b07cf39829da4 : SUCCESS in 16m 26s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/925e6f3851df46279e887933914fc504 : SUCCESS in 4m 54s\n- ironic-tox-codespell https://zuul.opendev.org/t/openstack/build/0666477767c2402882c4607361a43475 : SUCCESS in 5m 22s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/7b790d9d1f214397a870cbd8ecd2b443 : SUCCESS in 5m 41s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/ac6e0bd420bc45b4b21245c8dfb309f8 : SUCCESS in 8m 32s","accounts_in_message":[],"_revision_number":1},{"id":"9ed2faadca7fda6e0ed928f4eb99b3e0476bb83c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 15:18:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4980ed4a257d1ed23f2a0c82c227313c9bfcf4f4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 15:19:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7646c9d0518425c9e4cb779e059c88b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5858d55b04dc45beae93b186243589e4 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fda383098135414ca1d80ae98dadda00 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9a800922bcfcbfb23a6fcf116669a455dc9c3849","revisions":{"9a800922bcfcbfb23a6fcf116669a455dc9c3849":{"kind":"REWORK","_number":1,"created":"2024-09-25 14:29:32.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/77/930477/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/77/930477/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/77/930477/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/77/930477/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/77/930477/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/77/930477/1"}}},"commit":{"parents":[{"commit":"86487560f0957b87dc649d35fbf37fd916aa868d","subject":"Fix inspection if bmc_address or bmc_v6address is None","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/86487560f0957b87dc649d35fbf37fd916aa868d"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2024-09-24 08:27:18.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2024-09-25 14:29:32.000000000","tz":0},"subject":"Fix double transition to INSPECTFAIL on aborting in-band inspection","message":"Fix double transition to INSPECTFAIL on aborting in-band inspection\n\nBoth the driver and the conductor code try to transition the node to\nINSPECTFAIL, with the 2nd attempt failing. Rework the driver code to\nonly do implementation-specific clean-up. Also safeguard the conductor\ncode against this case.\n\nChange-Id: Ie1c64b4807ecf29fa0da54501798d363675977c8\n(cherry picked from commit 8a6b5eb8c3dc385b49bb8a7811029a2a60b57e71)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/9a800922bcfcbfb23a6fcf116669a455dc9c3849"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/9a800922bcfcbfb23a6fcf116669a455dc9c3849"}]},"branch":"refs/heads/bugfix/26.0"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
