)]}'
{"id":"openstack%2Fironic~930478","triplet_id":"openstack%2Fironic~bugfix%2F25.0~Ie1c64b4807ecf29fa0da54501798d363675977c8","project":"openstack/ironic","branch":"bugfix/25.0","topic":"inspect-abort-bugfix/25.0","attention_set":{},"removed_from_attention_set":{"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2024-09-25 15:21:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie1c64b4807ecf29fa0da54501798d363675977c8","subject":"Fix double transition to INSPECTFAIL on aborting in-band inspection","status":"MERGED","created":"2024-09-25 14:29:59.000000000","updated":"2024-09-25 15:22:54.000000000","submitted":"2024-09-25 15:21:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"930478-inspect-abort-bugfix/25.0","cherry_pick_of_change":930279,"cherry_pick_of_patch_set":1,"meta_rev_id":"9202803fa4b9f13a1717799cd0ec73ce6df6b2ca","_number":930478,"virtual_id_number":930478,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-25 15:21:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2024-09-25 14:58:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":1,"date":"2024-09-25 14:58:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-25 14:58:21.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2024-09-25 14:58:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8deef1d0387ea1002b8858e95fcb5d14a658ba9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2024-09-25 14:29:59.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"01582aa1dbcedc9a4beba88fe78f9d3210cea268","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2024-09-25 14:58:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"85953df7b4333be23718b380b48971d0a1e90508","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 14:58:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06c23e7b8ed84272891a86c36272196e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3047b986cb794419b15f3b7bc7b28b7f : SUCCESS in 10m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2fa608a2b834e99a01716c95e6656fb : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c052d6bbc807458e9282eb554e288b7a : SUCCESS in 13m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5cb043f698842579fbd53e23c8119f7 : SUCCESS in 22m 23s\n- ironic-tox-codespell https://zuul.opendev.org/t/openstack/build/0ae2606861084955879bcd7000874b53 : SUCCESS in 4m 28s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/60a2dc41640a4a6186e99d31e217ec9f : SUCCESS in 6m 52s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/3f9930c022dd4871959581c2af134d0a : SUCCESS in 8m 55s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/5193329315ac40c6b6ec309f1c719c10 : SUCCESS in 10m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8e9434e5f827177c59431196f8bb877dbcb9458a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 14:59:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"12c4bea9ed9b5373613af822afc3c14dcd5b386e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 15:21:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47e2a29a0490491992b632e3b13a3ab4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45b5f0c35f13406baeb6dc87b23ef0ab : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/baf042e586d14ae498aa2047e999e5fc : SUCCESS in 13m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/276e7bc78762401789d5e564ebfb5779 : SUCCESS in 20m 56s\n- ironic-tox-codespell https://zuul.opendev.org/t/openstack/build/d8a9059b3bc341cea5e95ce020423fa1 : SUCCESS in 3m 58s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/42ee3f7e2e9f4eb68e0df09f14c36d54 : SUCCESS in 4m 51s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/9e92f223a08b4783a71668fb4008a88f : SUCCESS in 10m 22s","accounts_in_message":[],"_revision_number":1},{"id":"c8997204af9ae404eb6a7507a5add398183a2514","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 15:21:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9202803fa4b9f13a1717799cd0ec73ce6df6b2ca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 15:22:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/321456a2c1644799a61821837f8318e4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ff573a6f7d546898cdd05df3c0eda7c : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50536bfe2530442abde8c9cb91bcb74d : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f1ad3e640d658d3743d609565b667e43aceb2210","revisions":{"f1ad3e640d658d3743d609565b667e43aceb2210":{"kind":"REWORK","_number":1,"created":"2024-09-25 14:29:59.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/78/930478/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/78/930478/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/78/930478/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/78/930478/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/78/930478/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/78/930478/1"}}},"commit":{"parents":[{"commit":"a1f63185c751326b021dfeb090fbdcd1b74604dd","subject":"Fix inspection if bmc_address or bmc_v6address is None","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/a1f63185c751326b021dfeb090fbdcd1b74604dd"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2024-09-24 08:27:18.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2024-09-25 14:29:59.000000000","tz":0},"subject":"Fix double transition to INSPECTFAIL on aborting in-band inspection","message":"Fix double transition to INSPECTFAIL on aborting in-band inspection\n\nBoth the driver and the conductor code try to transition the node to\nINSPECTFAIL, with the 2nd attempt failing. Rework the driver code to\nonly do implementation-specific clean-up. Also safeguard the conductor\ncode against this case.\n\nChange-Id: Ie1c64b4807ecf29fa0da54501798d363675977c8\n(cherry picked from commit 8a6b5eb8c3dc385b49bb8a7811029a2a60b57e71)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/f1ad3e640d658d3743d609565b667e43aceb2210"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/f1ad3e640d658d3743d609565b667e43aceb2210"}]},"branch":"refs/heads/bugfix/25.0"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
