)]}'
{"id":"openstack%2Fironic~932016","triplet_id":"openstack%2Fironic~master~I48ee1dd4adc2e5bcc18c5f116d979e7524248495","project":"openstack/ironic","branch":"master","topic":"grenade-venv","attention_set":{},"removed_from_attention_set":{"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2024-10-22 17:01:38.000000000","reason":"\u003cGERRIT_ACCOUNT_10342\u003e replied on the change","reason_account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-10-22 18:48:58.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"I48ee1dd4adc2e5bcc18c5f116d979e7524248495","subject":"Enable GLOBAL_VENV in ironic grenade jobs","status":"MERGED","created":"2024-10-10 02:55:02.000000000","updated":"2024-10-22 18:51:49.000000000","submitted":"2024-10-22 18:48:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"932016-grenade-venv","meta_rev_id":"976e3041cb2b85058027d29ee495ece373d58c66","_number":932016,"virtual_id_number":932016,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-22 18:48:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"recommended":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":1,"date":"2024-10-18 16:35:28.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":2,"date":"2024-10-22 17:01:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":2,"date":"2024-10-18 06:51:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":1,"date":"2024-10-22 17:01:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"date":"2024-10-18 03:56:01.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-10 05:29:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-10 05:31:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-15 18:39:26.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2024-10-18 06:51:46.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2024-10-18 16:35:28.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"}],"messages":[{"id":"76b37c1681681206d9b816fbd357f2dbfb6a855b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-10-10 02:55:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0b320eb8406bc56637789cb424796eea5eb05ecb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 05:29:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9cdd06c79374c85a9730882982dedde\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/83fa86cf1c964331aaa54584f6cdd829 : FAILURE in 26m 06s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1d036a455cd84b88ac0f5a2f88d50169 : FAILURE in 18m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"029684f13b83f2d0fb1a9865b54e0939f1ec442c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 05:31:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/33bd34ee92c2474a815a2d12707bc844\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48106d9d81af4c818236e6826c70fc5a : SUCCESS in 11m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e864fbae30144a5991153053d2c1761 : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6de86bc5dc4a4e42a64cbe2dd2f7e61d : SUCCESS in 8m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/60902a895f704c6a9ced880ddc721465 : SUCCESS in 10m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d273f2f6d8524dcea3975b97ddedd146 : SUCCESS in 15m 05s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/4bd39127b5804fa59c183f81ab7a77b7 : RETRY_LIMIT in 1m 22s\n- ironic-tox-codespell https://zuul.opendev.org/t/openstack/build/3f38ba1903db4c83a9183cc73a46afa0 : SUCCESS in 4m 32s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/0febba561850442da7aebb06a72ab714 : SUCCESS in 4m 52s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/459533779f0d40acaa2cb4d5a62b735d : SUCCESS in 9m 15s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/bf501e6f8ebd4e63abb41b778d7f8086 : SUCCESS in 39m 17s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/700b1c38891148d0969d4f683c256379 : SUCCESS in 1h 08m 41s\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/5d00e89ccadf45fd9c41a32c99367199 : SUCCESS in 57m 08s (non-voting)\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/aeb2d0a5b06349f998605b79ea559d26 : SUCCESS in 1h 04m 32s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/17ca11a5a80b443bba7ac58c8292b193 : SUCCESS in 40m 54s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/b3266ca09145466586b687f491c56c75 : SUCCESS in 41m 32s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/828b82ec26f2440c9e69b1df7f929847 : SUCCESS in 37m 22s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/18c2e0de026b4e06acaecf09da358860 : SUCCESS in 41m 57s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/64d350b380124e6f967053be94dbb36b : SUCCESS in 45m 37s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/5e17259476874795a2d06e9923427275 : POST_FAILURE in 2h 21m 08s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-shard https://zuul.opendev.org/t/openstack/build/74c225c655de40a8a752bf2a5da41048 : SUCCESS in 45m 21s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/8d89a15db6bf448787928271784c273e : SUCCESS in 58m 06s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/79f2c3a2520c4d3096b99fed60cbe786 : SUCCESS in 41m 34s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/d024792e41594a91b39ffe5e0fee77eb : RETRY_LIMIT in 1m 36s\n- metalsmith-integration-glance-centos9-legacy https://zuul.opendev.org/t/openstack/build/99318f92a0764e6b89341f64a8cbd0b8 : SUCCESS in 1h 34m 17s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/81d6c9015579455da8470716d7960291 : SUCCESS in 40m 37s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/95d9412872af4338a949424a634e5fca : SUCCESS in 49m 25s (non-voting)\n- ironic-inspector-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/30e8280e2adf4528bbca8d8f42adebe8 : SUCCESS in 48m 22s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/ee6a0c29c139421cb0761074d016d997 : SUCCESS in 30m 17s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/6ebf5ba96934417ba28db128532206f4 : SUCCESS in 44m 03s (non-voting)\n- ironic-tempest-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/9159d7bfd2134878ace459f6085bf13f : SUCCESS in 1h 04m 34s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/f71f3a8a201e448c98d73ff072ff9707 : SUCCESS in 55m 55s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8ec5cc11d8844f0b9aa810cc7cb290b6 : SUCCESS in 37m 23s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/8cfa60e43ead49839dd7ed0294cd0f2b : SUCCESS in 45m 20s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/26fefa2c0c4b4025aac1643e6a3071eb : FAILURE in 24m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6e97481db82d5ef456d7491951d2da608c781c48","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-10-15 18:39:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3baf74055727fd595eb639a70e27146bd67ee620","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-10-15 18:39:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"92d500686da3fbf34483a522c5403d520942006d","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2024-10-15 18:41:10.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":1},{"id":"47e444fbce646b7f24eca4f3984c921d02bc7bfd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 18:59:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7255e9a2791434ebe862a6749e2d7da\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1e7d1d32612c4b1b8cafded3f1f14e9f : SUCCESS in 17m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/968d951853a247d5b2c74573fffae2e0 : SUCCESS in 12m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"43d67c474d45822c283c6d127f73d5c76cfcc67e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 20:19:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1816564ba64845b99ffffc93d58c3a4b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5986a8d8347b4d58bbfdb098cb613eec : SUCCESS in 11m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1e5b34ee7da444daa6acdcef6190909 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7216c618b3a54eca995c6d185457ff35 : SUCCESS in 9m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/691dc2900fd0474fa6950fd8027a3793 : SUCCESS in 8m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86e73feb975e40ecbe1c92b02345a164 : SUCCESS in 13m 13s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/9ad52c65a31140b7abde77aa52cd2568 : SUCCESS in 5m 52s\n- ironic-tox-codespell https://zuul.opendev.org/t/openstack/build/6bcd36d147514574bcd26163ffa22eb5 : SUCCESS in 4m 32s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/5be61c925f0945b78c436023503ee977 : SUCCESS in 5m 00s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/2a903c5b37a842a793263d528d722a52 : SUCCESS in 10m 25s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/61492f713d304fcd89892decb6ee6fde : SUCCESS in 41m 59s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/7a4da021adf14002b8d60d85ae7f9afb : SUCCESS in 1h 11m 01s\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/bc620d40b0d34cb8bfeabce068394d37 : SUCCESS in 49m 25s (non-voting)\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/6d7e2181e4164751bd5cdaeeed29a70e : SUCCESS in 44m 42s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/b1ae26ab183a4c5b93485938641b6f2a : SUCCESS in 43m 10s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/f7e51a84405246d98eaee963494925d2 : SUCCESS in 33m 01s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/6161ab4ed40245a2b6bf79c0ec29724b : SUCCESS in 21m 40s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/7cc29da772cf496fac95ead12c03084a : SUCCESS in 40m 48s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/19feff91ea0349c2a9b4d82e69cb604b : SUCCESS in 35m 36s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/672c718ef3874bd0ab4a45f04218bedb : SUCCESS in 1h 08m 11s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-shard https://zuul.opendev.org/t/openstack/build/44ecc525a8534604b959dd7716378a31 : SUCCESS in 1h 03m 54s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/ae7542acda374e7aa49c458ca846974c : SUCCESS in 37m 44s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/7c8c2825dc164cbebe41d2e4eaa01cc1 : SUCCESS in 37m 03s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/0e98a7dce3de4ff1b26bf4aab580861a : SUCCESS in 45m 24s\n- metalsmith-integration-glance-centos9-legacy https://zuul.opendev.org/t/openstack/build/f4e0dbe62c074e789f45c5c458700055 : SUCCESS in 1h 39m 19s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/dedcd20c67b747629b39cf9c3bb7b2bb : SUCCESS in 41m 11s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/3d2fc40a6fc14c95b9dc9f82b26ad0cc : SUCCESS in 52m 51s (non-voting)\n- ironic-inspector-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/f1804c90cc0c418cbc16fc9adbb34b8a : SUCCESS in 53m 59s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/7a799fa7aff542e8a1676d851486e804 : SUCCESS in 43m 23s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/91af647afcb746fa82291d0f81bbbfce : SUCCESS in 41m 06s (non-voting)\n- ironic-tempest-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/cadad6498de74c77861424c2f2057913 : FAILURE in 1h 03m 43s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/f86480f4910f42a49411f647794f51c5 : SUCCESS in 34m 28s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e3bfccdec4a5448193172584c5f85096 : SUCCESS in 37m 38s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/c20b10734a104f48911b6f209a9edd60 : SUCCESS in 37m 17s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/1ab0480d3d4f487cbd120ee7ef5472ad : FAILURE in 24m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9a70c9cb3f867364cf30709d8c5fe7f18b974c1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-10-16 00:39:47.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"ae9b6191da387d19f4123dac3d8cd336b26aaded","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-16 01:05:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/972cd192949046349e37aabf62bf35cb\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dc91a3a28f4e44f281d00d27cfda8f3e : SUCCESS in 23m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/dbafeba5c1474f019979ef5a4f4c761b : SUCCESS in 14m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b8d1dcc4443ba38632e9fa8cb616831e67be1bbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-16 01:25:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/18ffca6129ce4097a08fe145d2b37245\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a03169f9a9e94b1ebfee72aa9252d545 : SUCCESS in 11m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a14524431da4e64b28b1bda680a58d8 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a6a2a63132ac4f76a1b4e69c9190a43a : SUCCESS in 8m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/04cfa9d4c02646d6b9750235755f4427 : SUCCESS in 8m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40629b7f5eeb48b19ddca05b1593b763 : SUCCESS in 10m 03s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/5a0dfb0c99554a2cb3d5891f5e5cdfe9 : SUCCESS in 4m 38s\n- ironic-tox-codespell https://zuul.opendev.org/t/openstack/build/df0a656287a54acf971c0366b4948f40 : SUCCESS in 3m 10s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/b1ffb2b1798c4be1b9f370ff184db1a5 : SUCCESS in 4m 38s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/e19d2d5e0fb8441a9a381fba835334d7 : SUCCESS in 7m 35s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/572386efae4047af8eef1800441054fd : POST_FAILURE in 9m 55s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/22b638a1d61e44f6a70930678871a703 : SUCCESS in 24m 47s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/3284d7b2c09a4df180f3cc01032cc064 : SUCCESS in 37m 38s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/3764be69c7c64d6c9ea7925e18a53e1e : SUCCESS in 29m 21s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/3dc3c480d7b74b2885d051ebba79b45a : POST_FAILURE in 17m 54s","accounts_in_message":[],"_revision_number":2},{"id":"3baf94971ce8895404c5df2452f76279592aa1ce","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-10-16 19:32:23.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"13cb8ed5bdfb4b1409e88aa2ef7ac05e37dfee3d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-10-16 19:43:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e5d5ec0e72675bf91e5c2091101dabaf1f9f73f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-10-17 17:07:33.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"e183545de90efd4eef8e89ec2eeaafebb12f391e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-10-17 17:07:50.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"d887f0e9bbfa40af768bbf7c92aea1ffe095bf8c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-17 17:31:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1678f0240ed42bd86c0fa95a7183143\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c37d0b10ec6a43d484cba06dbd22772c : SUCCESS in 22m 19s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7df0065ac38d4e6e9ac73d39594b00e2 : FAILURE in 17m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c54c44a9bdeceba08430edaa849d0ed03219619b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-17 18:57:34.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3d31ddcb10554c979e87ed6a7ce3b8ad\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2164ac0a08ac4baaa3d7a9638ea95443 : SUCCESS in 9m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9c80a5423004ad1b73d9c2dde1e09b0 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bac8051e9ec54953b1c66b4234acfdcd : SUCCESS in 7m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f82cecf591614bc79c4d0810d3e43ddf : SUCCESS in 10m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f21edf3c0a7a4ce38fb2805da0f61a69 : SUCCESS in 13m 50s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/5f78bc1b649e44d6b0477575522defac : SUCCESS in 5m 38s\n- ironic-tox-codespell https://zuul.opendev.org/t/openstack/build/db9c3eb5629d449190566aba9d398b0a : SUCCESS in 3m 58s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/9a1c7961e1c047308d24eedae6620d62 : SUCCESS in 4m 58s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/be4fd27b67024953b2cedf4782422d96 : SUCCESS in 10m 18s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/3fa7b31187f241beba1618e5c71a86d8 : SUCCESS in 7m 57s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/7d47159e01794f3b837c32442d99c5b3 : SUCCESS in 28m 03s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/e754eee822bb414d88f2aa9de9816891 : FAILURE in 44m 45s\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/0905cf35cf2f4eb8a86705545a6ba35a : FAILURE in 54m 33s (non-voting)\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/ec4b7154d9b9493fb410fb3ecededa82 : SUCCESS in 1h 43m 38s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/3875ea9991744b61ba0362c6e09a5682 : SUCCESS in 29m 23s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/384614a28db64cabaf9eab6f42bb4b99 : SUCCESS in 43m 09s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/5975911ffa1540e5b44b81e9593b4090 : SUCCESS in 36m 23s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/dfe068800b57409992ff2f66b24d1ee0 : SUCCESS in 31m 59s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/b43755e0b7434178ad745a618cc282dc : SUCCESS in 35m 39s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/d5e26349990d4a8ab7c58206bee27886 : SUCCESS in 1h 38m 30s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-shard https://zuul.opendev.org/t/openstack/build/f198cca6f36f48ffbf853bfa7780ac83 : SUCCESS in 1h 00m 33s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/5368523c09d447a6af9889071e870d93 : SUCCESS in 54m 08s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/2e44d9c2365c406b9e728320e4d8b5ac : SUCCESS in 28m 10s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/06723c717e0f4d94acb9c3ce064620a9 : SUCCESS in 30m 01s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/82a7f4f7bf09410888017e54d7549da1 : SUCCESS in 38m 40s\n- metalsmith-integration-glance-centos9-legacy https://zuul.opendev.org/t/openstack/build/96176fe391fe4ed394c9b8b91dfa1069 : SUCCESS in 1h 02m 52s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/85b25b945bba4736aac48d9de279a759 : SUCCESS in 26m 57s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/4acc2ce731e4427d850f1db8bdb4cd13 : SUCCESS in 34m 30s (non-voting)\n- ironic-inspector-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/2c54e20201e44c9098db8be0b276ffee : SUCCESS in 31m 23s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/5b092d5b447747f8979893a6dbbcb29c : SUCCESS in 24m 03s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/224d1961e86346f080f61388ce079d2f : SUCCESS in 27m 14s (non-voting)\n- ironic-tempest-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/2c1f8d57b8f74d32abc237a9e5a42852 : SUCCESS in 52m 46s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/c5c33aba6dbf48c8a81f2fce170d86f6 : SUCCESS in 45m 58s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d7b3e41d7084effac0a1869b8cf3b3e : SUCCESS in 38m 33s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/779696aa215d40f0af80d94721435712 : SUCCESS in 45m 08s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/7d6aa06d07db4129a5c3ae116c8e866a : FAILURE in 13m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7d0d004cdb394345c4d1947cc0616f1960ad464c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-10-17 21:48:18.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":4},{"id":"778a3b4b32139d0390e40e626151ab54bcdd7e43","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-10-17 21:48:30.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"07dbae5f59dc4a63170768c6fa7a56767c9ee3f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-17 22:11:28.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/228c9e3ec3b443dfb46f89266c8cd5c5\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ce12e471341f41c09c5961d2b2bc6328 : FAILURE in 20m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/def6a30eaf9b481a8fab8251834d1387 : FAILURE in 19m 34s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"96c8c1af1bd6ad23fee31ab9f20fefb30829ebde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 00:00:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ced6d460a6f94cdf90ce7f243631e921\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66e637a961bc467d802f12efe356f103 : SUCCESS in 10m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a46a0eee42e4e408b8e4f469e345c8d : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/53dd6e76812f48f1a3f3750444dda714 : SUCCESS in 7m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ba53b8cddf18461992d4692711b66ee6 : SUCCESS in 8m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ba18afd881442e0a45cdc1b0511d5d1 : SUCCESS in 12m 07s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/0b9a4dd4b90f46b39d269681853e8ed6 : SUCCESS in 4m 55s\n- ironic-tox-codespell https://zuul.opendev.org/t/openstack/build/d3339d69c5994378a881c74e8b642ef9 : SUCCESS in 3m 43s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/ccce8f6f37914124a39b3e36ab292a98 : SUCCESS in 3m 55s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/ed565abeab3d462e87ab64e9a62157ce : SUCCESS in 8m 48s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/203f592c60d34c79b4b5bb2386b6ee17 : SUCCESS in 7m 08s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/306f80f377714b51a2ed04ff9ddf83f2 : SUCCESS in 34m 13s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/cf96dda73ea047a2a69ce870094985c6 : SUCCESS in 54m 19s\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/5d574c83bf4d4b21a95eca893b0982b9 : SUCCESS in 44m 52s (non-voting)\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/f682e9acef8048869cc9fb49ce86bcfa : SUCCESS in 56m 00s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/4e2ece210c8c4c14b6c31a21ad016829 : SUCCESS in 38m 12s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/20f45c8ba17a4feaa3ce9ed7963c3154 : SUCCESS in 38m 16s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/9d5f68a37df5497c81cc507a44d96859 : SUCCESS in 29m 05s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/87a164641c9a42c59783823dc409a630 : SUCCESS in 39m 45s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/8ad1780e272f4bc6a7bc647ade9603ec : SUCCESS in 41m 38s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/b20b12775f9e4283a7b9ebb39dfbf55b : SUCCESS in 1h 12m 41s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-shard https://zuul.opendev.org/t/openstack/build/18c9e5ea4b344c71928587aa17126976 : SUCCESS in 57m 23s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/e54b0fb18d024692917719b310eb44ca : SUCCESS in 1h 03m 01s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/0d410b43ed0f47a1bced064432f11d7e : SUCCESS in 33m 07s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/859a026bc0e9483ea01d4b2ddc093437 : SUCCESS in 31m 10s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/23b805f956764543b861796e751bb11a : SUCCESS in 53m 38s\n- metalsmith-integration-glance-centos9-legacy https://zuul.opendev.org/t/openstack/build/5869bf84da254e539f98820e6d68d392 : SUCCESS in 1h 04m 20s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/23a2ffd863f84a5d99074c7da65d88ec : SUCCESS in 26m 16s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/38b19374098643b496cf9d5d73196ddb : SUCCESS in 44m 37s (non-voting)\n- ironic-inspector-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/52dd9d85a11049359d63659d60376fd5 : SUCCESS in 49m 33s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/414d3e1c40494645b16122da51b7b7a7 : SUCCESS in 24m 34s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/53a49818af9f40af861aa115d8b1ccf9 : SUCCESS in 33m 14s (non-voting)\n- ironic-tempest-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/7ded5dd8425848029c1d94ec3d24c963 : SUCCESS in 49m 05s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/04e32dd887da45f28574b05670b67b14 : SUCCESS in 48m 26s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/41474a2b018d4a71927ac2ec844cdd6d : SUCCESS in 39m 26s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/dc58a0f2f1b94e9db7c63c75573eb3c2 : SUCCESS in 27m 27s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/eb36c0c95fc8493db41c66b49fa48b9d : FAILURE in 14m 54s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f34c1ea9dc8977d2548886f4b380879409a3ec0c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-10-18 03:56:01.000000000","message":"Patch Set 4: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"86aa5de0c666b6b3e1ac1f3feca8f0012c9b9178","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2024-10-18 06:51:46.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"3d4be2f829512b8ca9767b519b0d107bf9feefe4","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2024-10-18 16:35:28.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8487f5c9ecfc898a54c6777ffdd3c892ffac375a","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-10-22 17:01:38.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"deda71eda0c488bf61e5b47a08834030f2f131a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-22 17:02:38.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e3f08e31021a3e878d9e891cbd9aee6926cdf200","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-22 18:48:57.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/130c0378af324614b0b91026dd13353e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c35d52e2a934d31a82d9d06068ece08 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/89c81eb2cee44f94b74c83fb6f05b8de : SUCCESS in 9m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e6b9b88f272c42739f2be06f3a5f4967 : SUCCESS in 11m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8accd66d9d242af8e34ec777c9d2bc3 : SUCCESS in 14m 12s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/f83e838683554a4c92673985b9c6b0af : SUCCESS in 5m 19s\n- ironic-tox-codespell https://zuul.opendev.org/t/openstack/build/da0764c5caa843df9694ef3974dd6f79 : SUCCESS in 3m 49s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/01d7cc6f982c473b857f6c9ff8314e23 : SUCCESS in 4m 51s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/49042863a5b148178b40fdf11d108620 : SUCCESS in 7m 36s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/f66f6ee6d9fa43d9802894dda0f2eb24 : SUCCESS in 20m 52s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/b34407f3617544b6ba97c2da0c2b815a : SUCCESS in 1h 10m 00s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/4c9eff8f6f03429ca3ee3b663f3c5620 : SUCCESS in 42m 13s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/d7143750f26242c5b66a12972971b7c9 : SUCCESS in 36m 29s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/2c3c98c677784567b7253bb6cb4146e0 : SUCCESS in 26m 48s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/6613fe50fcba4899b1b6f438cc733506 : SUCCESS in 41m 56s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/fbc1cd4afacf42c1a600e875af049256 : SUCCESS in 39m 59s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/181f16b776ab4330ad38d7cdff371821 : SUCCESS in 36m 34s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/1c6800e5f7cd4d24bed449c478ef0f80 : SUCCESS in 40m 15s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/83c9ec6635c24c8bb0b9c601691ea6bb : SUCCESS in 33m 40s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/bd68d476f56b438b8ce4aa6035e29af3 : SUCCESS in 1h 03m 45s\n- metal3-integration https://zuul.opendev.org/t/openstack/build/b94b6a9d5a2548c3b41767a49c5919e5 : SUCCESS in 36m 03s","accounts_in_message":[],"_revision_number":4},{"id":"cd18b05018da32a68a312e65b02e8d88344df3fa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-22 18:48:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"976e3041cb2b85058027d29ee495ece373d58c66","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-22 18:51:49.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd445fc90a2b4134a7a5c9f1f2be7d47\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acc7dcb26a69464aa0de76cebefde540 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7b807eff731df8b774a017e6c8a9f6c9a7e35da7","revisions":{"8a16205682e46a7bb40fc2ae0d0a884415576662":{"kind":"REWORK","_number":1,"created":"2024-10-10 02:55:02.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/16/932016/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/16/932016/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/16/932016/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/16/932016/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/16/932016/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/16/932016/1"}}},"commit":{"parents":[{"commit":"a47cb572d6ac20df4f300fde164f3e45e3d39911","subject":"Merge \"docs: Add context around asynchronous device initialization\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/a47cb572d6ac20df4f300fde164f3e45e3d39911"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-10-10 02:51:57.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-10-10 02:51:57.000000000","tz":-420},"subject":"Run ironic-dbsync in grenade testing from venv","message":"Run ironic-dbsync in grenade testing from venv\n\ngrenade is going to install and run everything from virtual env\n\n- https://review.opendev.org/c/openstack/grenade/+/930507\n\nironic-dbsync in grenade testing needs to be run accordingly\nand not from system level. Otherwise it will fail with below error\n- https://zuul.opendev.org/t/openstack/build/6482462065a7401b9f6707488d7d816e/log/controller/logs/grenade.sh_log.txt\n\nChange-Id: I48ee1dd4adc2e5bcc18c5f116d979e7524248495\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/8a16205682e46a7bb40fc2ae0d0a884415576662"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/8a16205682e46a7bb40fc2ae0d0a884415576662"}]},"branch":"refs/heads/master"},"5b4bcbba280af0d2bf3968e85f1dc48c7821aad4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-10-16 00:39:47.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/16/932016/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/16/932016/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/16/932016/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/16/932016/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/16/932016/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/16/932016/2"}}},"commit":{"parents":[{"commit":"d2cea0b35947eb283be99318f255c18e2b5add02","subject":"devstack: respect USE_VENV in Ironic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/d2cea0b35947eb283be99318f255c18e2b5add02"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-10-10 02:51:57.000000000","tz":-420},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-10-16 00:39:47.000000000","tz":0},"subject":"Run ironic-dbsync in grenade testing from venv","message":"Run ironic-dbsync in grenade testing from venv\n\ngrenade is going to install and run everything from virtual env\n\n- https://review.opendev.org/c/openstack/grenade/+/930507\n\nironic-dbsync in grenade testing needs to be run accordingly\nand not from system level. Otherwise it will fail with below error\n- https://zuul.opendev.org/t/openstack/build/6482462065a7401b9f6707488d7d816e/log/controller/logs/grenade.sh_log.txt\n\nChange-Id: I48ee1dd4adc2e5bcc18c5f116d979e7524248495\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/5b4bcbba280af0d2bf3968e85f1dc48c7821aad4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/5b4bcbba280af0d2bf3968e85f1dc48c7821aad4"}]},"branch":"refs/heads/master","description":"Rebase"},"c1d652492073b9f70abeb077e33bd60070182e4b":{"kind":"REWORK","_number":3,"created":"2024-10-17 17:07:33.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/16/932016/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/16/932016/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/16/932016/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/16/932016/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/16/932016/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/16/932016/3"}}},"commit":{"parents":[{"commit":"45ee7dc365bbe614769dfe68ad706933a907fbc0","subject":"Merge \"CI: Add a 4k disk CI job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/45ee7dc365bbe614769dfe68ad706933a907fbc0"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-10-17 16:43:57.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-10-17 16:43:57.000000000","tz":-420},"subject":"Enable GLOBAL_VENV in ironic grenade jobs","message":"Enable GLOBAL_VENV in ironic grenade jobs\n\ngrenade by default enable GLOBAL_VENV which means it\ninstall and run everything from virtual env\n\n- https://review.opendev.org/c/openstack/grenade/+/930507\n\nWe faced the error in ironic grenade scripts in virtual env\nso GLOBAL_VENV was disabled explicitly. This fixing the scripts\nand enable GLOBAL_VENV in ironic jobs also.\n\nChange-Id: I48ee1dd4adc2e5bcc18c5f116d979e7524248495\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/c1d652492073b9f70abeb077e33bd60070182e4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/c1d652492073b9f70abeb077e33bd60070182e4b"}]},"branch":"refs/heads/master"},"7b807eff731df8b774a017e6c8a9f6c9a7e35da7":{"kind":"REWORK","_number":4,"created":"2024-10-17 21:48:18.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/16/932016/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/16/932016/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/16/932016/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/16/932016/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/16/932016/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/16/932016/4"}}},"commit":{"parents":[{"commit":"45ee7dc365bbe614769dfe68ad706933a907fbc0","subject":"Merge \"CI: Add a 4k disk CI job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/45ee7dc365bbe614769dfe68ad706933a907fbc0"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-10-17 16:43:57.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-10-17 21:48:11.000000000","tz":-420},"subject":"Enable GLOBAL_VENV in ironic grenade jobs","message":"Enable GLOBAL_VENV in ironic grenade jobs\n\ngrenade by default enable GLOBAL_VENV which means it\ninstall and run everything from virtual env\n\n- https://review.opendev.org/c/openstack/grenade/+/930507\n\nWe faced the error in ironic grenade scripts in virtual env\nso GLOBAL_VENV was disabled explicitly. This fixing the scripts\nand enable GLOBAL_VENV in ironic jobs also.\n\nChange-Id: I48ee1dd4adc2e5bcc18c5f116d979e7524248495\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/7b807eff731df8b774a017e6c8a9f6c9a7e35da7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/7b807eff731df8b774a017e6c8a9f6c9a7e35da7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
