)]}'
{"id":"openstack%2Fironic~940413","triplet_id":"openstack%2Fironic~master~I5198d9169cff8474c7a990332639b2d0758e6e1a","project":"openstack/ironic","branch":"master","topic":"ironic-container-url-support","attention_set":{},"removed_from_attention_set":{"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2025-01-30 18:39:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5198d9169cff8474c7a990332639b2d0758e6e1a","subject":"Fix agent from being locked out with complex steps","status":"MERGED","created":"2025-01-29 20:53:22.000000000","updated":"2025-01-30 18:41:30.000000000","submitted":"2025-01-30 18:39:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"940413-ironic-container-url-support","meta_rev_id":"af5864e6a810adf9300c603c1e93e9cd3c035ad7","_number":940413,"virtual_id_number":940413,"owner":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-30 18:39:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":2,"date":"2025-01-29 21:25:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":2,"date":"2025-01-29 20:54:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":2,"date":"2025-01-30 16:57:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":1,"date":"2025-01-30 16:57:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":1,"date":"2025-01-29 21:25:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":1,"date":"2025-01-29 20:54:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-29 20:54:28.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2025-01-29 21:10:41.000000000","updated_by":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-29 21:25:24.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"},{"updated":"2025-01-29 22:00:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-29 23:45:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-30 16:57:13.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"}],"messages":[{"id":"8fbb5184b71cc96c8b770540c05064d8db0c5e29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2025-01-29 20:53:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"160488d0653b8b763a8f6cc1c8823545a3cae56a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2025-01-29 20:53:31.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"a1451e28d4b5bc3dfcbe3da8b210ddc4709414e3","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-01-29 20:54:28.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8ea0bc7313445a08658dbcf220af1a5c0aa652e7","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2025-01-29 20:55:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ff7a10c01f4acf559ff61061c4ee137a296a4868","author":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"date":"2025-01-29 21:10:41.000000000","message":"Patch Set 2:\n\nTempest test failed in Fujitsu iRMC CI environment.\n\nIn this CI, we prepare 2 environments, one with FIPS enabled \u0026 the other is FIPS disabled, both with CentOS 9 Stream\nand PXE boot for Ironic node boot, then we run following Tempest tests in both environments \n  - ironic_tempest_plugin.tests.scenario.test_baremetal_basic_ops.BaremetalBasicOps.test_baremetal_server_ops_partition_image\n  - ironic_tempest_plugin.tests.scenario.test_baremetal_basic_ops.BaremetalBasicOps.test_baremetal_server_ops_wholedisk_image\n\nIt\u0027ll take few minutes till log appears in following URL.\n\n\n- fujitsu-irmc-ci-job-fips fujitsu-irmc-ci-job-fips : ERROR Unable to find role in /var/lib/zuul/builds/3431436b7c854a00ad724d7abf9ca5de/ansible/pre_playbook_1/role_0/devstack-gate in 29s\n- fujitsu-irmc-ci-job-nofips fujitsu-irmc-ci-job-nofips : ERROR Unable to find role in /var/lib/zuul/builds/b4748f923ff2445ea1ddc32693858d5d/ansible/pre_playbook_1/role_0/devstack-gate in 28s\n\nAbout this CI test, see https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_iRMC_CI","accounts_in_message":[],"_revision_number":2},{"id":"77ca1cda2fbdaa18774ead090f91ab78f5ebecae","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2025-01-29 21:25:24.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1a1c0ef6bad5a07cc773450ae98ff06077be5dab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-29 22:00:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80dd5cd43787471a9f9f8f7d98d681eb\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ef9db74d4e624d5c837ef6c2ac46744d : FAILURE in 21m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1697ff510f5e4f3d91e6219d3115ae19 : SUCCESS in 18m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"327019b5667979edba012cb469617f20b1fc4601","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-29 23:45:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bee2a1aa5c6947879847f28802a07532\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6b7af84aaec14a18b078b5e29f67e9c6 : SUCCESS in 5m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8839133e737446fd96ef926ea482f72a : SUCCESS in 11m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dafddfe8228845df856165985edacb01 : SUCCESS in 6m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/31eee5a5a0944cc5848106bd8d23eb39 : SUCCESS in 8m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f348fad29b384337b413a80245103235 : SUCCESS in 8m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6ad6bd152b44475b68642f46d75b0ab : SUCCESS in 15m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/95c2b803d6774bcba5348f3f00f42d98 : SUCCESS in 14m 39s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/9e8bf27cd6ae4dceb1db50c9225c08a6 : SUCCESS in 5m 05s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/3cdb343bec1c433183d0165293c824cd : SUCCESS in 8m 30s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/2323c31cb9db407fb232737098658727 : SUCCESS in 21m 54s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/7edb490a3c8340b687d8403f90c04c4f : SUCCESS in 35m 44s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/f5030b5f8b834f228a740cebcdd27589 : SUCCESS in 53m 28s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/bafb4ed1b4554a3f8990713d11d813bf : SUCCESS in 37m 06s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/25b6769922384ea2bcfc157bc5d1e0e9 : SUCCESS in 1h 26m 29s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/7fbd8b9c08b845e6bdc355cffa171198 : SUCCESS in 24m 05s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/2cdf6dfc23304e7cbcdfb1ba440d7aff : SUCCESS in 39m 04s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/1a4dd2210c504f89bc99fceae4788168 : SUCCESS in 1h 17m 53s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/775a4156717945a98a16cca517a27a5e : FAILURE in 2h 29m 32s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-shard https://zuul.opendev.org/t/openstack/build/e3c7c713e70644be94a2e0579230e84e : SUCCESS in 56m 24s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/608bc892660144a484b35062937e0f19 : SUCCESS in 50m 52s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/8788e9803b554c6b916790c2b9c72ba1 : SUCCESS in 41m 20s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/b5a01ccd6470484881b3db4f3adc2871 : SUCCESS in 1h 34m 42s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/8ca05b84768a42bb915289238541e22a : SUCCESS in 1h 58m 32s\n- metalsmith-integration-glance-centos9-uefi https://zuul.opendev.org/t/openstack/build/52431fabd02348b5b3d2fbd1de421599 : SUCCESS in 49m 39s (non-voting)\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/f1ad855bb7134fa3970d118ade7b6a39 : SUCCESS in 1h 56m 45s\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/b9683d92b8984198b6a2700f03d0b762 : SUCCESS in 1h 53m 14s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/13261998bb0d4b3481427b1a2b11c34e : SUCCESS in 53m 54s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/b80f39907a6f45349ae3a1edc1f4a3d5 : SUCCESS in 1h 44m 59s (non-voting)\n- ironic-inspector-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/971d19dc7ae147f9870c48deaab693ba : SUCCESS in 1h 38m 39s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/3fa5b8c3fcc243c1a1a6f5a5c48beff5 : SUCCESS in 33m 02s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/a4526d7fcb4d4bdeba430d671c5032e8 : SUCCESS in 38m 42s (non-voting)\n- ironic-tempest-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/c1f2fed3546a4afeaa2a10f851a90372 : FAILURE in 1h 02m 38s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/69ccde6f0f0b48ef919361771eff8365 : SUCCESS in 1h 21m 18s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/028a1431360a48e79c9d9c27ac413892 : SUCCESS in 44m 53s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/c2d4095a114146a5b3e82b6b75f9c3ab : SUCCESS in 42m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"18d7b9099c1300bff7da99eb2369ee6f7183302d","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-01-30 16:57:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8695cdb0fc6dcff8a2e0053933ce252c0abd9d99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-30 16:57:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fe85061f82a14f4f9286cf046b73a73daeb3e911","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-30 18:39:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a65b8a7cf3e649f489ee9ed00376157c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8a47fee1acdf4729a24d6ef5ce9f7ca3 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2b59be4e773495bb487a09cbc527a84 : SUCCESS in 9m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/818246f88c39433ab87b74f76e63834d : SUCCESS in 9m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c01a25b838dc488887b7a89b5ad91879 : SUCCESS in 7m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e59e98e8b9d642838f0fc83c0c9d46b7 : SUCCESS in 13m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b17726093feb4657a48af42aa9635b53 : SUCCESS in 15m 17s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/b2ef293ee6ee44f09d8e0169280fbe7b : SUCCESS in 5m 30s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/ace75d694db64f76a264813a739df861 : SUCCESS in 12m 16s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/d3700d8485644d6793ba7dc1c9ba86ba : SUCCESS in 43m 30s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/ccf0bbeda95e40f98c81e23e6a7e167f : SUCCESS in 1h 15m 22s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/d25a580d5ebc4f6fb8d412339590d8dd : SUCCESS in 1h 08m 59s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/0b570fa0a78c41cb8f9a9b2297dbe3ff : SUCCESS in 47m 56s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/a3608ea281134bac88a26f4a15d5038d : SUCCESS in 42m 25s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/3ab0b546a6264f92922d0de5d229b5c8 : SUCCESS in 58m 13s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/9ea2b1e795a74f508ea52dee8417a47e : SUCCESS in 52m 08s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/8749ac310e3747fa877adab0210e1efe : SUCCESS in 1h 09m 17s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/153f42b065f549eb897f81dafb72e09f : SUCCESS in 48m 47s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/9e2a810ade794d3b8ef7986bc0f01558 : SUCCESS in 52m 32s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/84b74ba7720e4565a668b89e31fe95fe : SUCCESS in 1h 25m 36s\n- metal3-integration https://zuul.opendev.org/t/openstack/build/42e5639a2a6b450e9bb847fe2431a41e : SUCCESS in 57m 12s","accounts_in_message":[],"_revision_number":2},{"id":"5060a25d2603c6dadd3143d518860ac12845ab34","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-30 18:39:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"af5864e6a810adf9300c603c1e93e9cd3c035ad7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-30 18:41:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13094100463d403883a8c9b5b1d6c4e7\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/2611813d381e4c9c84f0d481fd7cf906 : SUCCESS in 55s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63d82085d09642318c4a6fe3cf23b0db : SUCCESS in 1m 27s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/059624a84bdb4e3d85f9b42c1042ef45 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5db194c5030f2e71a2bdb6c8bf9ec3d4cedc9f9a","revisions":{"f61c3e04b5eb3e82504d3d2b108ccc47318a0691":{"kind":"REWORK","_number":1,"created":"2025-01-29 20:53:22.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/13/940413/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/13/940413/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/940413/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/940413/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/940413/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/13/940413/1"}}},"commit":{"parents":[{"commit":"1ada2336a0f7645a5c5da0e084ba6aedc8be4429","subject":"bootc deploy interface - for bootable containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1ada2336a0f7645a5c5da0e084ba6aedc8be4429"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2025-01-29 20:49:50.000000000","tz":-480},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2025-01-29 20:52:30.000000000","tz":-480},"subject":"Fix agent from being locked out with complex steps","message":"Fix agent from being locked out with complex steps\n\nWhen mixing in-band out-of-band steps, the out-of-band status polling\nflag was not being cleared, and was being left to remain in the node\ndriver_internal_info field, thus preventing future heartbeat operations\nfrom the baremetal node from being processed to check the actual\ncompletion status of a step.\n\nWe now always clear the field based upon the workflow in-progress\nbefore starting a new step and should asynchronous steps also\nbe recorded as a result of any step\u0027s actions such as if a reboot\nis required.\n\nSpecial thanks goes to keekz for promptly providing upstream with\nthe information necessary for us to identify the root cause.\n\nCloses-Bug: 2096938\nChange-Id: I5198d9169cff8474c7a990332639b2d0758e6e1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/f61c3e04b5eb3e82504d3d2b108ccc47318a0691"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/f61c3e04b5eb3e82504d3d2b108ccc47318a0691"}]},"branch":"refs/heads/master"},"5db194c5030f2e71a2bdb6c8bf9ec3d4cedc9f9a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-01-29 20:53:31.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/13/940413/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/13/940413/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/940413/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/940413/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/940413/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/13/940413/2"}}},"commit":{"parents":[{"commit":"81b3612046b6b4de9a3d106c0f86ef7641c29973","subject":"Merge \"Add lsblk output to metal3 logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/81b3612046b6b4de9a3d106c0f86ef7641c29973"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2025-01-29 20:49:50.000000000","tz":-480},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2025-01-29 20:53:31.000000000","tz":0},"subject":"Fix agent from being locked out with complex steps","message":"Fix agent from being locked out with complex steps\n\nWhen mixing in-band out-of-band steps, the out-of-band status polling\nflag was not being cleared, and was being left to remain in the node\ndriver_internal_info field, thus preventing future heartbeat operations\nfrom the baremetal node from being processed to check the actual\ncompletion status of a step.\n\nWe now always clear the field based upon the workflow in-progress\nbefore starting a new step and should asynchronous steps also\nbe recorded as a result of any step\u0027s actions such as if a reboot\nis required.\n\nSpecial thanks goes to keekz for promptly providing upstream with\nthe information necessary for us to identify the root cause.\n\nCloses-Bug: 2096938\nChange-Id: I5198d9169cff8474c7a990332639b2d0758e6e1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/5db194c5030f2e71a2bdb6c8bf9ec3d4cedc9f9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/5db194c5030f2e71a2bdb6c8bf9ec3d4cedc9f9a"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
