)]}'
{"id":"openstack%2Fironic~950461","triplet_id":"openstack%2Fironic~master~If47e74751ba66a1296f16d9c43433033c04beffb","project":"openstack/ironic","branch":"master","attention_set":{},"removed_from_attention_set":{"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2025-05-22 02:41:23.000000000","reason":"Change was submitted"},"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2025-05-22 02:41:23.000000000","reason":"Change was submitted"},"4571":{"account":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"last_update":"2025-05-22 01:34:35.000000000","reason":"\u003cGERRIT_ACCOUNT_4571\u003e replied on the change","reason_account":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"}},"24828":{"account":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"last_update":"2025-05-22 02:41:23.000000000","reason":"Change was submitted"},"38045":{"account":{"_account_id":38045,"name":"sandzwerg","display_name":"sandzwerg","email":"launchpad@sandzwerg.de","username":"sandzwerg"},"last_update":"2025-05-22 02:41:23.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"If47e74751ba66a1296f16d9c43433033c04beffb","subject":"ci: Remove code which has been long-dead","status":"MERGED","created":"2025-05-20 20:03:31.000000000","updated":"2025-05-22 02:43:39.000000000","submitted":"2025-05-22 02:41:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950461","meta_rev_id":"b88f4457a2a7d220f8455cfc3a3fd9af1e13571d","_number":950461,"virtual_id_number":950461,"owner":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-22 02:41:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":38045,"name":"sandzwerg","display_name":"sandzwerg","email":"launchpad@sandzwerg.de","username":"sandzwerg"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":2,"date":"2025-05-22 01:34:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":38045,"name":"sandzwerg","display_name":"sandzwerg","email":"launchpad@sandzwerg.de","username":"sandzwerg"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":1,"date":"2025-05-22 01:34:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":38045,"name":"sandzwerg","display_name":"sandzwerg","email":"launchpad@sandzwerg.de","username":"sandzwerg"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":38045,"name":"sandzwerg","display_name":"sandzwerg","email":"launchpad@sandzwerg.de","username":"sandzwerg"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"_account_id":38045,"name":"sandzwerg","display_name":"sandzwerg","email":"launchpad@sandzwerg.de","username":"sandzwerg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-20 20:33:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-20 22:07:05.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2025-05-20 22:13:18.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"REVIEWER"},{"updated":"2025-05-21 00:23:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-21 10:09:26.000000000","updated_by":{"_account_id":38045,"name":"sandzwerg","display_name":"sandzwerg","email":"launchpad@sandzwerg.de","username":"sandzwerg"},"reviewer":{"_account_id":38045,"name":"sandzwerg","display_name":"sandzwerg","email":"launchpad@sandzwerg.de","username":"sandzwerg"},"state":"REVIEWER"},{"updated":"2025-05-21 14:10:08.000000000","updated_by":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"reviewer":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"state":"REVIEWER"}],"messages":[{"id":"6c8a9b6ddb81c17c373720f37f4beb6fbff38e24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-05-20 20:03:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4e8691b6a9ebc1ed8ed2dd843a58af2347ee23f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 20:33:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48bdde436e4141c588ccaeb127afbf80\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ba25f96b516c426abf1b4837e9fae0ac : SUCCESS in 16m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/390ffc1910b44667840357234de5c406 : SUCCESS in 8m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/086b2a7686584f93913d62df653f5055 : SUCCESS in 13m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f450adb65f5d489520b0ee8b448419cba8232ec0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-05-20 21:03:58.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"24137c5e0a12e66e77544c00ca1e274c5f1366c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-05-20 21:10:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"844f74be7d5c20db4777e97e8ac35af6007721d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-05-20 21:23:29.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"deeea6347bc6289f086966381e794761e8c31f0e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 21:45:06.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a9acc852c66477d86de93b1e3285c16\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8ecceaa9a38e49ec9bba8cfc7f7065e1 : SUCCESS in 8m 58s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cf25f24e7ba9469ea3e24144d2bb65e7 : SUCCESS in 9m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f5b5dac41f4e4afda93665833d797462 : SUCCESS in 9m 31s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b618b3cc56a7df634378bee67a6631bef89d3472","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-05-20 22:04:11.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":4},{"id":"f568e524b79e4ee397c4fd3c2d08ef4b6df3712f","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2025-05-20 22:07:05.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"c400a37fc2a5997c307ebabb0fb0a8d7f4620f96","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2025-05-20 22:13:18.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"43b96c54ec748d9e618b0f705996eacc594febfa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 00:23:45.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/320515d4ed8c42668b87c6ad1321d65e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a7f6f750e20248a9995de827becad18a : SUCCESS in 11m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f03102df89349c6bc30d77098c9ed96 : SUCCESS in 6m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e2c587537e5e41e4a142a7720a7bdb76 : SUCCESS in 8m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9f22d798afd24d9dae4007eb51f8a8b4 : SUCCESS in 7m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3c114a68ef284b2baa2c3ca7c6e218fe : SUCCESS in 10m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae16526bca494f8bb45da56439875fb4 : SUCCESS in 14m 33s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/75da0bda9d6d46ca8f54db8d54db4509 : SUCCESS in 2m 44s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/681a23af1b544b818cf58ac9e2bd2951 : SUCCESS in 7m 37s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/1e585fdb3d53459a87f5b0762c2b9f48 : SUCCESS in 38m 01s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/59ca19960e344aee813da7874a87ed0a : SUCCESS in 1h 00m 36s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/d9755a22e28a47c9ba62351f51a66394 : SUCCESS in 33m 10s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/787b04d4f0cc4488a73af7c9c7a78355 : SUCCESS in 20m 30s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/f8deb5e3a7a341b49724689cd2c7f321 : SUCCESS in 16m 21s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/6a476ea50eaa4cbca883354556692972 : SUCCESS in 40m 23s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/d9f7cabe47f94b6c9e542b8549950e1c : SUCCESS in 35m 16s\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/55cf2727b4c54f06b897353cf71ea0a3 : FAILURE in 52m 36s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/7600e0d27935436bab203861aab57357 : FAILURE in 2h 53m 42s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-shard https://zuul.opendev.org/t/openstack/build/5bcae6436f8b40ada2c7ad3ee4e9e9ae : FAILURE in 46m 58s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/c4a1dd2c9ad24f56814bef95be78d177 : SUCCESS in 1h 00m 33s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/aeb29567d14b42d881f65f333b4fb70c : SUCCESS in 33m 58s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/919096f0e5624997956a354e10e417bb : SUCCESS in 40m 07s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/91ec12afa5794254a335b02918868823 : SUCCESS in 1h 05m 36s\n- metalsmith-integration-glance-centos9-uefi https://zuul.opendev.org/t/openstack/build/169342561d80449d8747ca2392c06c83 : SUCCESS in 1h 20m 03s (non-voting)\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/9ee1c4de4fe345229304f523abc92c9e : SUCCESS in 56m 43s\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/f10e60596e7a40e0ae9885c36540466e : SUCCESS in 1h 03m 10s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/425f2b0c15cc4fd2b247c626df76c93b : FAILURE in 1h 08m 46s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/b520b1f85b7e4be1831c45bf187b6cec : SUCCESS in 35m 42s (non-voting)\n- ironic-inspector-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/90b674ba48e9427daa3a8e2ca60dd560 : SUCCESS in 35m 56s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/38459e88a18b477cb5bdb61ad2f1adbd : SUCCESS in 35m 21s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/8b0dc320b8ca439aa0461dfd0aada9c6 : SUCCESS in 35m 23s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/9879a4d1b813492093e658d238fb2933 : SUCCESS in 36m 30s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/776a4eec8afd45a49b0e4d829b806800 : SUCCESS in 39m 33s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/29429c77fdc84cb59c91d67ca8300c71 : FAILURE in 10m 33s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1a706cc2bfcd197cc9c58f9b57e4bc6668bccf72","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2025-05-21 00:53:18.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e5d62b4262684615324454af2a37b73a02773573","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 01:09:01.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15d6680de0964168ba67d8f6bc5b1be0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fa660cf7981f473aacc1b783a33d9b9e : SUCCESS in 10m 40s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/158d47d0f7034236933ddb3833886568 : SUCCESS in 10m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c986471935084851a8e1f795e5c80cb3 : SUCCESS in 13m 54s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"30961c1a0b06e2ce656feca2d016c6b16c97a219","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 03:52:34.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/28bfad36f35c4a258207fdeb59e55c9b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ccda34401ee745cfac0233b8a382f79a : SUCCESS in 11m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc4113994c6047678f0f4f15f67d5af8 : SUCCESS in 6m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6a265ba25854499397acc5e1d419f365 : SUCCESS in 4m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7fdaaa61c95544b3b06099dae875a19c : SUCCESS in 7m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/871e708d4c12441bb94cb44e329c4795 : SUCCESS in 11m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a46a59534404e7f9c7d2c8ad85418d8 : SUCCESS in 11m 11s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/a4ad0ce417c646fa9a43e2bd3392392f : SUCCESS in 4m 33s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/ff6c8d500d2b40f689f1a39cf49165da : SUCCESS in 8m 39s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/b9c72716fcf94c63b512d04a4c063300 : SUCCESS in 37m 06s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/86e59401148949d0b6f9b6e4ac34f599 : SUCCESS in 1h 29m 50s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/cbfa3d1cfc7e4fc58cb1edd643d2f9e4 : SUCCESS in 43m 14s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/796e20ed24d14a619a805aa75e3e26b4 : SUCCESS in 41m 36s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/e754c934baa8470b9d83c366fa57cf7d : SUCCESS in 30m 06s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/3672b09ec56b4bd4bfef0ce34e671612 : SUCCESS in 42m 32s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/07ae8a2cccae48abbbb8b46f728ddf3a : SUCCESS in 39m 39s\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/4c722d1542314543a9bfc29cabf3b44a : FAILURE in 1h 08m 49s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/8afb7c970ed34e8a880d2552d2983981 : FAILURE in 2h 54m 09s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-shard https://zuul.opendev.org/t/openstack/build/808b079882504455887818571167fcbc : FAILURE in 46m 14s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/f1392e45308d433fabf2e1b2eacc8bc3 : SUCCESS in 27m 10s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/8b1a94107c1e45cbb23f7a52fea30f60 : SUCCESS in 29m 21s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/ec19ffc966f94ead91e97c45557616ed : SUCCESS in 42m 21s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/92a26a08d9fe47a0931c90841a6b7327 : SUCCESS in 44m 51s\n- metalsmith-integration-glance-centos9-uefi https://zuul.opendev.org/t/openstack/build/3d9d8de5fdf74f33ba170ea03cf1a4cf : FAILURE in 32m 05s (non-voting)\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/84f80caf693b443e8ef7acf655e04ec6 : SUCCESS in 31m 13s\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/fc59cdbde1e5476db94d099de02f0f67 : SUCCESS in 53m 32s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/69b7557effec43c88391d31b4008999b : FAILURE in 1h 18m 04s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/2282f299475d42629e8b204014335db0 : SUCCESS in 33m 29s (non-voting)\n- ironic-inspector-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/d52d1f266c0b484b8714abb46d5e5e91 : SUCCESS in 28m 59s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/572e539823d0421693ed0eccbdfeb47c : SUCCESS in 39m 21s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/7d100453d71447a8a60ebb24fb676398 : SUCCESS in 32m 36s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/108d9e0d60744beb96a4df2239df2c19 : SUCCESS in 29m 14s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/985b06a2898f41ad966ab78069c41ded : SUCCESS in 19m 08s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/d3a40c08a670494c9e9ed9cec1818e86 : FAILURE in 10m 20s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"33ca3baa0f0903825195007eec31ed76be8fe8a7","author":{"_account_id":38045,"name":"sandzwerg","display_name":"sandzwerg","email":"launchpad@sandzwerg.de","username":"sandzwerg"},"date":"2025-05-21 10:09:26.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"d0fb4744392c4c15da62c9e6e6400dcf5416e0a6","author":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"date":"2025-05-21 14:10:08.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"39a0aa7b6e7afa215292e4c8da56a364c883edb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2025-05-21 16:47:48.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":5},{"id":"f006b23773f9f374eb273b0abd1dc57164e3d85a","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2025-05-21 16:49:12.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8d1c159ba773622af4fd8932012db288f9dedb6b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 17:23:04.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e18ab7d1a170452a8921b5a589bdd919\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/74b2211f9f7f4c589101adbe8d5482a0 : SUCCESS in 12m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/59a27df15b0248b6a74c69c4b132ea8d : FAILURE in 12m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/dad174bd9a7246d38ae2fda0bfc7f2ff : SUCCESS in 22m 12s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6c748aac29a1b8fd26774f51a9214fa731242c1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 19:47:24.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b92dad734444c16ae65c05ee23ea35a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c810fabc13c04a719d801cbd4770a309 : SUCCESS in 11m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4252bb407cf3410fb6c27977018bc30e : SUCCESS in 7m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c4be0d0150114d13aa2c733382fa079b : SUCCESS in 7m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0a2ff423382b42aa90d2185a5a4f58f6 : SUCCESS in 8m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a3abc78cdc5d487497b07e2dbebbb1ab : SUCCESS in 12m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbd814664b1941b484c42b92deb53327 : SUCCESS in 13m 25s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/4511c3a446904065a4ff60de07e20a72 : SUCCESS in 5m 15s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/c89780eafa22466f82b9f3e4cae9dd2f : SUCCESS in 9m 27s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/dbd6fd3fd796442daa5fc8c9b3cac751 : SUCCESS in 7m 55s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/cebae1fe51704371be61b2b997a32080 : SUCCESS in 42m 51s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/587bbe55265f43709acf44c770e8c6a7 : SUCCESS in 1h 31m 03s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/3307e2aa2c6e44cf9640abff388f06db : SUCCESS in 36m 09s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/3c73a802df204bf281901a82ec46b4f1 : SUCCESS in 41m 39s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/4af758b7ccf54447bed47703ad994db7 : SUCCESS in 32m 04s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/671847871ab3498fbd59b03ed8e5e1e0 : SUCCESS in 44m 32s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/73e6f73753164043811ba36f90dd5104 : SUCCESS in 41m 05s\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/0b7afb8efe3d494b9c9b95b583ac0c8e : FAILURE in 1h 13m 39s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/1fc10fbbdf45416b8513dd75561d34e6 : FAILURE in 2h 55m 17s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-shard https://zuul.opendev.org/t/openstack/build/4685e2f1d88e4536b6ba846502e63f4f : FAILURE in 1h 06m 34s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/091a5af3eb2c42fbb50179215f872faf : SUCCESS in 1h 47m 51s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/4348da21791e4f04a3b9eb32985ec221 : SUCCESS in 42m 10s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/e24b17e12fba410991e1fc2b765576f5 : SUCCESS in 43m 33s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/b20c942150a846348aee4023bd5290ee : SUCCESS in 26m 26s\n- metalsmith-integration-glance-centos9-uefi https://zuul.opendev.org/t/openstack/build/da34a4ce4a0c476abd91c4f2ea78ad65 : SUCCESS in 1h 15m 58s (non-voting)\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/3827dc304811456da04daa7d72aed51f : SUCCESS in 54m 14s\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/7607aee010ad4984bc9ca022c370c8d7 : SUCCESS in 56m 43s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/f24de49958fe47f89ba4aa2221533c22 : FAILURE in 1h 04m 18s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/37334351d86c41a0a534c55c54ba767f : SUCCESS in 38m 05s (non-voting)\n- ironic-inspector-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/b7a9407d227045da8d023128e05e437d : SUCCESS in 36m 57s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/32871b2722a646978b18302145b12639 : SUCCESS in 34m 35s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/498a827f1b8d4a1fa181fcdbf938ef7f : SUCCESS in 36m 54s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/efcbb8245bf541099bfafc6d8edde2c6 : SUCCESS in 39m 44s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c1e312f7794246eea226e46b787fa3d7 : FAILURE in 40m 06s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/d2d5576d502d424784421232ff1f9451 : FAILURE in 12m 50s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"0a24afaf57685b14fce82b658e065df8c895b0ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 19:47:57.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"19af412efa50c35f22566b99def23cc8ddcb6401","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 20:59:48.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bc46c5872fe744e1b5d06a9c7a3a3481\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6266aa432594eeaaaf0a35f350e6b64 : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2abd22aeba2f4e79ba511592b54fb3a9 : SUCCESS in 8m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4d35ea30c139462099f44f13ac8cc113 : SUCCESS in 6m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16a89c46949448228fdd6fd7dd1b75e9 : SUCCESS in 11m 25s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/16f817d7053142338f370fe457ae1b75 : SUCCESS in 5m 04s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/5d78a69ba5c040e18c4f18cac93a146a : SUCCESS in 8m 44s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/2d9cad0446a144cf9915243411742210 : SUCCESS in 27m 21s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/696fe8df93724e1caebbc6e8fcb0b656 : SUCCESS in 1h 07m 45s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/b9ca242d47e145c4b61be1c4f5639786 : SUCCESS in 39m 32s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/5874cebb8c9b410e89eaa337bc1a5dcf : SUCCESS in 40m 41s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/642ee89eadd74a09a93f0e4eca929a28 : SUCCESS in 31m 26s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/68b6d32ccc1a41b08c434be8554ee798 : SUCCESS in 42m 26s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/2e345f31465146a397d87687999c0059 : SUCCESS in 43m 02s\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/87022b746f5c45c2baa947a536d8b232 : FAILURE in 1h 11m 21s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/7e72c8498f3a447f9417c0992a0a3930 : SUCCESS in 59m 18s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/5567ae5c79c84dd1b9f1c55a71abac7d : SUCCESS in 42m 57s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/6426ba9f1ce140a98bdfc05b0a56104a : SUCCESS in 43m 33s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/6011d7978b6b48c8b61eb68abef36577 : SUCCESS in 57m 56s","accounts_in_message":[],"_revision_number":5},{"id":"989aa69c122f0faf0d08b30c41d7a50ab8c7e6d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-05-21 21:02:25.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":6},{"id":"8e493e4dfbb569d5a4d251a5a1f35e76d5ecd032","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 21:17:01.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1520f131aa64480921f5bcf1890c758\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a399352b60844995b8c5b32c95120c48 : SUCCESS in 8m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8dc438bf226a4ab7886500ce7ef5fb61 : SUCCESS in 10m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ae29342cfcaf4285bf24b874847042d3 : SUCCESS in 12m 51s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b144f38ec68dad6decfdda711d963d4b15ef5b47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 00:02:36.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f131a28101a4d7a82723d824e0c493f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f7483966cda4f31811d9f3a7d5896a7 : SUCCESS in 8m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0480f75dbffd44e4a3f4f7570cb4f2ef : SUCCESS in 6m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8a3cc5be9ac6415582e67f17d6d9cb5a : SUCCESS in 7m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f472c7f312ba41848765bca4477565a0 : SUCCESS in 8m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1b8a90ee99d649379169301da2262d39 : SUCCESS in 13m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b090fc5760349d5be617420e022f6db : SUCCESS in 11m 42s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/b6f969b843b1467fa7cb7eefefca98e2 : SUCCESS in 2m 41s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/c9fcd3168a1640e89dd2ecee06987e8b : SUCCESS in 9m 07s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/17f09448ff2e4d8b9cf4c121e2f9a938 : SUCCESS in 4m 18s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/738881b1f0e2437192b4c7d006dca3d1 : SUCCESS in 33m 14s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/fefacea48f984348ab74d1ea6e6e06e5 : SUCCESS in 1h 18m 31s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/c20feabc5e9d4f59a3986c711ae993dd : SUCCESS in 17m 38s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/f5e9420e1baa46c1a2228f65e158c946 : SUCCESS in 45m 42s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/1b8eb93f9f164e0dbd1e24295275b13a : SUCCESS in 31m 27s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/422312c79edf4432800239d13f20eaf7 : SUCCESS in 43m 20s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/b30feac7bd704ec5b0e751e2c0aa1793 : SUCCESS in 40m 59s\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/a46ee970407341fd85bd66d59f53b4b7 : FAILURE in 1h 06m 42s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/622b4c3d662e4646bc5b8d461736883d : FAILURE in 2h 48m 09s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-shard https://zuul.opendev.org/t/openstack/build/e809bc31efeb459b9baf8ad30f1cc2d0 : FAILURE in 46m 52s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/26af67ecd8cc4cf191e6ae9fbf736ed4 : SUCCESS in 59m 59s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/1eda7443af6640d5bec63ad184f68aa6 : SUCCESS in 42m 31s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/8190f151f0b94ac183dea7dc246433ad : SUCCESS in 34m 57s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/9d9dc0fa456c441fa034f41c76f47d96 : SUCCESS in 53m 14s\n- metalsmith-integration-glance-centos9-uefi https://zuul.opendev.org/t/openstack/build/2a244ab0bc1242de845b246354c29e9b : SUCCESS in 1h 27m 19s (non-voting)\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/322ec4d7be3147e5a1d5a69067905796 : SUCCESS in 1h 08m 28s\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/faff693e297e46e7bf79ba655d593448 : SUCCESS in 1h 01m 07s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/7b044531cbd54d14bab50fd025450c1e : FAILURE in 1h 26m 59s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/5508cf2de7134907a9e62d0b795c2a1a : SUCCESS in 34m 24s (non-voting)\n- ironic-inspector-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/93ec3a2935224b29a59341ba99efeef9 : SUCCESS in 43m 57s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/4a68bb18a35d4b81a56487697d3db0c8 : SUCCESS in 36m 40s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/998a61956e614d5aa5bc172aaff737e7 : SUCCESS in 44m 14s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/eb5d31be45c34eeaae7d5ae07441e6b1 : SUCCESS in 43m 24s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/96b1996a2d704a1e95abff8f6c9cc307 : SUCCESS in 37m 19s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/bfe679d532a94bbba93dbcfd82af0dc8 : FAILURE in 11m 19s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"dc35cf67f5e8b3d516116728d9640a5b0082ffc1","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2025-05-22 01:34:35.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"3ffec5cecf4e698f189116b4f18548b044db8689","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 01:35:06.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"8f41195af671629db28729c7a35a37e4f9fedacb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 02:41:23.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c5f589475af4e899a75f85d5f06c323\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7b229eb365141ac8fdb3ec0892f8988 : SUCCESS in 5m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4f4a31fe04cc4ee684c54ab23d46b51b : SUCCESS in 7m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b5606f3960bc48a39a77fac85cee4791 : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/642f506ed56f47f09271e4e5a9269ee0 : SUCCESS in 13m 18s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/076fdae30b7442dfb162516a8b98db2a : SUCCESS in 4m 30s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/5cfa9088808d4413833d48bbb0d03604 : SUCCESS in 8m 31s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/2c46f32727df4b0a81aa05e2d836a7a9 : SUCCESS in 29m 51s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/dfb7414f60ed4578a0615608cfa51176 : SUCCESS in 36m 00s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/daa8ab133a8c4a43ae101e3b9c6b26e0 : SUCCESS in 43m 29s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/258cd96ecfaf44e0915380ff42e77c60 : SUCCESS in 47m 54s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/ab8f50886fe047f89e00aae1c15d13e0 : SUCCESS in 17m 42s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/1936f78846fc402e8f1a77ada0790448 : SUCCESS in 45m 12s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/4a99a24f83b24ba080c352de760e0f68 : SUCCESS in 41m 56s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/039af1ba73d74152be8021cbe848cf4f : SUCCESS in 47m 18s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/2a0b22cc574b496d951e2da202aed37d : SUCCESS in 45m 05s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/0f409022858748f1a6a941fdbf9194d7 : SUCCESS in 41m 40s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/9672fff8bcc944c78a861d7498388ec7 : SUCCESS in 1h 02m 43s","accounts_in_message":[],"_revision_number":6},{"id":"13dc6f06b925dd3eb397b39a6e9491ca3156fdea","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 02:41:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"b88f4457a2a7d220f8455cfc3a3fd9af1e13571d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 02:43:39.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55f7c3020bb44cff8c6ae4554513a82b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73aeb2b298aa4807a6f3d742bb7415f9 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"56d43ff56d900377d379b54e4345ac7cb7f1381b","revisions":{"f27c2fd92c7b73bf5584479f30bc98e3b9ef2b4b":{"kind":"REWORK","_number":1,"created":"2025-05-20 20:03:31.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/61/950461/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/61/950461/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/61/950461/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/61/950461/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/61/950461/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/61/950461/1"}}},"commit":{"parents":[{"commit":"4f22bade301d6957743e3913c221f3872e39fdfc","subject":"Merge \"Fix redfish driver URL parsing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4f22bade301d6957743e3913c221f3872e39fdfc"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2025-05-20 20:03:10.000000000","tz":-420},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2025-05-20 20:03:10.000000000","tz":-420},"subject":"Science: replace start_neutron_api with start_neutron_service_and_check","message":"Science: replace start_neutron_api with start_neutron_service_and_check\n\nBreaking devstack changes makes pixie boots sad\n\nChange-Id: If47e74751ba66a1296f16d9c43433033c04beffb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/f27c2fd92c7b73bf5584479f30bc98e3b9ef2b4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/f27c2fd92c7b73bf5584479f30bc98e3b9ef2b4b"}]},"branch":"refs/heads/master"},"816df84cbf067922411204593d3d5f19fa7ae652":{"kind":"REWORK","_number":2,"created":"2025-05-20 21:03:58.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/61/950461/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/61/950461/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/61/950461/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/61/950461/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/61/950461/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/61/950461/2"}}},"commit":{"parents":[{"commit":"4f22bade301d6957743e3913c221f3872e39fdfc","subject":"Merge \"Fix redfish driver URL parsing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4f22bade301d6957743e3913c221f3872e39fdfc"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2025-05-20 20:03:10.000000000","tz":-420},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2025-05-20 21:03:58.000000000","tz":0},"subject":"Science: replace start_neutron_api with start_neutron_service_and_check","message":"Science: replace start_neutron_api with start_neutron_service_and_check\n\nBreaking devstack changes makes pixie boots sad\n\nChange-Id: If47e74751ba66a1296f16d9c43433033c04beffb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/816df84cbf067922411204593d3d5f19fa7ae652"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/816df84cbf067922411204593d3d5f19fa7ae652"}]},"branch":"refs/heads/master"},"d5b2c012d4048cd9a69d6397126ec60a95d2a438":{"kind":"REWORK","_number":3,"created":"2025-05-20 21:10:02.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/61/950461/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/61/950461/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/61/950461/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/61/950461/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/61/950461/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/61/950461/3"}}},"commit":{"parents":[{"commit":"4f22bade301d6957743e3913c221f3872e39fdfc","subject":"Merge \"Fix redfish driver URL parsing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4f22bade301d6957743e3913c221f3872e39fdfc"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2025-05-20 20:03:10.000000000","tz":-420},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2025-05-20 21:09:54.000000000","tz":-420},"subject":"Science: replace start_neutron_api with start_neutron_service_and_check","message":"Science: replace start_neutron_api with start_neutron_service_and_check\n\nBreaking devstack changes makes pixie boots sad\n\nChange-Id: If47e74751ba66a1296f16d9c43433033c04beffb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/d5b2c012d4048cd9a69d6397126ec60a95d2a438"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/d5b2c012d4048cd9a69d6397126ec60a95d2a438"}]},"branch":"refs/heads/master"},"07dfb04b827bbe55ad87171ee7687c246f2d7521":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-05-20 21:23:29.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/61/950461/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/61/950461/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/61/950461/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/61/950461/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/61/950461/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/61/950461/4"}}},"commit":{"parents":[{"commit":"4f22bade301d6957743e3913c221f3872e39fdfc","subject":"Merge \"Fix redfish driver URL parsing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4f22bade301d6957743e3913c221f3872e39fdfc"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2025-05-20 20:03:10.000000000","tz":-420},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2025-05-20 21:21:33.000000000","tz":-420},"subject":"Remove code which has been long-dead","message":"Remove code which has been long-dead\n\nOur CI broke this morning with an error around start_neutron_api\nfunction not existing. That function was removed from devstack\nin Nov 2022 in a52041cd3f067156e478e355f5712a60e12ce649.\n\nUpon further research, is_service_enabled neutron-api appears to have\nbeen returning false, even for neutron-enabled jobs, for a long time. A\nrecent fix to this behavior in devstack exposed this dead code and the\nlurking breakage which we\u0027ve now experienced.\n\nI\u0027m making the assumption that since our CI has been fine for 2.5 years\nwithout this code block running, it\u0027ll be fine for now too. We likely\nwant to follow up on if missing these calls have a side effect.\n\nChange-Id: If47e74751ba66a1296f16d9c43433033c04beffb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/07dfb04b827bbe55ad87171ee7687c246f2d7521"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/07dfb04b827bbe55ad87171ee7687c246f2d7521"}]},"branch":"refs/heads/master"},"251ce0dcf305f05cc7fad268ed517a661853dd2c":{"kind":"REWORK","_number":5,"created":"2025-05-21 16:47:48.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/61/950461/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/61/950461/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/61/950461/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/61/950461/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/61/950461/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/61/950461/5"}}},"commit":{"parents":[{"commit":"4f22bade301d6957743e3913c221f3872e39fdfc","subject":"Merge \"Fix redfish driver URL parsing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4f22bade301d6957743e3913c221f3872e39fdfc"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2025-05-20 20:03:10.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2025-05-21 16:46:17.000000000","tz":-420},"subject":"ci: Remove code which has been long-dead","message":"ci: Remove code which has been long-dead\n\nOur CI broke this morning with an error around start_neutron_api\nfunction not existing. That function was removed from devstack\nin Nov 2022 in a52041cd3f067156e478e355f5712a60e12ce649.\n\nUpon further research, is_service_enabled neutron-api appears to have\nbeen returning false, even for neutron-enabled jobs, for a long time. A\nrecent fix to this behavior in devstack exposed this dead code and the\nlurking breakage which we\u0027ve now experienced.\n\nI\u0027m making the assumption that since our CI has been fine for 2.5 years\nwithout this code block running, it\u0027ll be fine for now too. We likely\nwant to follow up on if missing these calls have a side effect.\n\nThis change also disables voting for the\nironic-tempest-ovn-uefi-ipxe-ipv6 CI job which is in a similar state\nas the other CI jobs which use networking-generic-switch.\nThis is a side-effect due to a lack of the uwsgi launched neutron having\nthe configuration files to load plugins. That issue is being worked\nseparately and once networking-generic-switch is fixed, the job will\nbe returned to voting status.\n\nChange-Id: If47e74751ba66a1296f16d9c43433033c04beffb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/251ce0dcf305f05cc7fad268ed517a661853dd2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/251ce0dcf305f05cc7fad268ed517a661853dd2c"}]},"branch":"refs/heads/master"},"56d43ff56d900377d379b54e4345ac7cb7f1381b":{"kind":"REWORK","_number":6,"created":"2025-05-21 21:02:25.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/61/950461/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/61/950461/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/61/950461/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/61/950461/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/61/950461/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/61/950461/6"}}},"commit":{"parents":[{"commit":"4f22bade301d6957743e3913c221f3872e39fdfc","subject":"Merge \"Fix redfish driver URL parsing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4f22bade301d6957743e3913c221f3872e39fdfc"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2025-05-20 20:03:10.000000000","tz":-420},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2025-05-21 21:02:25.000000000","tz":0},"subject":"ci: Remove code which has been long-dead","message":"ci: Remove code which has been long-dead\n\nOur CI broke this morning with an error around start_neutron_api\nfunction not existing. That function was removed from devstack\nin Nov 2022 in a52041cd3f067156e478e355f5712a60e12ce649.\n\nUpon further research, is_service_enabled neutron-api appears to have\nbeen returning false, even for neutron-enabled jobs, for a long time. A\nrecent fix to this behavior in devstack exposed this dead code and the\nlurking breakage which we\u0027ve now experienced.\n\nI\u0027m making the assumption that since our CI has been fine for 2.5 years\nwithout this code block running, it\u0027ll be fine for now too. We likely\nwant to follow up on if missing these calls have a side effect.\n\nThis change also disables voting for the\nironic-tempest-ovn-uefi-ipxe-ipv6 CI job which is in a similar state\nas the other CI jobs which use networking-generic-switch.\nThis is a side-effect due to a lack of the uwsgi launched neutron having\nthe configuration files to load plugins. That issue is being worked\nseparately and once networking-generic-switch is fixed, the job will\nbe returned to voting status.\n\nChange-Id: If47e74751ba66a1296f16d9c43433033c04beffb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/56d43ff56d900377d379b54e4345ac7cb7f1381b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/56d43ff56d900377d379b54e4345ac7cb7f1381b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
