)]}'
{"id":"openstack%2Fironic~950540","triplet_id":"openstack%2Fironic~master~Ic6142bcfa7ea2746d704625548edbef3bb57cad1","project":"openstack/ironic","branch":"master","attention_set":{},"removed_from_attention_set":{"36770":{"account":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"last_update":"2025-05-22 11:09:29.000000000","reason":"\u003cGERRIT_ACCOUNT_36770\u003e replied on the change","reason_account":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."}},"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2025-05-22 13:07:49.000000000","reason":"Change was submitted"},"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2025-05-22 13:07:49.000000000","reason":"Change was submitted"},"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2025-05-22 13:07:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic6142bcfa7ea2746d704625548edbef3bb57cad1","subject":"Revert \"Fix redfish driver URL parsing\"","status":"MERGED","created":"2025-05-21 13:37:59.000000000","updated":"2025-05-22 13:08:59.000000000","submitted":"2025-05-22 13:07:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"revert_of":949597,"submission_id":"950540","meta_rev_id":"35d652e5a17def678d6eedb248030a48a56a25f8","_number":950540,"virtual_id_number":950540,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-22 13:07:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"all":[{"value":2,"date":"2025-05-22 11:09:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-05-22 10:12:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2025-05-22 10:58:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2025-05-22 11:36:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}],"reviewers":{"REVIEWER":[{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-21 13:37:59.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"state":"REVIEWER"},{"updated":"2025-05-21 13:37:59.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2025-05-21 13:37:59.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2025-05-21 13:37:59.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-22 10:12:23.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2025-05-22 10:58:39.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"}],"messages":[{"id":"7072a20c2176ae6e31952b091f834b9436a198ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-05-21 13:37:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"602a0a3e38aa9819ae499b82bd9a7c7a8d3df131","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-05-21 13:38:39.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"746135389d7dd0a3bc1f8603f856e7c54077daeb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 14:02:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b99e06ab8ddf46819c7bea0f92e01723\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d7d68208b2d746059c71601de64a04fc : FAILURE in 20m 04s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2d5ea713235c470ea62cbdd7753f333b : SUCCESS in 11m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3a51a50ccb504e9489c0ebc903345f70 : SUCCESS in 10m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"de26c7f08a97ab73ce2ffb37b620fb0c52fd983b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 15:07:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f57bf4fcab24506a69c3c1bb0530cb8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5ce703ddc8f4331845628ced251399c : SUCCESS in 13m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78a32ecb76b0492bb7c1e464707a7399 : SUCCESS in 6m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/708571315ca945c3a9dd7d7d91f3880b : SUCCESS in 10m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c3df633a85804536b626c3dacb5cbbb7 : SUCCESS in 6m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/35dbee805abb4a3db2f578503fed159a : SUCCESS in 11m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f36c3155ead74093906e685712d55046 : SUCCESS in 15m 48s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/2367a01fd48f4cd39faa926e02accf5e : SUCCESS in 5m 49s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/eb688202215345478551610ab9adead1 : SUCCESS in 9m 49s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/be803e664a684a43a076cb9a95324218 : SUCCESS in 9m 03s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/98f4a7adf0db4f9b90f768e083ea98fa : SUCCESS in 38m 51s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/7f4e0efe827d436bba731263a85074ce : SUCCESS in 1h 11m 57s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/eaf0a91fe9b847fcb800529b6a0a3761 : SUCCESS in 37m 35s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/d1e990960fee4f96a7a9736e5f9444b3 : SUCCESS in 38m 39s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/731a1321b6a94aa7b90d662a48a5015b : SUCCESS in 26m 56s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/13cb87bd301d4ba18d2785fb13ff532a : SUCCESS in 38m 38s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/8bac5562a0c84e0e9a91419012b38ac9 : SUCCESS in 37m 18s\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/b5926fc51811410cb0d25b99ced1605c : FAILURE in 26m 13s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/18e098a9970745f594f285c4ab1b6dac : FAILURE in 35m 19s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-shard https://zuul.opendev.org/t/openstack/build/cc0d2282209246419917ed77b346b4ec : FAILURE in 32m 40s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/c5ad92beabaf4a94a9dd7280a827b173 : SUCCESS in 1h 04m 06s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/dacce896ab0144fba11f2965b1cf451b : SUCCESS in 37m 23s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/ddd860182bb0459ba28e2e58bcadfb79 : SUCCESS in 43m 09s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/36790d801f4d447dba0799569f1beb45 : SUCCESS in 1h 04m 02s\n- metalsmith-integration-glance-centos9-uefi https://zuul.opendev.org/t/openstack/build/f04c973fbe654f08b7752d456b26dcd2 : SUCCESS in 1h 24m 33s (non-voting)\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/e34db1c900a94d81b7030e55d1232099 : SUCCESS in 29m 58s\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/59dd70faa9cc4d339be5653b86263011 : SUCCESS in 1h 08m 26s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/aac90e95e1094714aa461bf521c75e3f : SUCCESS in 39m 02s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/d17a54d6bcf84a3f95013de17b22ff66 : SUCCESS in 40m 07s (non-voting)\n- ironic-inspector-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/45aeffaeacd444fcab93d60299979819 : SUCCESS in 39m 46s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/8b776fc8972f4c64b14b4e7a541894bc : SUCCESS in 38m 22s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/8c827272d9a84dc7803f5bc507bf232f : SUCCESS in 43m 45s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/163a3afb1db5462288e7d62d86b8a927 : SUCCESS in 47m 07s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/07891eea2165436487e8a401d9c02437 : SUCCESS in 38m 06s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/b2214bf2da9049c69b4de3956bff2d30 : SUCCESS in 35m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6f38077031c7d76af20ef4658b85b0307ddc4e2e","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-05-22 07:28:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dd945f15ff06aa568a65dd1e07e5811ef4dc3c5d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 07:45:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d37aff061534b0591d0d60d7e8586d0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5e39b0b520e14745ad25a62044f28503 : SUCCESS in 8m 38s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4aaae851cb6b42519142e398c6f4b8ba : SUCCESS in 9m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/00e20ed355cd437384eb8307e5272357 : SUCCESS in 15m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c23346bfd5e1c99be2774698f9774df634a465bc","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-05-22 08:58:11.000000000","message":"Patch Set 1: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f01c6ae8e4a74eed0755f57355ec423c954a0c7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-05-22 08:59:57.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"2b6c4077811874e54276db4dc54c87903d4a6377","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 09:28:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40df646190694155a33bff7d96d81bd4\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ec0ba0b20a894096a7c1ee1d96f89904 : SUCCESS in 15m 10s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2ff06198cb124c48866622fd7065076d : SUCCESS in 9m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/047f980f82424ffc911363c8c5a2e4ee : SUCCESS in 13m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bc407384b8f39c37abc8e192118f332fc6fc0feb","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-22 10:12:23.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1878cdc64c08c23df235d9bba746fc421fb3f15f","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2025-05-22 10:58:39.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c0d8a9c2483b6156e33581543d56385e74e29cc7","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2025-05-22 11:09:29.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"40ba22a55dd2fead9408bda3fd2eaac50de18be1","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2025-05-22 11:36:21.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9d1e0f38bd7ec6d71dff9875de78814e76e48076","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 12:00:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8491db3b989f43a6be9bd37072d45846\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04cb88a3cab94a13899bc50a20f72016 : SUCCESS in 14m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dd3529a14b044f5a740e658ad0a8166 : SUCCESS in 6m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7052bdd810ee4033af382c2c9c2a1923 : SUCCESS in 7m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0d6661d28d0748e6a3a470efe4c69464 : SUCCESS in 7m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c028d074d7a745e78cd11bc5f6f87b04 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/550d7ad2aa42444fa44fa4a4d996df02 : SUCCESS in 14m 09s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/dcee528bab10415b9158936399160778 : SUCCESS in 3m 03s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/11910d134b2f4b838af1d7092e5a5bdb : SUCCESS in 9m 10s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/35fa88a19469462faabc2f61d840ee78 : SUCCESS in 10m 05s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/8ffeee76e4e8489299861e99f7b454f9 : SUCCESS in 34m 59s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/1111aa322b43467a912a87c017cf2151 : SUCCESS in 1h 07m 42s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/14204455056a4950a8aefc977a711f27 : SUCCESS in 38m 33s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/8c9a14f85b884f42b92d83a7e34709d4 : SUCCESS in 39m 44s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/5ea5fbcce3b24a22802a52ea6918a625 : SUCCESS in 26m 58s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/c1d1860876734850aa10f9f54a1f641a : SUCCESS in 45m 13s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/682b7529013647cf85c5a7a16e3e38dc : SUCCESS in 43m 07s\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/e9d4b927c43f4b54ab2b624012447292 : FAILURE in 1h 08m 18s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/c7f36de283ca450f9fcd8fd3653a6dcd : FAILURE in 2h 55m 26s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-shard https://zuul.opendev.org/t/openstack/build/1663ab1e25404b15814d88175a597f6d : FAILURE in 1h 01m 20s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/da55ebd5bbbf48f5a65c180cfb451c6a : SUCCESS in 1h 02m 28s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/27ec351df37e45599cedfe80b85c4acf : SUCCESS in 45m 55s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/d66cd47da0f3414cb195a45b73284825 : SUCCESS in 42m 59s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/463bdaa885e148968524081122a2594a : SUCCESS in 1h 17m 24s\n- metalsmith-integration-glance-centos9-uefi https://zuul.opendev.org/t/openstack/build/e2acb44b62fa47adae22d043f5ecb8e8 : SUCCESS in 1h 26m 09s (non-voting)\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/e8536070c58741b5a34a7c82e6d93402 : SUCCESS in 1h 23m 50s\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/f06f68d42aad4d1e92d4b7f9c74ba7d4 : SUCCESS in 1h 03m 19s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/8d7aceb2d3f14218ba00a8227b576765 : SUCCESS in 51m 25s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/3dfebaec3d7c4740b7cea26ffb64dbc2 : SUCCESS in 44m 03s (non-voting)\n- ironic-inspector-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/364fb36fcc174258a3d67376d4b6b34c : SUCCESS in 39m 01s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/c30d2f93e7ef4836b0d1e4120174254d : SUCCESS in 40m 50s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/1f895bb037094ccba04666db2f8928a7 : SUCCESS in 43m 04s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/de0d1db7a30f4baf8748b60acb608e53 : SUCCESS in 38m 12s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/699f6e99bc17495285965cde5f4ffb74 : SUCCESS in 35m 39s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/fbc77b129f7642c98167abbcdd2d66f0 : SUCCESS in 40m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8c8f023a85f30dc0a52d9e392c7530377028a311","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 12:01:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fcfb7bb54e326079b9eaedb419f0945ed163928f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 13:07:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee7f92aa31124e648d01a04d1c829d75\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3a780635360411086a182cd2149e204 : SUCCESS in 7m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01f207dbee334a7a8c26c6466ea47654 : SUCCESS in 8m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5389a9d8355f4f5ea1f6c6b0c6a4c348 : SUCCESS in 8m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79d8eb6bbb3e46d8928c0701eb0ad216 : SUCCESS in 15m 42s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/6cb6ec22f7da424faa1c9028ac382888 : SUCCESS in 2m 41s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/0d385ede6d5945acad75829a78ac0ee8 : SUCCESS in 9m 56s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/ab9e663675374f299e5b3c78bce4b5f3 : SUCCESS in 39m 35s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/3f37afeab02949ba9dd75a095baee54d : SUCCESS in 1h 04m 55s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/5cec2f4cbe524b47ab74cd50e9b7a403 : SUCCESS in 36m 37s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/bfbcadf7839e4d8f943bd87bb59a0a35 : SUCCESS in 42m 25s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/573ebe5a82ff4191b18f247cc21517cb : SUCCESS in 25m 43s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/534b564b1ffb412b8870b3bd8d5575c4 : SUCCESS in 37m 52s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/7754789f0cc14487ab17746170bc0aef : SUCCESS in 38m 29s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/9312ecf64cd849e485cc96eaa78f68d9 : SUCCESS in 51m 02s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/8af4780e206d47deb9305bcef223e6c8 : SUCCESS in 38m 21s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/9bd9100604e349a4ae36aa9c078b6b23 : SUCCESS in 43m 37s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/ce2c34f9ff4a4a52b34a53ed89202bc7 : SUCCESS in 55m 57s","accounts_in_message":[],"_revision_number":2},{"id":"1717ac99005499906c83a7fcc711be79d999ebb5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 13:07:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"35d652e5a17def678d6eedb248030a48a56a25f8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 13:08:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/baac8e94835144aab1d897c35cd000af\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f40e9f571eb4ad1b1a4ef5085d9a630 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c7ccec9d8b0c8afa99ad8f900dd9128a3c69e5fb","revisions":{"f2f0ceb2346e2a19783c3b6cdb6d721d27b5dabe":{"kind":"REWORK","_number":1,"created":"2025-05-21 13:37:59.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/40/950540/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/40/950540/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/40/950540/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/40/950540/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/40/950540/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/40/950540/1"}}},"commit":{"parents":[{"commit":"5f7c7dcd041e95a7f1283ab12e9d708844fd0974","subject":"Fix redfish driver URL parsing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/5f7c7dcd041e95a7f1283ab12e9d708844fd0974"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-05-21 13:37:59.000000000","tz":0},"committer":{"name":"Gerrit Code Review","email":"review@openstack.org","date":"2025-05-21 13:37:59.000000000","tz":0},"subject":"Revert \"Fix redfish driver URL parsing\"","message":"Revert \"Fix redfish driver URL parsing\"\n\nThis reverts commit 5f7c7dcd041e95a7f1283ab12e9d708844fd0974.\n\nReason for revert: seems to be causing an authentication loop between sushy and sushy-tools \n\nChange-Id: Ic6142bcfa7ea2746d704625548edbef3bb57cad1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/f2f0ceb2346e2a19783c3b6cdb6d721d27b5dabe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/f2f0ceb2346e2a19783c3b6cdb6d721d27b5dabe"}]},"branch":"refs/heads/master"},"c7ccec9d8b0c8afa99ad8f900dd9128a3c69e5fb":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-05-22 08:59:57.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/40/950540/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/40/950540/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/40/950540/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/40/950540/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/40/950540/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/40/950540/2"}}},"commit":{"parents":[{"commit":"56d43ff56d900377d379b54e4345ac7cb7f1381b","subject":"ci: Remove code which has been long-dead","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/56d43ff56d900377d379b54e4345ac7cb7f1381b"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-05-21 13:37:59.000000000","tz":0},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-05-22 08:59:51.000000000","tz":120},"subject":"Revert \"Fix redfish driver URL parsing\"","message":"Revert \"Fix redfish driver URL parsing\"\n\nThis reverts commit 5f7c7dcd041e95a7f1283ab12e9d708844fd0974.\n\nReason for revert: seems to be causing an authentication loop between sushy and sushy-tools \n\nChange-Id: Ic6142bcfa7ea2746d704625548edbef3bb57cad1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/c7ccec9d8b0c8afa99ad8f900dd9128a3c69e5fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/c7ccec9d8b0c8afa99ad8f900dd9128a3c69e5fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
