)]}'
{"id":"openstack%2Fironic~951901","triplet_id":"openstack%2Fironic~master~I504f5e04c46190b09013875c5d99eb1f0298f2a0","project":"openstack/ironic","branch":"master","topic":"jay/RunbookCleaning","attention_set":{},"removed_from_attention_set":{"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2025-07-09 16:50:25.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"I504f5e04c46190b09013875c5d99eb1f0298f2a0","subject":"[docs] Automated cleaning by runbook","status":"MERGED","created":"2025-06-05 19:08:50.000000000","updated":"2025-07-09 16:53:07.000000000","submitted":"2025-07-09 16:50:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951901-jay/RunbookCleaning","meta_rev_id":"8e3188a6825788b88dddf7247dc7738d43827cb1","_number":951901,"virtual_id_number":951901,"owner":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-09 16:50:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"all":[{"value":2,"date":"2025-06-18 16:34:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-07-09 16:13:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-07-09 16:13:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-05 19:49:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-11 15:20:27.000000000","updated_by":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"reviewer":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"state":"REVIEWER"},{"updated":"2025-07-09 16:13:02.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"}],"messages":[{"id":"d0c41d344840bd3052f588db14916360ae025f7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-06-05 19:08:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"022179e43e0868ed4b16ce306705500805b4f081","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-06-05 19:35:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3886062dfb35528057da3ed39d7fc6330fe93ea7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 19:49:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5faf706cd1ed484d9202e129daa09807\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/168b09439208441dad2a43d37cf351be : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90d565566ec64bfc90bfcad49baf1ed4 : SUCCESS in 12m 55s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/09b684d2dad4476c9bfa19e18654769b : SUCCESS in 5m 04s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/d55000a7081742ba9c2f2c78e5cd3e05 : SUCCESS in 8m 02s","accounts_in_message":[],"_revision_number":2},{"id":"7f606ee70390dfd503620afcac24af52eeb0ac90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-06-11 14:49:01.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"409868dc068be9d62cd46948076a65646a1c19b7","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-06-11 14:49:09.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":3},{"id":"6c5be03504651a5ac266476eabc3f0cc88ed411b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-11 15:09:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e0ec11477a6491fb14fa0a9aba734d3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14693154d15d4d27a3bf5883ec51ff8e : SUCCESS in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec4b3f62bee94d178a6de48db3dcf8ac : SUCCESS in 8m 02s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/3ea0f98301a343189db6609cc991da44 : SUCCESS in 5m 20s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/e49b22fb03934a24aafccb06574dd1bc : SUCCESS in 9m 54s","accounts_in_message":[],"_revision_number":3},{"id":"351323ae6ba828e922b3833d11f1d1afa9b5d62c","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2025-06-11 15:20:27.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"af99b5ac81cfe9d7e0273f8bd65b1ef8adc00cc6","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2025-06-11 15:23:35.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"67a5d7fdb8c4993ce5c83bbcdd663bd3e09b6b8e","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2025-06-18 16:34:30.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"fdd329a67f691fc27d270757517c0dacf90c1385","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2025-07-09 16:13:02.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"7886e8a06e439363082cd4a10549ab5389001a65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 16:13:17.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"499739790545d74d3329ed454f50b8050db67cf7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 16:50:25.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af45d3004241472f88ed6a55d0c84114\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d44517b95704a66b8ce8171c66c162c : SUCCESS in 8m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25b0ee1a9e35451b869f32fb6836d805 : SUCCESS in 16m 30s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/91abd7fbaf744d04aebbc096eba247fc : SUCCESS in 5m 01s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/bbb1a41cd8bc4ac5bfca520721077fb6 : SUCCESS in 10m 21s","accounts_in_message":[],"_revision_number":3},{"id":"16ed1d3d43b1fb863a87b531795ace5076f561d0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 16:50:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"8e3188a6825788b88dddf7247dc7738d43827cb1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 16:53:07.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/067165b7bce342eeac8a79f545aef75e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e898b8003b349e6971d69db74f62912 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b551ec5e151032242b3ccf85e16410fa5366da4c","revisions":{"51179fbe143a4feec691fe9ae21d5f5ee14265fc":{"kind":"REWORK","_number":1,"created":"2025-06-05 19:08:50.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/01/951901/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/01/951901/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/01/951901/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/01/951901/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/01/951901/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/01/951901/1"}}},"commit":{"parents":[{"commit":"263ab17a4980ea8510b491f5d17a86c227bfcd46","subject":"Automated cleaning by runbook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/263ab17a4980ea8510b491f5d17a86c227bfcd46"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2025-06-05 19:08:23.000000000","tz":-420},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2025-06-05 19:08:23.000000000","tz":-420},"subject":"[docs] Automated cleaning by runbook","message":"[docs] Automated cleaning by runbook\n\nThis adds documentation for the new automated cleaning by runbook\nfeature.\n\nChange-Id: I504f5e04c46190b09013875c5d99eb1f0298f2a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/51179fbe143a4feec691fe9ae21d5f5ee14265fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/51179fbe143a4feec691fe9ae21d5f5ee14265fc"}]},"branch":"refs/heads/master"},"eb1da76a360b72ddfc29889154008e2bc7ee1f6c":{"kind":"REWORK","_number":2,"created":"2025-06-05 19:35:47.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/01/951901/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/01/951901/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/01/951901/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/01/951901/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/01/951901/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/01/951901/2"}}},"commit":{"parents":[{"commit":"0ee5858294b1f492776d3b0a2f877dbe25735e2b","subject":"Automated cleaning by runbook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/0ee5858294b1f492776d3b0a2f877dbe25735e2b"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2025-06-05 19:08:23.000000000","tz":-420},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2025-06-05 19:20:35.000000000","tz":-420},"subject":"[docs] Automated cleaning by runbook","message":"[docs] Automated cleaning by runbook\n\nThis adds documentation for the new automated cleaning by runbook\nfeature.\n\nLLM tooling created an extremely basic draft which I enhanced.\n\nGenerated-by: claude-code\nChange-Id: I504f5e04c46190b09013875c5d99eb1f0298f2a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/eb1da76a360b72ddfc29889154008e2bc7ee1f6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/eb1da76a360b72ddfc29889154008e2bc7ee1f6c"}]},"branch":"refs/heads/master"},"b551ec5e151032242b3ccf85e16410fa5366da4c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-06-11 14:49:01.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/01/951901/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/01/951901/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/01/951901/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/01/951901/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/01/951901/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/01/951901/3"}}},"commit":{"parents":[{"commit":"febb6e24a0948cddb985f76ecb500aa859505361","subject":"Automated cleaning by runbook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/febb6e24a0948cddb985f76ecb500aa859505361"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2025-06-05 19:08:23.000000000","tz":-420},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2025-06-11 14:49:01.000000000","tz":0},"subject":"[docs] Automated cleaning by runbook","message":"[docs] Automated cleaning by runbook\n\nThis adds documentation for the new automated cleaning by runbook\nfeature.\n\nLLM tooling created an extremely basic draft which I enhanced.\n\nGenerated-by: claude-code\nChange-Id: I504f5e04c46190b09013875c5d99eb1f0298f2a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/b551ec5e151032242b3ccf85e16410fa5366da4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/b551ec5e151032242b3ccf85e16410fa5366da4c"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
