)]}'
{"id":"openstack%2Fironic~955713","triplet_id":"openstack%2Fironic~master~I5100144a330602996c27ed18d2bbde09be6e9571","project":"openstack/ironic","branch":"master","topic":"clif/trait_based_port_scheduling/portgroup_object_category_attribute","attention_set":{},"removed_from_attention_set":{"36770":{"account":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"last_update":"2025-11-05 04:20:31.000000000","reason":"Change was submitted"},"5890":{"account":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"last_update":"2025-11-03 15:40:54.000000000","reason":"\u003cGERRIT_ACCOUNT_5890\u003e replied on the change","reason_account":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}},"14228":{"account":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"last_update":"2025-11-05 04:20:31.000000000","reason":"Change was submitted"},"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2025-11-05 04:20:31.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"I5100144a330602996c27ed18d2bbde09be6e9571","subject":"Add a new \u0027category\u0027 field to the Portgroup object","status":"MERGED","created":"2025-07-23 17:19:11.000000000","updated":"2025-11-05 04:21:54.000000000","submitted":"2025-11-05 04:20:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"955713-clif/trait_based_port_scheduling/portgroup_object_category_attribute","meta_rev_id":"e2ccad6703da18687bf3bc70ec67b1d4b4ea6e64","_number":955713,"virtual_id_number":955713,"owner":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-05 04:20:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":2,"date":"2025-11-05 01:20:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":2,"date":"2025-11-05 01:22:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":1,"date":"2025-11-05 01:22:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-23 17:40:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-23 18:42:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-11 17:45:26.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2025-08-23 15:37:34.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"CC"},{"updated":"2025-09-11 20:24:09.000000000","updated_by":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"reviewer":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"state":"REVIEWER"},{"updated":"2025-11-03 15:40:54.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"},{"updated":"2025-11-05 01:22:41.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"}],"messages":[{"id":"3274c37c477d5df13ab4297f5518f7d772eeae78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2025-07-23 17:19:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d5c5f91554160075e41eb5ed0b0f6226b029839d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-23 17:40:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2f8f22ad7124ded8c08d226333feb7a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/49e173a7dc244e6a9d5c7671e2850dfc : FAILURE in 13m 04s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/407d7d77168445dfb12d4b05bddf5b70 : FAILURE in 17m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e856f11e83d74ce6a058366c54b607b7 : FAILURE in 19m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c187ba8da88aade25c34eeda611b0011655980c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-23 18:42:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8e7a85c638b7430abaa034849d2427bb\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c0f5c2f7ded6410587aa83e2ef76d8be : SUCCESS in 4m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/845d2d077c7643e8bcf92d6b7283f4ed : FAILURE in 7m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a86cfcfd88f48658356a008c7f5b21c : SUCCESS in 7m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6e858cc7b91a4a5c8d5df01259ac10c5 : FAILURE in 6m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7d41820d941a4068a0dff6abf798cac6 : FAILURE in 11m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bcbd4e2abae74a2f87a6f28c1c8cc828 : FAILURE in 10m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69513f519cc445ecbd071722d81ad0ac : SUCCESS in 8m 34s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/700be9df2fa6466e9786eb0e58c5a312 : SUCCESS in 6m 50s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/f90a0849e3e641d7a0f4243642f76bf5 : FAILURE in 7m 02s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/3918b76dda89440098f6eb7a75f23ea5 : FAILURE in 6m 25s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/6d30654806104bdf8789cc29332ba1ee : FAILURE in 47m 12s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/3b190e7a7b1d4bb599b86041c72d417f : FAILURE in 29m 03s (non-voting)\n- ironic-tempest-standalone-advanced https://zuul.opendev.org/t/openstack/build/58e67e05a5b64462bd7eceb7b74e41cf : FAILURE in 1h 01m 43s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/5a23ca01f04d42eabcc457d605264aad : FAILURE in 22m 56s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/6102aa2c0e7a40f2a9224c6861a3b7b1 : FAILURE in 15m 25s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/42dd96aa0a59461196e9b849dcae933e : FAILURE in 26m 08s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/1c45b40a9153417d86463cd950759b58 : FAILURE in 15m 21s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/2f9a6d056a2f498996bcc91090a72df8 : FAILURE in 35m 09s\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/8a90df532eb84fc3bdc77c9097e85c14 : FAILURE in 15m 38s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/dc9730cfa5a644f5bc71f1a591e26b13 : FAILURE in 33m 49s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-shard https://zuul.opendev.org/t/openstack/build/8764374d8094480ea973f0b8e152536b : FAILURE in 1h 17m 26s (non-voting)\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/68043e2b5efe44418f3ac4afde45c8b2 : FAILURE in 1h 00m 46s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/6a096a99387f401cba1706dc2efa5a8d : FAILURE in 38m 13s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/d2b7c01b00c348a1a8b28bc0193c98bc : FAILURE in 30m 29s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/38980b116c1b4fd292d6c357ec2f6f8d : FAILURE in 32m 52s\n- metalsmith-integration-glance-centos9-uefi https://zuul.opendev.org/t/openstack/build/490418b031804d499638136ace5dbef3 : FAILURE in 49m 01s (non-voting)\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/e62ae7aea9cf42dd99ea87fe716e8f3f : FAILURE in 1h 00m 27s\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/ae2f4e33e2a246ea9e02824c20b078b7 : FAILURE in 42m 15s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/f11b77a52c8844d7add2759a18ed1b3b : SUCCESS in 44m 51s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/3b417437d79b4272800c546b49c472e9 : FAILURE in 1h 10m 56s (non-voting)\n- ironic-inspector-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/5426a112d268402484598603d6360a21 : SUCCESS in 31m 36s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/4cf81691de4c4bec9b09328345f86814 : FAILURE in 12m 35s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/41fae5f457694d5db5a2a7e75ef8a614 : FAILURE in 27m 17s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/518a1ab13d6240059e9bf42563eb5398 : FAILURE in 35m 54s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b6d33092b23d414d8787585e5737ea4d : SUCCESS in 40m 20s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/2bcb2af0788f4a7eb5ee90f3c3ad362a : SUCCESS in 44m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fad77dd42ec88cc4f0f020d6794a09f0d4196dcc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2025-07-23 21:20:33.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7390715b66c144fc8a29da0ba1bcfeb419ee99b9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-23 21:45:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27c48698a18c496aa6f4081cd192ada4\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9eb517153e1e45b6a3558dd6e17b97e8 : SUCCESS in 22m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2d3072f1100445ad9b4b212360a35589 : SUCCESS in 18m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f026ac879bb647e79c292ac3dce2aa1c : FAILURE in 14m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"808870aaaba29a4f5ba1d7b693b7e0a23daa4366","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-23 22:55:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d68eaba52df94207832f1d2d61fd80bc\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f44b1ffc6b1341c5b95b30df0ad7061b : SUCCESS in 4m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b043c8ef6b024bcf92f75ec4bb4ac134 : SUCCESS in 15m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39975b9b062d46fcb14921d14bbefd69 : SUCCESS in 7m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e8248c4c236346bdb307788d6f5a74d6 : SUCCESS in 8m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a520b207bb64446baee1d4cdda222158 : SUCCESS in 9m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/80db42c574ed4f0ca76d326cbb59f450 : SUCCESS in 12m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efef54a813ab4ecea314176a2f1f0cdc : SUCCESS in 15m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5794304f33244a36b93128d26928a80e : SUCCESS in 7m 09s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/a10f6be420a942928b31279f4a21d451 : SUCCESS in 3m 05s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/a513b4f7a80b41b799202cda980b8fde : SUCCESS in 10m 58s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/7a1ef8721ac14d24969d8c3dd6b0dd62 : SUCCESS in 8m 51s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/034fcca6ceab443d80fda49be7c1fa93 : FAILURE in 47m 29s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/5c5c1c0bb61e430f94fd1e88c087a0bf : SUCCESS in 1h 01m 46s (non-voting)\n- ironic-tempest-standalone-advanced https://zuul.opendev.org/t/openstack/build/7cf5f6f87ecd46a1a869751332ab57da : SUCCESS in 1h 26m 30s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/13f2a4af991849b5b78da66370553f43 : SUCCESS in 43m 43s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/40e1003044854e0fa19e591bf068a66f : SUCCESS in 53m 33s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/5df46b31f0c1478a8cf8cbf08d7fff60 : FAILURE in 26m 33s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/b33e58c4286e48249a7b50ebafe362f8 : SUCCESS in 46m 12s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/113f3760572e4c2ea6c65cd3463e7b55 : SUCCESS in 45m 15s\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/9532293d6b544ea68db1ec7c51659c3a : SUCCESS in 1h 01m 29s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/ef2a85e756a44de0a2e4a3c62249c1c0 : FAILURE in 57m 29s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-shard https://zuul.opendev.org/t/openstack/build/b5194de781e24bdc948ea917a202e902 : SUCCESS in 1h 11m 09s (non-voting)\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/65773969981f45e5a242a3db9e44efc7 : SUCCESS in 1h 05m 07s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/1263b5626bd0449bb6cd6b5a99e85903 : SUCCESS in 38m 17s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/164a4f4e6e444533a8f0fcca13d39286 : SUCCESS in 45m 50s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/812af293e0e44f76a4764ba844e646fa : SUCCESS in 1h 07m 56s\n- metalsmith-integration-glance-centos9-uefi https://zuul.opendev.org/t/openstack/build/d208f29eeb4a4b08bc11a2d27c5417b6 : SUCCESS in 1h 28m 25s (non-voting)\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/13cfefebd98744bd831bd71780920fec : FAILURE in 56m 14s\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/edb7dd897a7648bdab74ff025d1efa42 : FAILURE in 57m 07s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/8e9be67e9ff647abb69fb883db0fc7bf : SUCCESS in 1h 19m 41s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/d9367c54523e4086a189d51c29562bec : SUCCESS in 42m 32s (non-voting)\n- ironic-inspector-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/f2af7f7336a247d69d6fec1836d0ca55 : FAILURE in 53m 55s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/ee26c0c26f53476ab2a65bb29046bc02 : SUCCESS in 34m 16s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/b8533f8b053d46b2a2faad81fe29eb01 : SUCCESS in 40m 58s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/de3fef4e815a471facfa4aead6d866aa : SUCCESS in 53m 43s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5af794370cd948a39ae8fc7f4cac0a30 : SUCCESS in 41m 41s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/439057cfb71644e58e20776e17bf8bb5 : SUCCESS in 45m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c3cf01df2d8aed08804f11fca30eac096a0dbc52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2025-07-24 15:16:40.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"aaa10bcd7f75b80d3482530b797a4d9388eb0604","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-24 15:47:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1ef47ab901f46f8a6e7917cb4cf3a77\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1181e996ba134899b65d5d7bdb1197f3 : SUCCESS in 14m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/349e346518a14ef8b078c5ecb615729c : FAILURE in 12m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9cd58defc5ff48d68ef67e54772821be : SUCCESS in 21m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f4deb5e3bb7c507f649866ae067a7a7e4eb345a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-24 18:20:30.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5bd590a6e4264d14bebcc6eb6de5ce29\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9333cba035914bf08f48b97eb3144200 : SUCCESS in 3m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/90bd2fa10fca46cf881b776b06e8c9fc : SUCCESS in 8m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10707ca0a7514a6ab51b427a3e34d653 : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5fb35faf46334443b70811652efdfd59 : SUCCESS in 5m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/48bc7c88a53044efa8ff7c997416c0b6 : SUCCESS in 6m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/591f1b589c244568a02e0fa09096b6cf : SUCCESS in 14m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f89d26c461d41568094edfac6cb1d58 : SUCCESS in 13m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63e8bcbd802f42f7903baef3d2793e7b : SUCCESS in 17m 17s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/c29fed99e7bb4214b1f06f25bf09dee6 : SUCCESS in 3m 50s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/d1ceee7d47d148ce949e965d1374d378 : SUCCESS in 10m 41s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/4c27379de63041b18385b0d633120c43 : SUCCESS in 9m 45s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/8e4abc8832d740799b8f3335ae52e26a : SUCCESS in 44m 33s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/a04b0afd39f14fa99273037a00374bd8 : FAILURE in 54m 47s (non-voting)\n- ironic-tempest-standalone-advanced https://zuul.opendev.org/t/openstack/build/16a7e0e0f7c143a58dcd5755af5b9dc3 : SUCCESS in 39m 17s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/40805c11fa174c038a8ca6e7142b4fa3 : SUCCESS in 22m 28s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/4fe6b6dada3447c8bd2c17174f254a75 : SUCCESS in 36m 27s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/d418e3383ff242cfab65948fea5d778d : FAILURE in 17m 01s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/bd663837c1d24f42abbc9891da3dd660 : SUCCESS in 49m 55s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/63341abd419846fdb636ae9ecbb72bf2 : SUCCESS in 23m 29s\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/a22c11c08e614c8999621e749765b684 : SUCCESS in 1h 03m 20s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/b159da72ce6640e1b758dbc62cee1c9e : FAILURE in 2h 50m 05s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-shard https://zuul.opendev.org/t/openstack/build/be0b554b5afa49418cbed795f017bcbe : FAILURE in 48m 40s (non-voting)\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/88afc385559f4ba0970af02c08538039 : SUCCESS in 33m 20s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/d4ee0afba73b4741b27207997cd4b979 : SUCCESS in 37m 11s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/f10a11edbc1d4fb5afe317ac9a06010a : SUCCESS in 31m 32s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/10f827c3812a4c28b34c219e08608a16 : SUCCESS in 44m 43s\n- metalsmith-integration-glance-centos9-uefi https://zuul.opendev.org/t/openstack/build/c4310b6011d449aa90126978ab667aa0 : SUCCESS in 1h 15m 15s (non-voting)\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/863b3057b3904157a5ac58524959f89e : FAILURE in 55m 14s\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/4e8e7cee8c86488bae48adab06159f93 : FAILURE in 56m 14s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/0623cf799c9e4b13bce97cdd08d5df0d : TIMED_OUT in 2h 32m 40s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/6cf10cfd6e4448f39dc255ab5697a5d4 : SUCCESS in 33m 57s (non-voting)\n- ironic-inspector-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/ae3b9efed4354cd8a31bff6bd0ad7d30 : SUCCESS in 35m 06s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/83c1c51de656416896c81103702025f0 : SUCCESS in 20m 16s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/5b234e810de74ae7acf08fe0dd4c3f5f : SUCCESS in 35m 45s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/6695b170ebbd4e65ba3293f75f1b711c : SUCCESS in 39m 32s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ebd43b31e3744808283907b17c5f8a7 : SUCCESS in 23m 41s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/4eef616bbe90475ebe31280af6fd5866 : SUCCESS in 42m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6ba9dd2c77ec2d39c05feb7d4c9be3e1aa9020e0","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-08-11 17:45:26.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"7f1661bdb3fd83a717cb109ff602038ff8174e3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2025-08-13 16:41:52.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"2a9e53269c1123211b91671aa4dfcb99d77e4615","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2025-08-13 16:42:22.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"55f65b061b9376f87187643f4530be534d3153c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 16:43:40.000000000","message":"Patch Set 4:\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/955625, https://review.opendev.org/955799 are needed.","accounts_in_message":[],"_revision_number":4},{"id":"2accc6ffc919cc608bbb2dc70ffd4a248b282d59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 16:43:54.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/955625, https://review.opendev.org/955799 are needed.","accounts_in_message":[],"_revision_number":4},{"id":"3c9aea3f0b8da03f8d9df4ccd19fa937b15f0bb0","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2025-08-14 12:25:36.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f5f736be67ef6c57fbf5b423bbecb558428e07b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-14 13:16:16.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7a54d0349814abcad80f97c1539ff41\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e36f42a0406d44f998be29ef6a01ed62 : SUCCESS in 18m 51s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/37b6a188505649168c33221feceefd4b : FAILURE in 18m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/34196ddaaba145e7b2702a3d3419170a : FAILURE in 25m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"85ff99f0a05e03878fe7b32ec60365e8024b8f4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 05:54:24.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b3d633d04ca94f25a07fcec980286c65\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/719162ea44f9470d9bbe82ab2fa3477b : SUCCESS in 3m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1967c16c15404033981719e18c030208 : SUCCESS in 7m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90e429875bed4e02a08bec413eb5d530 : FAILURE in 6m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5655cc19fa5b4db49995021c281d6864 : SUCCESS in 5m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e1ed7df88a774560b77fea9ee195fa1a : SUCCESS in 8m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/35e5924440b94abbbfc49571a9386fc2 : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f083ddfdd6e24e919c11b73f6d9eaf01 : SUCCESS in 7m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c16a0ab8124d41eba13b24a496ef548e : SUCCESS in 12m 38s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/8370fc0bd2e648328d838df012174241 : SUCCESS in 2m 53s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/0715dc162820447ca90053d8d06a7419 : SUCCESS in 5m 29s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/3a3783420db04ae4acdf429f83e17063 : SUCCESS in 6m 42s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/62907dfe982c41d1be79a07e4eb51ccd : SUCCESS in 34m 44s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/8103a8c194a84ea3a653b4094dbac54d : SUCCESS in 1h 01m 56s (non-voting)\n- ironic-tempest-standalone-advanced https://zuul.opendev.org/t/openstack/build/1b74b4188c5149309c89e77bb31a56d2 : SUCCESS in 1h 09m 25s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/6681d1c328fc427f9291bc331a876d6c : SUCCESS in 23m 55s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/fecc70acb3e747178d6da9c08a24efa6 : SUCCESS in 19m 34s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/219c704e9b0646fe88d0cbbbf5496f56 : FAILURE in 11m 16s (non-voting)\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/ab65f021ad25448bb9bf5274cd798363 : SUCCESS in 19m 02s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/7f0998445064425f96bdc4d7174c1925 : SUCCESS in 36m 34s\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/c54fde64e5cd4c988c8139c9921638f2 : SUCCESS in 50m 48s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/62d55b63a70f43d298a9393926c21603 : SUCCESS in 2h 05m 49s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-shard https://zuul.opendev.org/t/openstack/build/aae0a81a15fc4b46ad40d3ab83247c05 : SUCCESS in 1h 22m 11s (non-voting)\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/724d7887aeaa4c3dab3c3544b2946412 : SUCCESS in 29m 05s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/879e78f53ff14588928d157ee0a3b0f5 : SUCCESS in 19m 51s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/e6a21b8102aa44258369935fdedc8ccc : SUCCESS in 32m 54s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/8b959e7d98674e0ba6adbba1b490f390 : SUCCESS in 28m 03s\n- metalsmith-integration-glance-centos9-uefi https://zuul.opendev.org/t/openstack/build/3b4cf5ce1b074d5e8b999489632a511c : SUCCESS in 39m 17s (non-voting)\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/bac0369e942645049038deb3ae759447 : FAILURE in 31m 07s (non-voting)\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/b45f602bb11e4555b51b2f86b9f132a4 : FAILURE in 29m 35s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/f1e95892e09e4b5fa3f56c1b63028673 : SUCCESS in 32m 12s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/aca7185112974554b486d8ebc142b988 : SUCCESS in 20m 32s (non-voting)\n- ironic-inspector-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/ad162cb08f124689ace9e2ac8ba09ecb : SUCCESS in 29m 05s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/35fd6c750d194195a31487a2d059a5f6 : SUCCESS in 21m 16s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/bb343fdd54484d838d8fdda15ee895e3 : SUCCESS in 36m 53s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/6a893ad0b07d41a5803f02da45127cff : SUCCESS in 33m 13s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/573fe60c148c456195a533bb86e6bdff : SUCCESS in 19m 59s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/35760e2b0a6f49799868de159d48daa1 : FAILURE in 6m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5234d27544f10725922963fe8447158a7acad3d0","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-08-15 18:28:05.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"ffbf2db5ec09cc4641441c3ce4044b4604db84ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2025-08-15 18:41:20.000000000","message":"Uploaded patch set 5: New patch set was added with same tree, parent tree, and commit message as Patch Set 4.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"564e4a6c17106d48ba7287384d1d8db20a76b7d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2025-08-15 18:44:27.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":6},{"id":"f2c85f7e16792ce1140e1bb444b9ddc939acbf5a","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2025-08-15 18:44:48.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"042dc1b90b4e2a472df5a6f6d5cef871b6677153","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 19:51:33.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/187ec3ff5793428fb6481fe1727c3384\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b06cbed7541a46e8b00f3d79e095c9fd : SUCCESS in 15m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e5fd0c0cb8584e7cb9e0e37a652f689f : SUCCESS in 19m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bb5c3c9d32204e699c91c3c3276bf51c : SUCCESS in 24m 54s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b221b7f6431f75e35046a870c903d6e023ae099a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 21:02:53.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/838b46fc9de14de981d18adfab0f173e\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7e4a2242248749708b92e52e4ae87900 : SUCCESS in 7m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b14e06f637c40b3a6679d4d6aafa341 : SUCCESS in 13m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57e0add9bc534d588662c2912cadf31b : SUCCESS in 9m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2f9efe76b4914dff98a5cf079bfd26af : SUCCESS in 5m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fdaf460f9f8949948b27ad2a3ad98ab5 : SUCCESS in 9m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c7aa76080b8a45fdb8c7dfb696d71633 : SUCCESS in 12m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edb0e118c2ff400cb48eafcb502c9fd0 : SUCCESS in 8m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/09c8c666623445b59b85a73f9373e07b : SUCCESS in 9m 11s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/76b480fa288e4c59a03f69fd5696ed60 : SUCCESS in 3m 26s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/8237b59fe21246c0ad5ca8a9bda9fa7c : SUCCESS in 6m 22s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/7e8b1d0c92b240698110c990c549827d : SUCCESS in 5m 17s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/fd6aaacc025e425c9934b02780172dea : FAILURE in 51m 19s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/a747ee49e42142baa1df9bbbfe48b67a : SUCCESS in 55m 49s (non-voting)\n- ironic-tempest-standalone-advanced https://zuul.opendev.org/t/openstack/build/d978264c65b24026b283c05d17fab334 : SUCCESS in 1h 14m 44s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/bda6d8a0efac49ceb599412810fa6bdf : SUCCESS in 21m 34s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/269627fb6fff44c2b246917233612879 : SUCCESS in 23m 36s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/9882fc6d25fa4cdb9c101abe88b33b59 : FAILURE in 25m 32s (non-voting)\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/2e80800b34b648dd9dc4a50c15ebc37a : SUCCESS in 44m 19s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/010faa6314b744d18a5264b308a07333 : SUCCESS in 43m 07s\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/01b2361400304979987ac04932733aa9 : SUCCESS in 44m 02s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/4bc01df6d9f148ebb2775785bd050658 : SUCCESS in 2h 14m 18s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-shard https://zuul.opendev.org/t/openstack/build/17739391ad554c998a38e29c61b55916 : SUCCESS in 46m 52s (non-voting)\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/e3e2da94acac42c28abd09e8659fb595 : SUCCESS in 27m 47s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/e2a8dfe299e24df7bcfbb4ce7e3c672a : SUCCESS in 18m 25s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/04eebe0001f74871adedf91a8b953e23 : SUCCESS in 24m 16s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/3d5e038f002b4c09a654a1671559b7c0 : SUCCESS in 31m 36s\n- metalsmith-integration-glance-centos9-uefi https://zuul.opendev.org/t/openstack/build/5750474ffc4240d2ab341c3c22681204 : SUCCESS in 1h 17m 17s (non-voting)\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/a783812622c64826ad6b9ae4af86eb3f : SUCCESS in 38m 10s (non-voting)\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/94ac58fa365a46f98a2230db355c8369 : SUCCESS in 57m 13s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/151749bc7f1740b5878c88acf5fed1ab : SUCCESS in 33m 21s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/c9ebf8f07d434fbd8209cbabae0193c5 : SUCCESS in 20m 36s (non-voting)\n- ironic-inspector-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/8607489652c44ab78bd122e0b520ac26 : SUCCESS in 34m 09s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/4c3b3d1224214374b021d031c8bbc93c : SUCCESS in 28m 13s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/6ea2d53850fe48d282f23c822ecfba50 : SUCCESS in 24m 34s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/d8dd50c8b7b8450ba14fe52871fd475d : SUCCESS in 33m 59s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/32c68729b722480ab87b72c38d13c72a : SUCCESS in 21m 07s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/dcbff91a49254feeb6a5a90263e87449 : FAILURE in 8m 15s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"0f8e32e3bd2f392a5d933e20631d112d8e4642f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2025-08-18 18:52:31.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"baf8a826fa5ae6f3adc7b6822fa119c5291e3e51","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 19:13:44.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54e9437d3c344615845b25a5fad8abfe\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c83f4a0801fe47c5b860bddd3b174648 : FAILURE in 6m 47s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/dbce219ea7214c0a8df79bdfaacc0a4a : FAILURE in 9m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0e17fc9c31ca41f48b54099c56fb7d8d : FAILURE in 8m 14s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"156bda87cafddb4ab6eb0ba59fd06afeb1b9f095","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 20:55:11.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18f246dc239e414abd1878d0caefbd69\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3262280dc13a48f78c39b7b757e6a367 : SUCCESS in 4m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7aca99e4fe9a41cfb0e58e8c7f17aaa6 : SUCCESS in 11m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc68dea7d1174460957289e1db963c40 : SUCCESS in 7m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/28609fe03b5b43a489d348cbc3c568f6 : SUCCESS in 9m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/43848a16c987422c9f0823f92097e2c4 : SUCCESS in 8m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7b516e2d5abc4f0f8fe750101f2816ad : SUCCESS in 9m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cb90b642c4749cda95d93264efcbd6f : SUCCESS in 9m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/235704fef75244cca2c851e1c21982d9 : SUCCESS in 15m 18s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/0c7207034f6b4833ac035226ab4dd94e : SUCCESS in 5m 32s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/d12cf02d7ff344668e89314efc8970ca : SUCCESS in 9m 59s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/0067a7c119ab4e12a8c81b764cd722d5 : SUCCESS in 9m 12s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/36a7ca09222f4eb19ec9899b081fda08 : SUCCESS in 30m 51s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/6818dab6803b4351ba055b4fe1d72b40 : SUCCESS in 1h 02m 03s (non-voting)\n- ironic-tempest-standalone-advanced https://zuul.opendev.org/t/openstack/build/3265c9726ac64463939e4481a103d91a : SUCCESS in 40m 29s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/1fa95fea4f9c49e2b5f5c55901ae7bf1 : SUCCESS in 33m 21s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/e53f868cd75c49e9a6f1e2ba43de58ef : SUCCESS in 34m 28s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/b42f45d835384e32a05b17f98d30348c : FAILURE in 20m 30s (non-voting)\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/10ae944668144c3e93c74b1f3e9eef5b : SUCCESS in 21m 41s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/419ea0cbdb07429fa5ce44c2075606cc : SUCCESS in 21m 48s\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/0025f25b16d343c79c0f30c49924e991 : SUCCESS in 48m 47s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/c2d6df9b5c1e4a45b3c148d4b800ed36 : SUCCESS in 1h 05m 55s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-shard https://zuul.opendev.org/t/openstack/build/a6a1ca2d3e0749db9cd423723c1062c8 : SUCCESS in 1h 20m 08s (non-voting)\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/a5f8a7d3fc234a9796ce61432750fec6 : SUCCESS in 32m 59s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/00a86e03e23c40a4ac7e99826517eaf8 : SUCCESS in 33m 42s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/69a0b2e9d1ea4ed3a6103c6cf8274309 : SUCCESS in 39m 02s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/019872977800402db43d2afb5a897dd3 : SUCCESS in 1h 06m 53s\n- metalsmith-integration-glance-centos9-uefi https://zuul.opendev.org/t/openstack/build/8e84cba4e45a45eaa0c5f328cc1d3de1 : SUCCESS in 1h 09m 33s (non-voting)\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/0d15a7ae1ab449eabbe39f04df056954 : FAILURE in 57m 02s (non-voting)\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/0adf9ef4e08247378d89d597adc7d6fe : SUCCESS in 1h 15m 19s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/1ba08ccf91344e609fcdbec21b9afaa5 : SUCCESS in 35m 38s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/bc5137c3ab9a4c9ba2b209c113f56776 : SUCCESS in 35m 22s (non-voting)\n- ironic-inspector-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/7da0f48a94f24b4c9f0091458e2d8fe5 : SUCCESS in 22m 38s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/b7a707894d50415bbbf95b90f73e2f5f : SUCCESS in 35m 51s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/482607f1cb724a949ca0873b7b434cbf : SUCCESS in 36m 48s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/7c87b315d97d465c8f57219d90feaa7b : SUCCESS in 48m 27s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d141f614bae4c8eb9c9fef9bfa8aeb2 : SUCCESS in 22m 52s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/640ae90f8a8b4559be0432b631fc2a60 : FAILURE in 13m 22s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"87d10ef7912fd699f53681b4ce1591fb9a61036d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2025-08-19 14:12:46.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"87faf821b72e0b12adf4738958c6a9f922f5249c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 14:43:46.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4f1d3170bee441d88aa75198289e2ee\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/267f7d1c156a4c1995ca69d427ffff1d : FAILURE in 9m 56s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2ea1f59b21a24369a7de72c6c97fa02e : FAILURE in 12m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3be33bfa51c1414e833604ef082c844f : FAILURE in 15m 50s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"2cfecf2ef646cf7a0b4cf30794312dd2e18fb6a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 16:44:27.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d8f2d773f8fa43b596920577f06b252d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9de1ab87cf23433a81b0c9e68e4a7147 : SUCCESS in 4m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4bd7899aed6149c98e773e6265f93373 : SUCCESS in 11m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f24507f2a0b345778471ceb1db6259ef : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ce6bce6188f14dc19ca327def35f8e52 : SUCCESS in 10m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/75d8bf1f6b8045f9b593f7b53ec11075 : SUCCESS in 9m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b7fa2ddda4aa4741a152cdf481c83076 : SUCCESS in 10m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/068d11c940224b29b83885a310a8e800 : SUCCESS in 10m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63f35730d62c424f8f37a886c6ca4ea8 : SUCCESS in 13m 59s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/83a8566ac69c4d128d425432652df253 : SUCCESS in 3m 07s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/8e7303def86746a9bb4dfb3bcc601c3d : SUCCESS in 12m 06s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/667a2f18cd2c495fa38084f3c761c345 : SUCCESS in 6m 20s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/0356b47739604f13a8fd15f64b741f24 : SUCCESS in 33m 39s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/a79a65e5e7344646b320e324437a825d : FAILURE in 19m 17s (non-voting)\n- ironic-tempest-standalone-advanced https://zuul.opendev.org/t/openstack/build/42baa76fe4b74ce584cec2497f9327e8 : FAILURE in 33m 15s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/8075708671044007a95b1dcd11997a18 : SUCCESS in 23m 59s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/df002b874e9b4028ac1a7f29b2fd4a62 : SUCCESS in 18m 54s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/16734b012d8c40388b94ff13c3bdda38 : FAILURE in 11m 46s (non-voting)\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/054cc37d41c64bb2b2de3421e8ec0d6d : SUCCESS in 39m 14s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/5251d041eb904b999282269c4c07b5cb : SUCCESS in 18m 54s\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/70faeed16e474d0a9b346d4686b6a043 : SUCCESS in 27m 46s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/0395e81cb7274397b0df97376f8b018a : FAILURE in 1h 11m 58s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-shard https://zuul.opendev.org/t/openstack/build/8fd104de0c3344b8ad69ec6bf0f7b5e1 : FAILURE in 1h 16m 16s (non-voting)\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/68e30ec5c94a4499a9884e17b075037c : SUCCESS in 36m 32s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/86aa0fb5472446ecae14c0cd93693efe : SUCCESS in 41m 46s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/f466d007896b494d903598c6dba12dc1 : SUCCESS in 19m 50s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/a827f19c81bb49baac21aa0777854d3a : FAILURE in 13m 03s\n- metalsmith-integration-glance-centos9-uefi https://zuul.opendev.org/t/openstack/build/cf2f327bda914176b424e8b21724d51a : FAILURE in 29m 07s (non-voting)\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/c8074000e131442dad66485d8c91fd51 : SUCCESS in 38m 50s (non-voting)\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/e885f89a53e34fe993138b5c6304b394 : FAILURE in 31m 34s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/ef061815314e4c30bc640e488c9d2c16 : SUCCESS in 36m 01s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/58a68d934b7b4a6d95bf9dd9e640a652 : SUCCESS in 46m 30s (non-voting)\n- ironic-inspector-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/c150d1f8a7cc469783d709da0cb1b4be : SUCCESS in 19m 47s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/4ec1f556584e459aa80ce29cd2c85c7b : SUCCESS in 32m 08s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/3f3f5a6564de4fc3ba3158c1547eb118 : SUCCESS in 19m 14s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/b70014d8b1a14df187c8bbb51cfaa185 : FAILURE in 17m 54s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/757a89c29ec34c13b884aa9ba5c5799c : SUCCESS in 20m 53s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/5f493f01b92b468e9fc0afa76fbca153 : FAILURE in 6m 10s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"6b12e832f2c3c420a75283a3ee9eb01ca9f12975","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2025-08-21 13:21:44.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"7f33d73256c36bf977b15145e89d9ea5c8446f41","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 14:30:28.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bce7a021b7dd4a309627a6867fd055a1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8d881311561f47f79e9a079b1fd2bb6a : FAILURE in 10m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/82bd355906a14e649559d1f6d46b27ef : FAILURE in 13m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/94cc5746fdbd442fb400f34432adeb6d : FAILURE in 14m 55s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"27acf86af90204bd9753f9a699befb045eae354e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 15:24:43.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7c874d8619447a4b0ef4da0e65c62b0\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/062da5ceb501453fb4e616b223444fce : SUCCESS in 6m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f93f24b45254f74a8361412572a87fe : SUCCESS in 15m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e00c4fb6675043569502ac86edef6745 : SUCCESS in 6m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bd26ecd0493d4311a2d4fa18e1d90958 : SUCCESS in 6m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a2ae9f0ea3f4458cbc5c70012030dfa6 : SUCCESS in 11m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/485d9ee05c5f4dcc95687abded469629 : SUCCESS in 13m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30c378b792aa4e22a0a950049341a208 : SUCCESS in 17m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f153204cb0294bc49e6053ad2fd35146 : SUCCESS in 16m 17s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/a89d0fbc85b944aa956b959491458b07 : SUCCESS in 7m 00s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/b011ed12cf8140749d5b248c26dcc967 : SUCCESS in 11m 10s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/4f9f700c23d94b8c87186e4ff1916483 : SUCCESS in 11m 07s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/3d621668caac452ab7e99576ab4622e4 : SUCCESS in 35m 31s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/c60993bcba894ba088f374ca23f6c2aa : FAILURE in 18m 15s (non-voting)\n- ironic-tempest-standalone-advanced https://zuul.opendev.org/t/openstack/build/60c2050a7127467c916d4d887a6dcd6d : FAILURE in 24m 18s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/1261be30bf854d88a61d94e3cbb14db4 : SUCCESS in 36m 43s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/15e6197e97144cca9fb8f390136d0062 : SUCCESS in 38m 47s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/ea875b1d177c4027886369d41f95979f : POST_FAILURE in 7m 31s (non-voting)\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/4e695f0d55804204a78634a16b693114 : SUCCESS in 18m 13s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/d7dea2db3c0a4f0980713f48c41e4e66 : SUCCESS in 46m 54s\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/9163a6d76a4b435ca65a14bb9d367901 : SUCCESS in 1h 06m 33s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/e325d4be9b954805bfbd24408f4297c6 : FAILURE in 1h 59m 01s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-shard https://zuul.opendev.org/t/openstack/build/46ea1d6746be4c0ca32163f07089465e : FAILURE in 47m 43s (non-voting)\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/9ef46b82719e47c68bb451edc442ee5a : SUCCESS in 46m 52s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/186d5ed5a5c64532a4a4103c2ad71df6 : SUCCESS in 19m 04s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/2ce9bf32d68547d4aab8c913eff73591 : SUCCESS in 33m 56s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/d567be9c067a403ea6ad686efd0f0184 : FAILURE in 21m 25s (non-voting)\n- metalsmith-integration-glance-centos9-uefi https://zuul.opendev.org/t/openstack/build/5b73f54a60dc48ef8edda833ac7f4ec6 : FAILURE in 40m 56s (non-voting)\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/5a94ff5a5b04433a813e12f7ea49edda : SUCCESS in 1h 14m 22s (non-voting)\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/2307c682c7464f5f9cbcd2a995a91ec6 : FAILURE in 1h 04m 31s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/2fedee5013644eedaf9d17efee722b32 : SUCCESS in 39m 39s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/7fa9bcde3d824d479e780ea861fb8a7a : SUCCESS in 50m 46s (non-voting)\n- ironic-inspector-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/e6b7953daeae4da8ae6e8e5999dafe9d : SUCCESS in 17m 20s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/e921fd074dac423bbdd7a48c5af0c9a7 : SUCCESS in 17m 51s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/75d32cd5f51349ba8039b1a07de74aa1 : SUCCESS in 44m 11s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/12720740e2634d21b672ddb2b54cd276 : FAILURE in 26m 30s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/59f0d9af1cc442b293853df4be603589 : SUCCESS in 42m 11s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/5f02696c0d704bcd8a4eeedceb286106 : FAILURE in 12m 01s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"38269cdfe7c61e4a3ef5dd245283416300dfe1b5","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2025-08-21 21:29:18.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":8},{"id":"d33fca0bfe69d6c0c8cfaaf475ba6da5d346f496","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-08-22 17:03:16.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"0105f33e8772cd5be9884dff4aaf9fc433d66fe2","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2025-08-23 15:37:34.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"147dd7238c30e610044aa1e1381a54b47536042f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2025-08-25 15:00:13.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"be28a5c932efbfcd9765f7debb66c5aabbf5a480","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-25 15:36:35.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50ceda35bacb4e33a979b7a94c009230\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2f696a0645a441a79849652283ada5f9 : FAILURE in 8m 07s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7a69b024a2ad45f9a0bbaa5fcd70143f : FAILURE in 7m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/50eb22143802431dbd9a51e8833cf880 : FAILURE in 15m 20s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"6e991e9ae721bd35d0a8b73b7c1850689dbdb1ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-25 16:17:37.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec037eaad90c4034b909ea5ae24f02f2\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3c4fd3658768436686d329e91f6fc48a : SUCCESS in 6m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee079473909d41dd96e6d4bdec7925a7 : SUCCESS in 14m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7788b1fc2bde4b36b1d92afd0b8b4935 : SUCCESS in 6m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/793e1957b27344669d9309068228b415 : SUCCESS in 9m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ccf6909699dc4f0691fe5ce14bd072c6 : SUCCESS in 9m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8ec6d0198b7742a0a1e5bfe395cffafd : SUCCESS in 12m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96b017f8230348888db0507206da697a : SUCCESS in 9m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2578df555c8640739ecf1b2266ba25e0 : SUCCESS in 15m 25s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/9834f8b8acc94bc0a3502942d33bfe0b : SUCCESS in 5m 27s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/6f154853bff14bd2bcfd34069af936b6 : SUCCESS in 8m 00s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/01c8a070440d411fb7cfbb7229e91cb3 : SUCCESS in 7m 26s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/98140c4ff021482d9f8bd795ca1cf0f8 : SUCCESS in 40m 00s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/78df2205da7b46a78ef0f05f7bec6026 : FAILURE in 18m 11s (non-voting)\n- ironic-tempest-standalone-advanced https://zuul.opendev.org/t/openstack/build/355e896da794482bbbe40da5b4f35400 : FAILURE in 26m 40s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/d771300a4cee4390bec8ed38b5cf08a1 : SUCCESS in 31m 12s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/76f013f7e7bd4702858fe58c3ee01643 : SUCCESS in 44m 10s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/a8c2604a38e741fe8b035fb7085f7d32 : FAILURE in 15m 42s (non-voting)\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/5d8029dd164d43328236bf7acbd2cf7d : SUCCESS in 30m 59s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/edc9b08547994b3b900b6b8587fead59 : SUCCESS in 44m 15s\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/5dfbb7c20b4c40c48cf39abb9b03246f : SUCCESS in 1h 07m 57s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/855f79f0a1574ecd9e4958dea2573bc2 : FAILURE in 59m 00s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-shard https://zuul.opendev.org/t/openstack/build/05db2d207c5048e2a2e0b11964578ee3 : FAILURE in 40m 06s (non-voting)\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/0acd569060f54b5aa06a5e8ed880a39a : SUCCESS in 32m 08s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/ea1527bca63842c4bdf03220b5db0cf8 : SUCCESS in 41m 55s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/02f390642dd2473b971b8051345398f0 : SUCCESS in 42m 17s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/88ac811e31a44fa6b735cfa29624b9ae : FAILURE in 29m 13s (non-voting)\n- metalsmith-integration-glance-centos9-uefi https://zuul.opendev.org/t/openstack/build/9a2951fda207437ab33869f0c0214826 : FAILURE in 23m 09s (non-voting)\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/b1928d184fcc4510aa8268f321842cf7 : SUCCESS in 1h 14m 53s (non-voting)\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/70fa09d9fd0a4e149535b2c4211b61d1 : FAILURE in 57m 14s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/6ab1b2eaf65a4a6aa9419cdfa1ec287b : SUCCESS in 33m 55s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/9520172848844bb48116e1d8c1b729c4 : SUCCESS in 52m 52s (non-voting)\n- ironic-inspector-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/13364b8c0dcd474caa6955cdba446f8a : SUCCESS in 44m 53s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/9218dd6a5ec34f6c83b9254c7f444c43 : SUCCESS in 44m 20s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/d05434558c614d7d880adf67b75a1a27 : SUCCESS in 43m 02s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/11a1a5f7c22149658914056bfa9004ec : FAILURE in 24m 54s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/72ddaddf17104e1ea9495f8f6916b3ce : SUCCESS in 44m 13s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/f9e0bb3357334b14bdf18baa44856848 : FAILURE in 11m 58s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"98dad302001e28496f1f75f53d108f39b0601866","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2025-09-04 16:31:19.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"83a05d3bddb57e971c7e23d9b3fe263b1538e0e4","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2025-09-04 16:32:49.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"ee3efacea7b890a91fd020edcd526ba794c556c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 17:18:17.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fd4ff5020e64bad8848deac8d6d4fd1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/987f279629dc4109be9d6f59c3c4f8e2 : SUCCESS in 14m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/19f3dbbe69104e66abe896595bae070f : SUCCESS in 20m 09s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d19a6aadb064449f8cf08f9dceea7a43 : SUCCESS in 21m 26s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"90b4d37a445c7b0759631d0fa54c895bdfcbf168","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 18:19:35.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/99f48d72aad5416d9913099aa40ad845\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ef3e52a4171340458ea026a73e0e79cd : SUCCESS in 4m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8572dd24b19c4eb69d8ee846401fa4d3 : SUCCESS in 10m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1fe7b651b644c20b247f4ed48630a9c : SUCCESS in 6m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c4635094afbb4a83bbd14716756fd1f4 : SUCCESS in 8m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b34c99402cb2488b99713b1e9f82e41e : SUCCESS in 7m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/59df7eb0e509479dac3f349e02ce439f : SUCCESS in 11m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/faa1be9c130443af8d1ffa66d3c0af00 : SUCCESS in 14m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86c123978cc4413eb6d94520e098570e : SUCCESS in 14m 27s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/23f683dc55ec46aab6beb4b1d4a6d860 : SUCCESS in 7m 12s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/eed5cae30dab4b80a8ed4474868a05b8 : SUCCESS in 8m 54s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/f4ad21d53a5e47b097c3ef54dd57c101 : SUCCESS in 10m 05s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/74322c4de2e24120b14a780ee8272883 : FAILURE in 39m 30s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/a311453adaed4522bd2799ccd07eb43b : SUCCESS in 55m 27s (non-voting)\n- ironic-tempest-standalone-advanced https://zuul.opendev.org/t/openstack/build/4b323b4f2e084047bba8c57040d56054 : SUCCESS in 1h 23m 42s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/a3b5e9e48d394ab592312e26a90b0cd9 : SUCCESS in 41m 12s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/b70ded07a33d45108bebb420ed959a8d : SUCCESS in 46m 52s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/693d95ed4fb04988a2226a714c617796 : FAILURE in 19m 31s (non-voting)\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/2947e34d40bc495aaf76f18204a64818 : SUCCESS in 35m 17s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/8974017688244e31a0de22ff4e3c3647 : SUCCESS in 37m 07s\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/d15b79b81d1c4e97860c23f4839e4a4e : SUCCESS in 1h 00m 02s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/9bce5070f2bf4975814ce4ba923f4888 : SUCCESS in 1h 45m 21s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-shard https://zuul.opendev.org/t/openstack/build/7f0e68b600634e78a8452cab42156a52 : SUCCESS in 1h 17m 39s (non-voting)\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/8fe896a178a7498d89558cc5cb03f9a9 : SUCCESS in 32m 12s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/af11783ca81b42c499068685c690556a : SUCCESS in 28m 07s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/cc011f3b50ea4cba984266d869f7707c : SUCCESS in 47m 45s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/e296c0a07f8b4bcd83e9e198a8bb4fdb : SUCCESS in 36m 28s\n- metalsmith-integration-glance-centos9-uefi https://zuul.opendev.org/t/openstack/build/1ca4e7f3fd19451d871f2803d8ac5c85 : SUCCESS in 49m 45s (non-voting)\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/1a59561cc5584ddc928360d3bb0ea578 : SUCCESS in 1h 11m 32s (non-voting)\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/60e539f2d42e4c739c50dac8d84ad98f : SUCCESS in 49m 32s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/4ce9203d35ff43f6acbc99568d90dca9 : SUCCESS in 33m 01s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/3a8f2d5650b54e4eae907888acfe9e0a : SUCCESS in 30m 07s (non-voting)\n- ironic-inspector-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/97a3add0ccee40bc804fc6e16bb9f218 : SUCCESS in 28m 34s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/031a7e54bced4dea8bf44242c144c0cb : SUCCESS in 26m 16s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/05ab5710dae747d3b0a99ca88d807b68 : SUCCESS in 46m 30s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/87739343e4124af89f7f4a729fc5b6f3 : FAILURE in 1h 34m 09s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e5297f5262304ab8832aac6d35f5662b : SUCCESS in 21m 43s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"73a02d8e7f56ea766928f9251de8ce9665585692","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2025-09-11 20:24:09.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"29ef4618b7f1c65c02ef7092436fda15fbc10ee8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2025-10-14 19:57:18.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"c1b9cce006b0fc17b7df8e328fd6d3ac33d55186","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 20:17:47.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/604a6b6f582d43fbacef5502ff9aa45a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8ac6880fda0d41df8f623d5b10ad6ae6 : SUCCESS in 14m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7d416d6dcaa0409f8f9fcd99c3d9a7ec : SUCCESS in 15m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/56a6f3a966e94085a484e3dc9eafc3cc : SUCCESS in 18m 17s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"d3f2963bdce4033db2620b584f6c5b5840c6c493","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 22:45:13.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d6672cece1a424fbff3d239c185c74d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9f0a818aa9294a93bacf9f0641c2f802 : SUCCESS in 7m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7805e5f438941bbb37ebdbd162eb17d : SUCCESS in 9m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e0e7027ae1a4d5ba593a8c9196454d4 : SUCCESS in 7m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/353ceee6cb9a402d9f9b99f2f1e42891 : SUCCESS in 6m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/79f5bdc6008641819b3b94ccfbf20aac : SUCCESS in 10m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e1923e68b9749d583f523523965cd11 : SUCCESS in 9m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b89886ce5d004577a678af1bdeaf7127 : SUCCESS in 19m 02s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/6d5f11cc7ffe4c65be224e69bda0d0f9 : SUCCESS in 3m 51s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/2f0b01f01a9041dfa83a6f2761384430 : SUCCESS in 7m 39s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/2e6c9d1c931a4b4ab5c6f6230b936957 : SUCCESS in 10m 03s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/30f0dd64c7334d859304604ca2ce3054 : SUCCESS in 35m 49s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/2bf1aaebfbc84fe7933e26164dcf3e8e : SUCCESS in 1h 05m 05s (non-voting)\n- ironic-tempest-standalone-advanced https://zuul.opendev.org/t/openstack/build/18fe54195db94f349568458a5a18af5f : SUCCESS in 54m 51s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/e69caa29aa9c4f34bf3181d8b21c40c4 : SUCCESS in 22m 21s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/72b733cf51a0473ba7375602325dada6 : SUCCESS in 22m 44s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/36703088f9704a1396fb63e91fc50022 : FAILURE in 27m 43s (non-voting)\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/7be4766e141149cea882531704263301 : SUCCESS in 19m 24s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/423bda31b1d94e03852965618a7ee73d : SUCCESS in 48m 30s (non-voting)\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/84a7da439dc14037aa686d9c1c121eac : SUCCESS in 27m 41s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-multinode https://zuul.opendev.org/t/openstack/build/d6a5211226c74c0b96f79455f7db2fd0 : FAILURE in 2h 42m 16s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-multinode-shard https://zuul.opendev.org/t/openstack/build/dea93823e101413bad58a8c3ba0f3a28 : SUCCESS in 39m 20s (non-voting)\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/9d8c9f95de2a424898145847536382d4 : SUCCESS in 1h 12m 49s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/a4fa7b5480e645e9ade034ee19ff5443 : SUCCESS in 18m 58s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/40c7e8fbf31445939330f6c866f6549c : SUCCESS in 44m 47s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/07238cdb121341c2b5475b176deee040 : SUCCESS in 1h 07m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c99562bddfbf4fb9b822723425fabe71 : SUCCESS in 1h 05m 18s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/c1abc875702a4ed18c2764ae79b45ae1 : SUCCESS in 1h 06m 34s (non-voting)\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/fe6514f0047a405eac570b9fdd3475fe : SUCCESS in 1h 07m 59s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/5ca48e3fbdf34823ae46900cd308fe00 : SUCCESS in 34m 08s\n- metal3-ironic-standalone-operator-functional https://zuul.opendev.org/t/openstack/build/dbe405d31b56412c94b4d23f960f35f8 : SUCCESS in 36m 22s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/eeb2b26434b84752ad11e1709f8fa9a5 : FAILURE in 18m 21s (non-voting)\n- ironic-inspector-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/34e684491a5a4e2f8fe66ec923c27e87 : FAILURE in 33m 35s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/94c8b09ad71b447c8cefee6198c519e3 : SUCCESS in 18m 00s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/9c3af7e41c3e463b9c0354f5d6315cac : SUCCESS in 47m 20s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/17772af8ab4e4e8ab6019ba14e92fabc : SUCCESS in 57m 21s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4c8b66ac9bb04f98b8603580916751d4 : SUCCESS in 43m 57s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8dde8546743c4a9e9976515bffd4582a : SUCCESS in 45m 21s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"d1cf0ea886f552b4d7ab6d51fa1d3002b86b54a4","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2025-10-15 14:25:49.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"3292039fd2db726850aa1960eb1958197e36fcbc","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2025-11-03 15:40:54.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"5fd66efc674ff8cbe7c6b803e1ee7734d6858528","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2025-11-05 01:20:07.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"1c182ee0d39d36534c4aad26a2f7be7d3a43a879","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2025-11-05 01:22:41.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"138752c03ee8cbc7b618ff4345724f299154c180","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-05 01:38:10.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ea5dcee3e954fdcb195ee965d23be93\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/99d88c9208e84e3f9d692c56547ffc32 : SUCCESS in 16m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0af3ea22862e40748664ba709e13a3f3 : SUCCESS in 14m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/dee7e802c8e74414bc6fcdbc36e527cf : SUCCESS in 12m 07s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"209b6753b4f2d8a84b8f51df8df903909aa79b61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-05 02:57:33.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fa4b75cbb4c4bcba6dbb4310e2512d2\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c460465d7ecf4cfdbd6b229433d1521c : SUCCESS in 8m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4a600e7481e44828ae87fc8b7ad4442 : SUCCESS in 9m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aec9f4c895334bbf8e729b24ae65fc8b : SUCCESS in 6m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e374ce3d755d4a5db75cbe62b122ee52 : SUCCESS in 11m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/28d01b6189b44355b6ce5854f4eb4936 : SUCCESS in 9m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdacf85a0deb45f89284ef1bb7654d08 : SUCCESS in 9m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0195807603bd4e2b946553160e23f1a4 : SUCCESS in 19m 58s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/227baea14dba48449e8fe79cdff8fe01 : SUCCESS in 5m 20s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/db1e6ae02ca94135b2b77e7818842e4f : SUCCESS in 8m 47s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/72f59ac279204d43bd6b591b389be737 : SUCCESS in 7m 51s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/6ebb01928dc74d8a970362aa8a3f72d8 : SUCCESS in 25m 08s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/99a577c498fd4dd28522a6c50dc3df01 : SUCCESS in 1h 00m 09s (non-voting)\n- ironic-tempest-standalone-advanced https://zuul.opendev.org/t/openstack/build/2d9b0a9bc08d47b7a617234789c5ec38 : FAILURE in 33m 02s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/43bc5814c61e47e1a230d0ea21951d6a : SUCCESS in 43m 41s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/94b8a247c4394c41bd5a511d774da455 : SUCCESS in 26m 00s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/64035df3018b4d6ea59fb9d26eb95f18 : FAILURE in 17m 37s (non-voting)\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/8982eeb05ff94a00b2c23d290ab30385 : SUCCESS in 53m 07s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/c88704b1f329499ba0dca8d27771ec8c : SUCCESS in 46m 14s (non-voting)\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/82060311dc4547a388e5c0997d53a4a8 : SUCCESS in 1h 03m 47s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-multinode https://zuul.opendev.org/t/openstack/build/d8d9f4522275477fa2f24259f95c2b96 : FAILURE in 56m 16s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-multinode-shard https://zuul.opendev.org/t/openstack/build/3587c45a0dc44483ae7d610ea3d6a49c : SUCCESS in 1h 10m 36s (non-voting)\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/d2575b3ba37d4c2c9d1644843f9f221c : SUCCESS in 35m 29s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/1e407f098c934e118c12f25c82297ebe : SUCCESS in 42m 32s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/295704e25a1942169cbcfd943e880bd5 : SUCCESS in 49m 34s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/0717a5adae814cea8c6665839ac5c767 : SUCCESS in 1h 06m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/895a171db6fe45a6a7edc9c7525926b6 : SUCCESS in 45m 12s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/000b13cefbf9449ca476cce849a3a64b : SUCCESS in 1h 32m 23s (non-voting)\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/35bdae4daa90419196db66041cddd833 : FAILURE in 1h 04m 59s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/8e5190d70dcc4d4ea4d535580f81f880 : SUCCESS in 31m 18s\n- metal3-ironic-standalone-operator-functional https://zuul.opendev.org/t/openstack/build/1587f14935e14fcc876441ed19319806 : SUCCESS in 33m 10s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/32ff307dbdf34a5f8aa96c718d9676b9 : FAILURE in 22m 27s (non-voting)\n- ironic-inspector-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/a0d100938d07499fb3587cac20a6368d : FAILURE in 23m 09s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/0af05c83af444a3ea9e8c408592fe9ee : SUCCESS in 31m 52s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/617657f0fa9941578dadb931c570d6b4 : SUCCESS in 46m 28s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/4b9612dbd1a84976a8471fddbc3c992c : SUCCESS in 52m 41s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fbbde9af71514831b92fa7860bdfa1f2 : SUCCESS in 48m 02s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d67d8cd3cc9a410a872904b76f4022a9 : SUCCESS in 29m 25s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"5cb0fc5b0a3dbc472e95fcfb5d3187fb9e6b9fb5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-05 02:58:00.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"01c1c8edf598610bbcb5a488acd062daa4a11561","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-05 04:20:31.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a55e261565b44552a19bdb4f0ce9c744\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c124ded584a34d13a1f3229e63c9ad1d : SUCCESS in 8m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/876e4454c53840ecb06c2631d2fdeff8 : SUCCESS in 6m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b2d6147b136b4051bd1948428b068586 : SUCCESS in 9m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ce525c8f42284d9ebec1f2ef162b5b3a : SUCCESS in 12m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91806b74b6dc4736b67c90f16d7f58ec : SUCCESS in 14m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb315dd8561940c2a4f448a48a7ab60c : SUCCESS in 20m 53s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/c26a7b8cda674610aa2881102f3c4870 : SUCCESS in 7m 37s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/c8c942e43dad4dd98d41227cd6f3ba9d : SUCCESS in 10m 04s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/2a3b7cebd8414561b8767674049f94e8 : SUCCESS in 38m 59s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/b22a2d4528bf43d6bc14f208dd13f372 : SUCCESS in 44m 59s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/0ea966b4b9db43b99049b2d2c0bce966 : SUCCESS in 46m 02s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/ba0c27048b28422096e1b5a7cf5ca369 : SUCCESS in 45m 59s\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/9fd6310559ad4f75815d336673fed496 : SUCCESS in 1h 06m 16s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/0f3815587fa449c991ce2641504f22bb : SUCCESS in 34m 14s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/d4a05610fbc04be092f2ad3320d9e796 : SUCCESS in 44m 34s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/96847be8ab744fa793e9287b24cfaf84 : SUCCESS in 1h 04m 39s\n- metal3-integration https://zuul.opendev.org/t/openstack/build/020ad7bf1e814087977c68720dc5327a : SUCCESS in 31m 42s","accounts_in_message":[],"_revision_number":12},{"id":"2dee06ce4f48a6ba41002fa7c6a545423039bcf5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-05 04:20:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":12},{"id":"e2ccad6703da18687bf3bc70ec67b1d4b4ea6e64","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-05 04:21:54.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e18eada3f4b4b6b9ad5acc819c98686\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/37edd161295e408da0afa7112bd7cb88 : SUCCESS in 42s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1212b3591bb94aa089d7f03fb33302b0 : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1f70de36b2a4607b592a363d73eb38b : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"7a67d4f18319ca0230c99ca2679f94a2cd9c5950","revisions":{"182eb6bbe9348e967ba05aa76537d8c4c110fc36":{"kind":"REWORK","_number":1,"created":"2025-07-23 17:19:11.000000000","uploader":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"ref":"refs/changes/13/955713/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/13/955713/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/955713/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/955713/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/955713/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/13/955713/1"}}},"commit":{"parents":[{"commit":"acd5f1bfe22eb5b082d92397ff5de265007a4f46","subject":"Add a new \u0027physical_network\u0027 field to the Portgroup object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/acd5f1bfe22eb5b082d92397ff5de265007a4f46"}]}],"author":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2025-07-23 17:18:03.000000000","tz":-300},"committer":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2025-07-23 17:18:03.000000000","tz":-300},"subject":"Add a new \u0027category\u0027 field to the Portgroup object","message":"Add a new \u0027category\u0027 field to the Portgroup object\n\nAdds a new category field to the portgroup object. Foundational\nwork for first milestone of trait based port scheduling.\n\nDepends-On: https://review.opendev.org/c/openstack/ironic/+/955625\nChange-Id: I5100144a330602996c27ed18d2bbde09be6e9571\nSigned-off-by: Clif Houck \u003cme@clifhouck.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/182eb6bbe9348e967ba05aa76537d8c4c110fc36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/182eb6bbe9348e967ba05aa76537d8c4c110fc36"}]},"branch":"refs/heads/master"},"e0e99511c717aaaa7e1dddcd6bb7dc763ac2a945":{"kind":"REWORK","_number":2,"created":"2025-07-23 21:20:33.000000000","uploader":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"ref":"refs/changes/13/955713/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/13/955713/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/955713/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/955713/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/955713/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/13/955713/2"}}},"commit":{"parents":[{"commit":"acd5f1bfe22eb5b082d92397ff5de265007a4f46","subject":"Add a new \u0027physical_network\u0027 field to the Portgroup object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/acd5f1bfe22eb5b082d92397ff5de265007a4f46"}]}],"author":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2025-07-23 17:18:03.000000000","tz":-300},"committer":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2025-07-23 21:20:22.000000000","tz":-300},"subject":"Add a new \u0027category\u0027 field to the Portgroup object","message":"Add a new \u0027category\u0027 field to the Portgroup object\n\nAdds a new category field to the portgroup object. Foundational\nwork for first milestone of trait based port scheduling.\n\nDepends-On: https://review.opendev.org/c/openstack/ironic/+/955625\nChange-Id: I5100144a330602996c27ed18d2bbde09be6e9571\nSigned-off-by: Clif Houck \u003cme@clifhouck.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/e0e99511c717aaaa7e1dddcd6bb7dc763ac2a945"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/e0e99511c717aaaa7e1dddcd6bb7dc763ac2a945"}]},"branch":"refs/heads/master"},"3a9b1be61e70ce1eb8ff224eafe264196246e974":{"kind":"REWORK","_number":3,"created":"2025-07-24 15:16:40.000000000","uploader":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"ref":"refs/changes/13/955713/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/13/955713/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/955713/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/955713/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/955713/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/13/955713/3"}}},"commit":{"parents":[{"commit":"bf53b7acf1d73f8ad6b10fbf6d7537908193c89d","subject":"Add a new \u0027physical_network\u0027 field to the Portgroup object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/bf53b7acf1d73f8ad6b10fbf6d7537908193c89d"}]}],"author":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2025-07-23 17:18:03.000000000","tz":-300},"committer":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2025-07-24 15:16:18.000000000","tz":-300},"subject":"Add a new \u0027category\u0027 field to the Portgroup object","message":"Add a new \u0027category\u0027 field to the Portgroup object\n\nAdds a new category field to the portgroup object. Foundational\nwork for first milestone of trait based port scheduling.\n\nDepends-On: https://review.opendev.org/c/openstack/ironic/+/955625\nDepends-On: https://review.opendev.org/c/openstack/ironic-tempest-plugin/+/955799\nChange-Id: I5100144a330602996c27ed18d2bbde09be6e9571\nSigned-off-by: Clif Houck \u003cme@clifhouck.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/3a9b1be61e70ce1eb8ff224eafe264196246e974"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/3a9b1be61e70ce1eb8ff224eafe264196246e974"}]},"branch":"refs/heads/master"},"9b4be60faab10bde9e60b0154442d1346208f45b":{"kind":"REWORK","_number":4,"created":"2025-08-13 16:41:52.000000000","uploader":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"ref":"refs/changes/13/955713/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/13/955713/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/955713/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/955713/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/955713/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/13/955713/4"}}},"commit":{"parents":[{"commit":"299c461304e641715c4aca1b35f4448d0be28286","subject":"Add a new \u0027physical_network\u0027 field to the Portgroup object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/299c461304e641715c4aca1b35f4448d0be28286"}]}],"author":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2025-07-23 17:18:03.000000000","tz":-300},"committer":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2025-08-13 16:41:44.000000000","tz":-300},"subject":"Add a new \u0027category\u0027 field to the Portgroup object","message":"Add a new \u0027category\u0027 field to the Portgroup object\n\nAdds a new category field to the portgroup object. Foundational\nwork for first milestone of trait based port scheduling.\n\nDepends-On: https://review.opendev.org/c/openstack/ironic/+/955625\nDepends-On: https://review.opendev.org/c/openstack/ironic-tempest-plugin/+/955799\nChange-Id: I5100144a330602996c27ed18d2bbde09be6e9571\nSigned-off-by: Clif Houck \u003cme@clifhouck.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/9b4be60faab10bde9e60b0154442d1346208f45b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/9b4be60faab10bde9e60b0154442d1346208f45b"}]},"branch":"refs/heads/master"},"81be0e37cd168ad3ef6817b24f7d55ec574bbbf6":{"kind":"NO_CHANGE","_number":5,"created":"2025-08-15 18:41:20.000000000","uploader":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"ref":"refs/changes/13/955713/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/13/955713/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/955713/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/955713/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/955713/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/13/955713/5"}}},"commit":{"parents":[{"commit":"20b1b9ef26418810b8177b91cb08be8a2e1fef67","subject":"Add a new \u0027physical_network\u0027 field to the Portgroup object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/20b1b9ef26418810b8177b91cb08be8a2e1fef67"}]}],"author":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2025-07-23 17:18:03.000000000","tz":-300},"committer":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2025-08-15 18:40:57.000000000","tz":-300},"subject":"Add a new \u0027category\u0027 field to the Portgroup object","message":"Add a new \u0027category\u0027 field to the Portgroup object\n\nAdds a new category field to the portgroup object. Foundational\nwork for first milestone of trait based port scheduling.\n\nDepends-On: https://review.opendev.org/c/openstack/ironic/+/955625\nDepends-On: https://review.opendev.org/c/openstack/ironic-tempest-plugin/+/955799\nChange-Id: I5100144a330602996c27ed18d2bbde09be6e9571\nSigned-off-by: Clif Houck \u003cme@clifhouck.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/81be0e37cd168ad3ef6817b24f7d55ec574bbbf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/81be0e37cd168ad3ef6817b24f7d55ec574bbbf6"}]},"branch":"refs/heads/master"},"c69cc6d05c3d9eef64145297afa6e449e54c9cf5":{"kind":"REWORK","_number":6,"created":"2025-08-15 18:44:27.000000000","uploader":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"ref":"refs/changes/13/955713/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/13/955713/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/955713/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/955713/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/955713/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/13/955713/6"}}},"commit":{"parents":[{"commit":"20b1b9ef26418810b8177b91cb08be8a2e1fef67","subject":"Add a new \u0027physical_network\u0027 field to the Portgroup object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/20b1b9ef26418810b8177b91cb08be8a2e1fef67"}]}],"author":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2025-07-23 17:18:03.000000000","tz":-300},"committer":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2025-08-15 18:44:16.000000000","tz":-300},"subject":"Add a new \u0027category\u0027 field to the Portgroup object","message":"Add a new \u0027category\u0027 field to the Portgroup object\n\nAdds a new category field to the portgroup object. Foundational\nwork for first milestone of trait based port scheduling.\n\nDepends-On: https://review.opendev.org/c/openstack/ironic/+/955625\nDepends-On: https://review.opendev.org/c/openstack/ironic-tempest-plugin/+/955799\nChange-Id: I5100144a330602996c27ed18d2bbde09be6e9571\nSigned-off-by: Clif Houck \u003cme@clifhouck.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/c69cc6d05c3d9eef64145297afa6e449e54c9cf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/c69cc6d05c3d9eef64145297afa6e449e54c9cf5"}]},"branch":"refs/heads/master"},"94ff9436b019e66b325f94e176e5bc8b785fe25d":{"kind":"REWORK","_number":7,"created":"2025-08-18 18:52:31.000000000","uploader":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"ref":"refs/changes/13/955713/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/13/955713/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/955713/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/955713/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/955713/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/13/955713/7"}}},"commit":{"parents":[{"commit":"e7c3caa80f2307de1ff9e0d454f654eda1dbc1cd","subject":"Add a new \u0027physical_network\u0027 field to the Portgroup object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/e7c3caa80f2307de1ff9e0d454f654eda1dbc1cd"}]}],"author":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2025-07-23 17:18:03.000000000","tz":-300},"committer":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2025-08-18 18:48:50.000000000","tz":-300},"subject":"Add a new \u0027category\u0027 field to the Portgroup object","message":"Add a new \u0027category\u0027 field to the Portgroup object\n\nAdds a new category field to the portgroup object. Foundational\nwork for first milestone of trait based port scheduling.\n\nDepends-On: https://review.opendev.org/c/openstack/ironic-tempest-plugin/+/955799\nChange-Id: I5100144a330602996c27ed18d2bbde09be6e9571\nSigned-off-by: Clif Houck \u003cme@clifhouck.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/94ff9436b019e66b325f94e176e5bc8b785fe25d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/94ff9436b019e66b325f94e176e5bc8b785fe25d"}]},"branch":"refs/heads/master"},"9b8876cfdc87743cbac865b4c9246fbcc3b88d4f":{"kind":"REWORK","_number":8,"created":"2025-08-19 14:12:46.000000000","uploader":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"ref":"refs/changes/13/955713/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/13/955713/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/955713/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/955713/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/955713/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/13/955713/8"}}},"commit":{"parents":[{"commit":"81e5b52167c640e19aeff7dde8a3f75057a3095f","subject":"Add a new \u0027physical_network\u0027 field to the Portgroup object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/81e5b52167c640e19aeff7dde8a3f75057a3095f"}]}],"author":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2025-07-23 17:18:03.000000000","tz":-300},"committer":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2025-08-19 14:10:19.000000000","tz":-300},"subject":"Add a new \u0027category\u0027 field to the Portgroup object","message":"Add a new \u0027category\u0027 field to the Portgroup object\n\nAdds a new category field to the portgroup object. Foundational\nwork for first milestone of trait based port scheduling.\n\nDepends-On: https://review.opendev.org/c/openstack/ironic-tempest-plugin/+/955799\nChange-Id: I5100144a330602996c27ed18d2bbde09be6e9571\nSigned-off-by: Clif Houck \u003cme@clifhouck.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/9b8876cfdc87743cbac865b4c9246fbcc3b88d4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/9b8876cfdc87743cbac865b4c9246fbcc3b88d4f"}]},"branch":"refs/heads/master"},"8f915239a14dd61262af9c75f783c62afc8bc5c3":{"kind":"REWORK","_number":9,"created":"2025-08-25 15:00:13.000000000","uploader":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"ref":"refs/changes/13/955713/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/13/955713/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/955713/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/955713/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/955713/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/13/955713/9"}}},"commit":{"parents":[{"commit":"8303ea4351485c2cb9a15d30e23027fd3d7ee470","subject":"Add a new \u0027physical_network\u0027 field to the Portgroup object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/8303ea4351485c2cb9a15d30e23027fd3d7ee470"}]}],"author":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2025-07-23 17:18:03.000000000","tz":-300},"committer":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2025-08-25 15:00:06.000000000","tz":-300},"subject":"Add a new \u0027category\u0027 field to the Portgroup object","message":"Add a new \u0027category\u0027 field to the Portgroup object\n\nAdds a new category field to the portgroup object. Foundational\nwork for first milestone of trait based port scheduling.\n\nDepends-On: https://review.opendev.org/c/openstack/ironic-tempest-plugin/+/955799\nChange-Id: I5100144a330602996c27ed18d2bbde09be6e9571\nSigned-off-by: Clif Houck \u003cme@clifhouck.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/8f915239a14dd61262af9c75f783c62afc8bc5c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/8f915239a14dd61262af9c75f783c62afc8bc5c3"}]},"branch":"refs/heads/master"},"98b913ded10b8400a9bd92c02ed202f757dfd33d":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2025-09-04 16:31:19.000000000","uploader":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"ref":"refs/changes/13/955713/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/13/955713/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/955713/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/955713/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/955713/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/13/955713/10"}}},"commit":{"parents":[{"commit":"8a3302d7d22b2150a9d63d4e9e4a9641a25d89b9","subject":"Add a new \u0027physical_network\u0027 field to the Portgroup object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/8a3302d7d22b2150a9d63d4e9e4a9641a25d89b9"}]}],"author":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2025-07-23 17:18:03.000000000","tz":-300},"committer":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2025-09-04 16:31:10.000000000","tz":-300},"subject":"Add a new \u0027category\u0027 field to the Portgroup object","message":"Add a new \u0027category\u0027 field to the Portgroup object\n\nAdds a new category field to the portgroup object. Foundational\nwork for first milestone of trait based port scheduling.\n\nDepends-On: https://review.opendev.org/c/openstack/ironic-tempest-plugin/+/955799\nChange-Id: I5100144a330602996c27ed18d2bbde09be6e9571\nSigned-off-by: Clif Houck \u003cme@clifhouck.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/98b913ded10b8400a9bd92c02ed202f757dfd33d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/98b913ded10b8400a9bd92c02ed202f757dfd33d"}]},"branch":"refs/heads/master"},"a93ce4fb3143a5fb671877e337794956cdd167ce":{"kind":"REWORK","_number":11,"created":"2025-10-14 19:57:18.000000000","uploader":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"ref":"refs/changes/13/955713/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/13/955713/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/955713/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/955713/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/955713/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/13/955713/11"}}},"commit":{"parents":[{"commit":"4a4ea08fce2c82e53fc76104df012e192532ac44","subject":"Add a new \u0027physical_network\u0027 field to the Portgroup object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4a4ea08fce2c82e53fc76104df012e192532ac44"}]}],"author":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2025-07-23 17:18:03.000000000","tz":-300},"committer":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2025-10-14 19:54:24.000000000","tz":-300},"subject":"Add a new \u0027category\u0027 field to the Portgroup object","message":"Add a new \u0027category\u0027 field to the Portgroup object\n\nAdds a new category field to the portgroup object. Foundational\nwork for first milestone of trait based port scheduling.\n\nDepends-On: https://review.opendev.org/c/openstack/ironic-tempest-plugin/+/955799\nChange-Id: I5100144a330602996c27ed18d2bbde09be6e9571\nSigned-off-by: Clif Houck \u003cme@clifhouck.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/a93ce4fb3143a5fb671877e337794956cdd167ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/a93ce4fb3143a5fb671877e337794956cdd167ce"}]},"branch":"refs/heads/master"},"7a67d4f18319ca0230c99ca2679f94a2cd9c5950":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2025-11-05 01:20:07.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/13/955713/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/13/955713/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/955713/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/955713/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/955713/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/13/955713/12"}}},"commit":{"parents":[{"commit":"d9fc0686308c8d955eaa714962e41b8688b331c0","subject":"Merge \"api: Add schema for inspection rules API (requests)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/d9fc0686308c8d955eaa714962e41b8688b331c0"}]}],"author":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2025-07-23 17:18:03.000000000","tz":-300},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2025-11-04 20:44:01.000000000","tz":-480},"subject":"Add a new \u0027category\u0027 field to the Portgroup object","message":"Add a new \u0027category\u0027 field to the Portgroup object\n\nAdds a new category field to the portgroup object. Foundational\nwork for first milestone of trait based port scheduling.\n\nDepends-On: https://review.opendev.org/c/openstack/ironic-tempest-plugin/+/955799\nChange-Id: I5100144a330602996c27ed18d2bbde09be6e9571\nSigned-off-by: Clif Houck \u003cme@clifhouck.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/7a67d4f18319ca0230c99ca2679f94a2cd9c5950"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/7a67d4f18319ca0230c99ca2679f94a2cd9c5950"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
