)]}'
{"id":"openstack%2Fironic~960283","triplet_id":"openstack%2Fironic~master~Iba84bbe5da541700d20a445818a4a0d584f1eca8","project":"openstack/ironic","branch":"master","topic":"oci-fixes","attention_set":{},"removed_from_attention_set":{"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2025-09-24 08:36:44.000000000","reason":"Change was submitted"},"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2025-09-24 08:36:44.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"Iba84bbe5da541700d20a445818a4a0d584f1eca8","subject":"Fix OCI artifacts pointing to a single manifest","status":"MERGED","created":"2025-09-09 15:43:33.000000000","updated":"2025-09-24 08:37:46.000000000","submitted":"2025-09-24 08:36:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"960283-oci-fixes","meta_rev_id":"ee5c476a01ce7e33e694d0492c78162a3a69b220","_number":960283,"virtual_id_number":960283,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-24 08:36:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":2,"date":"2025-09-23 18:13:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":2,"date":"2025-09-24 07:21:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":1,"date":"2025-09-24 07:21:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":1,"date":"2025-09-23 18:13:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-09 16:21:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-09 17:06:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-18 22:48:00.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2025-09-23 18:13:04.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2025-09-24 07:21:40.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"}],"messages":[{"id":"ae38d03c09714498124f21cfc1f9771a91ff25f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2025-09-09 15:43:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"62a89fa82f69712547ad6f6798b35f345aeccc72","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2025-09-09 15:43:52.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c897c6e1273b3bc2d5ffb3e54ecc84111fb3bbf6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-09 16:21:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a7afc00d2c14214888d92ba3da9815d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a3f260985d4f47a1831d2f3f237d54c2 : SUCCESS in 19m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a62777ab675842c79ce00b630ff02258 : FAILURE in 16m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1ab37d7b010c41abbdfe8ebda8989b1d : SUCCESS in 24m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"929f99b8fb714a4f91bb8d57cb00ed78cf8e2a26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-09 17:06:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/374dbe78f4254c5ab4905c9e948198bc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f1539621e3a45fbaaee13da0bd5e727 : SUCCESS in 13m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/174d263c81414b46b4200119a8a860e5 : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dca92c5683f640b8aa74935e17f221b3 : SUCCESS in 5m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2d91d5245fe34b2fb468d6033190215e : SUCCESS in 6m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5646370a40814364a314628442a55e55 : SUCCESS in 12m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cbc791158b34f00ab8387dccbb7816f : SUCCESS in 15m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/128b6fc326e244f096e1984b665c0384 : SUCCESS in 16m 06s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/d55e9ec50fb94f828da766073a38aca1 : SUCCESS in 5m 35s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/be2c5c6b567e4ba5bd48d607776773bf : SUCCESS in 10m 58s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/a06eb02aa4b54358bb2a496f128277d3 : SUCCESS in 8m 41s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/d94e7cdfd35a44d4a4c3f71a5b4d32c9 : SUCCESS in 32m 53s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/b68ac26d998d486dba234973b24f6e06 : SUCCESS in 56m 21s (non-voting)\n- ironic-tempest-standalone-advanced https://zuul.opendev.org/t/openstack/build/36c5946d03a142cfaa10a2bd0c774c4b : SUCCESS in 33m 18s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/fb81d720d9cf4b4b9447396798c251e9 : SUCCESS in 50m 55s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/a1e37cccec884353967279079670c151 : SUCCESS in 34m 30s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/312ad4f6bbfd4dbd91938a3fd3dbd2c1 : FAILURE in 25m 50s (non-voting)\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/3f9351ba02fb4ddeb23f6e1d296c6b62 : SUCCESS in 22m 12s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/c0ccdc2767d94b9c8b21af3d7156d108 : SUCCESS in 22m 12s\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/62aa34f8b56d4aa6af3b88e21894beec : SUCCESS in 27m 56s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/bab725b050f54b83a2373e0c3684139a : SUCCESS in 1h 02m 46s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-shard https://zuul.opendev.org/t/openstack/build/72bcf4ab36b94bc187e159bfd1bd4436 : SUCCESS in 36m 59s (non-voting)\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/952bbc2e6d55452787a1942bc389199e : SUCCESS in 1h 11m 09s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/b1a41bede61d4490b01c6f64e32b7e4e : SUCCESS in 30m 28s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/973532cba5444596b3e758fdb1c20598 : SUCCESS in 54m 19s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/bddf206e71b74a2eb3e7ed53e9919564 : SUCCESS in 49m 00s\n- metalsmith-integration-glance-centos9-uefi https://zuul.opendev.org/t/openstack/build/499ce6c4afe84146aa69f2be4979093c : SUCCESS in 1h 20m 36s (non-voting)\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/8b88348fd8154517bc3385fc234a18d6 : SUCCESS in 1h 14m 46s (non-voting)\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/d86e645d122b445fb23786636140d82e : SUCCESS in 1h 18m 05s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/bde4d13a5f084258a1303c2afa31f9a4 : SUCCESS in 33m 56s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/16bcedbeefb042b786da892ef94a7d95 : SUCCESS in 51m 00s (non-voting)\n- ironic-inspector-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/8494830df31c43639c5c39cd24c505f9 : SUCCESS in 44m 34s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/49ef8bdd6db44c2ca51071cbfea67331 : SUCCESS in 44m 09s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/d590f544df874232bb664735218fcf1d : SUCCESS in 20m 08s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/071781e4591e4ce1b941ef3fc7aee522 : SUCCESS in 56m 50s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/028712de3140475ba2e39b7454172eab : SUCCESS in 36m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cf63a21cc503b4191a56a22e2c0404151847f0a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2025-09-17 12:12:20.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"bb6a264cb075748fded631de43a6b28dcad0173f","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2025-09-17 12:13:36.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"aa6211671ff2e06f3d2ce7d7496d358efbc69225","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 12:34:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eaa57c21dd484ee58e0847e53e8c2150\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cdbbe280db65449a8c77a14a0a1ded58 : SUCCESS in 14m 04s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1030b8014a5543af964eba25073cb813 : SUCCESS in 18m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/360575b35ad643b284d159840428acec : SUCCESS in 19m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"84680cf1f67973cbec6ac75074005ca7d195caee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 14:06:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8527ef361c0543d08fabbec619752883\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40b30600fc89412e97fb56fd820f4aae : SUCCESS in 12m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0887d48182854d4da4f5652f96271203 : SUCCESS in 8m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cacd390e253345a3b493eda863fbe206 : SUCCESS in 9m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/77d6ef943c104064b6bcde25db810119 : SUCCESS in 7m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b9fddbfbb77e4696b68120fdf28882db : SUCCESS in 10m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/891b6beac0504fb8bee264f1b4f5bd1e : SUCCESS in 16m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa6e4914ebab40be9def66883ac559dc : SUCCESS in 18m 39s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/e65bfaa0cc4f4d7e81d1e36dc8dfb593 : SUCCESS in 4m 46s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/f9506575598046f9927fcace85c3bfb9 : SUCCESS in 11m 02s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/309c0d11a051403f9c1664c19c3f33fb : SUCCESS in 8m 31s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/1e5db0a81df44a26b66744b81736d115 : SUCCESS in 32m 11s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/a9194c8e0a5543b49226896d444a7acc : SUCCESS in 48m 32s (non-voting)\n- ironic-tempest-standalone-advanced https://zuul.opendev.org/t/openstack/build/997e9c61bc1c4b71adb59e2e0fecc0af : SUCCESS in 1h 03m 05s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/4e602e7f8fab415c99b70a66b7e4fa15 : SUCCESS in 41m 40s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/72fd001657c44d96b8c81ac87557a322 : SUCCESS in 44m 39s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/111e36fd8d0342d4aa24479fc86057e1 : FAILURE in 21m 01s (non-voting)\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/358504cb73f5465fbfd844c3cba78cd0 : SUCCESS in 36m 56s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/52f0ea9ca3164311ba1b1776a7afaeb7 : SUCCESS in 33m 31s\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/85c6993417f94c99a02fac56092d998d : SUCCESS in 50m 08s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/cf6b4c17ee63446ba860a5d10d4e57cc : SUCCESS in 1h 37m 40s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-shard https://zuul.opendev.org/t/openstack/build/85fdbbf2d87f4e48b732c152051870a2 : SUCCESS in 1h 03m 44s (non-voting)\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/d41f39f94f8d4d0c8e88546af486cab7 : SUCCESS in 55m 03s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/6968e389a5044e9c9354cfd74696cd20 : SUCCESS in 28m 42s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/749f9ee0e07b48a6b6f03397f7b0bd46 : SUCCESS in 43m 01s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/a33b600f78034fe3b55786373258b1dc : SUCCESS in 58m 27s\n- metalsmith-integration-glance-centos9-uefi https://zuul.opendev.org/t/openstack/build/329a4c2fde1244a5bd3e9f45453de49d : SUCCESS in 1h 15m 19s (non-voting)\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/72ecbbcc220243dab3c1d781a52440a7 : SUCCESS in 1h 07m 59s (non-voting)\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/37e645f84e714a28b4d8b7313f932110 : SUCCESS in 1h 04m 40s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/3588f210c0cd406c8a1722b41b9a4298 : SUCCESS in 35m 41s\n- metal3-ironic-standalone-operator-functional https://zuul.opendev.org/t/openstack/build/429540c3f89e4e1588653c41f61ffa35 : SUCCESS in 35m 36s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/61d749f4f8cd46499baec0808cbce876 : SUCCESS in 37m 19s (non-voting)\n- ironic-inspector-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/1fe5bdeb8cfa45ea8d8dfceee133dd76 : SUCCESS in 35m 10s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/f6f51d0525674a998160400f647c7e1a : SUCCESS in 33m 44s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/b96b3db32cdb4d1ca4ddfd4239779208 : SUCCESS in 33m 06s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/423c29dfc207497f85b49984008806ab : SUCCESS in 43m 08s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f449f8e6477c4e11b907839ed4002181 : SUCCESS in 33m 08s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d9759e81606475a9e90fe43db2142eb : SUCCESS in 34m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c52da2336cfff27d45600339c2e178407084a5ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2025-09-17 14:35:13.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"e3b08fef76127f27163fd9dd4962d7d9784e06b5","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2025-09-17 14:39:37.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"97bc3678a8af95c0c3909b0b442cc6371df91a19","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 15:21:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/042652e7149147d4b55099e9355d32b0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dfa17eeb508e4be88236458994451215 : FAILURE in 14m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/28af72e67f57428f8620209832183ef4 : FAILURE in 11m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6fc4a9c100eb498a8042b6e071f6c52e : FAILURE in 25m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"981b6796cbad6a6cb2aa46f015a0684a9722100f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2025-09-17 17:19:49.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Workflow-1\n","accounts_in_message":[],"_revision_number":4},{"id":"d0d1e82e282b2a192749222dc3f8d7f5f2243dc3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 18:59:36.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/507bc19f579f465b851add1d8aebf037\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/eed26010bfc24ace81e285cfa4f18c4f : FAILURE in 34m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4fb4d46f8a404e98a82f173ce81ad360 : FAILURE in 32m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/95a4812ffa7f40e484e2ccbbe4f0e4ad : FAILURE in 22m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b63d7b10166076971a95a4650c61e6034b0fee12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 22:32:52.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/54040ca2283d486180bac77d01f3b245\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f57b16df9ef144e9b9102486d5df75a7 : FAILURE in 12m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e0ac94290414d86a2eee4d7ca5e0327 : SUCCESS in 8m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b10b0d7ca968457594afcef0110ba52d : FAILURE in 7m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/48471b69ac3f482f82838e72a52f66a9 : FAILURE in 10m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0c78b784053346aab24948c5017dde26 : FAILURE in 10m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5783bc2e44894c05bd14c106b1e43745 : SUCCESS in 14m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91811034dc5647f6a73b4dafef987c09 : SUCCESS in 17m 02s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/bcc0b5a3be0049acbb03ef9ffb786c18 : SUCCESS in 6m 47s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/dafeebac8ddf4da1ad6df088f5958347 : FAILURE in 9m 36s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/2425c18e43174b97ba869286b2eb32fe : FAILURE in 9m 00s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/ce5e26f210414d2e8009bfeafb7c3ede : SUCCESS in 39m 07s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/6f807bcbb6fc49538f548c96689f7eb1 : SUCCESS in 58m 10s (non-voting)\n- ironic-tempest-standalone-advanced https://zuul.opendev.org/t/openstack/build/7f44e87f14f84c8086c9b315d8d45eab : SUCCESS in 57m 40s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/7e809623b48e40febd12a7ca0d58c75b : SUCCESS in 36m 54s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/1bc78e56e4c341c582006a9cc8448c91 : SUCCESS in 51m 58s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/c53b780379c343bcba7f7dbe27fc95ef : FAILURE in 17m 30s (non-voting)\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/5190f6180f644a95abd282e449d14438 : SUCCESS in 37m 14s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/bfa978501ee7472e99901d79fdfccd1e : SUCCESS in 45m 16s\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/b1148f31aa6e456a890f218de701297a : SUCCESS in 57m 01s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/d5ccfd864dd84a92a040d52799321f06 : SUCCESS in 1h 44m 17s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-shard https://zuul.opendev.org/t/openstack/build/00dff1819083420f92fe2f4fb9613a49 : SUCCESS in 1h 15m 45s (non-voting)\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/fd4a6f8d70fc41e5a317fc9fd63be171 : SUCCESS in 1h 02m 46s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/80ece2ac2166468186210371a7050fd9 : SUCCESS in 37m 21s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/2c113c0594034cc6b269fda07a35636d : SUCCESS in 42m 42s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/8666f7348d3e4607b15c76d560eadce2 : SUCCESS in 1h 00m 54s\n- metalsmith-integration-glance-centos9-uefi https://zuul.opendev.org/t/openstack/build/7c858c9d827640e298bbfd6ddfc2eb56 : SUCCESS in 1h 04m 09s (non-voting)\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/d87224125af44765bfc5b0bd5a1b7cc4 : SUCCESS in 1h 27m 33s (non-voting)\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/9ad7a86707fb49c99a5c2f8e8fd69bf4 : SUCCESS in 1h 22m 54s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/4a01099c3e8b4a8893e30b539aded2f6 : SUCCESS in 34m 39s\n- metal3-ironic-standalone-operator-functional https://zuul.opendev.org/t/openstack/build/bcfd48e6ced5499f8e24b8a51da7bbf1 : SUCCESS in 36m 35s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/f69417aab0ee4c16a46c2fa421b6b495 : SUCCESS in 35m 08s (non-voting)\n- ironic-inspector-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/670287a5e3f54b5dac5a84ddc5a27f2f : SUCCESS in 35m 49s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/66519380bb8444a4a78f7d6dc130014e : SUCCESS in 30m 54s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/71539fa737af4580a9547c9199f33067 : SUCCESS in 37m 16s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/d66b4ed4a103438081134869c79c3abf : SUCCESS in 40m 39s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b1f372d58f5f494ead4af7256e1c6397 : SUCCESS in 35m 40s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2cf2e915eed6479eb15941df08e1bf0b : SUCCESS in 31m 57s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"99c8af1302e3dc281a559f557956b2f941100e5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2025-09-18 11:19:51.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"901d8f0a0175f0fdf7513146983f335db1dd4fc8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-18 12:01:49.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09b538718822432c9d1519e7bd31699e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/70242c456c274b99b7a494bfb2fadb91 : SUCCESS in 12m 27s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3d2b3d78a96d4fc38231a6340ecaba2c : FAILURE in 21m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6057f4d0d5a74afcbb74fa33dc5c454c : SUCCESS in 22m 44s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1b91415e9b7d3671c8bc3a3137a32c3d917046c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-18 13:19:23.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5796d0543332427abfd2d6e6a9592866\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b75131d9417d4ff681793ad1be2d2d4c : SUCCESS in 10m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8da9d7290f9645e58ce5d77bb7b51f21 : SUCCESS in 8m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c2fbfc52dbdf4a9d997d3ffab97f0ea1 : SUCCESS in 8m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/828d549b80a84dea8b158b0472110e8a : SUCCESS in 10m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7830337551734cc3a31f54c42f5bbd30 : SUCCESS in 13m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7830390b5cbc4f4484a2ad6e05dc41c7 : SUCCESS in 13m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d3d9b79496e441b8d8e0400fe1249b3 : SUCCESS in 19m 27s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/177b1cb1b8e94ac78854413a50b113b7 : SUCCESS in 5m 24s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/6a3094e64f3c4060b2ea1bf0fbdae126 : SUCCESS in 9m 46s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/cc15dfcdeea442a18f7eed3cdf9d2fd3 : SUCCESS in 8m 12s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/16e1a0f7c31345229dfa9808f1d768dd : SUCCESS in 15m 40s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/dfd78b07a12148d19c4ba8eff33913c6 : SUCCESS in 1h 02m 43s (non-voting)\n- ironic-tempest-standalone-advanced https://zuul.opendev.org/t/openstack/build/c806b9043b874baa97cc25db34bc5d49 : SUCCESS in 1h 17m 07s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/ee266239d700413297550470e36aadb1 : SUCCESS in 17m 19s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/8346a2e3771e442c9523a18f88ff0997 : SUCCESS in 45m 52s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/c8b8223f5329477ab311d6ebbf1e45b6 : FAILURE in 11m 31s (non-voting)\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/c90533a253034c579beea74ce9c92e67 : SUCCESS in 45m 53s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/4561368c962f4f579aa7e354aca48440 : SUCCESS in 43m 30s\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/73b737c2b82a4441977708e7c24e6f37 : SUCCESS in 49m 39s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/4d79ad2ced7549af8df154dc0083fe60 : SUCCESS in 1h 48m 36s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-shard https://zuul.opendev.org/t/openstack/build/61d99d90e70f45679eab718f8692d7dc : SUCCESS in 1h 25m 05s (non-voting)\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/4c9b6963d15e4b8499ad9f4edb3baaf6 : SUCCESS in 1h 11m 23s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/ed994ca26c324828898efe4cf993b8b5 : SUCCESS in 36m 15s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/91ace0805e3540f6a5f150a62dfdd4bb : SUCCESS in 37m 25s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/fab5593662fa47f88e27f6b779c8adc4 : SUCCESS in 1h 09m 29s\n- metalsmith-integration-glance-centos9-uefi https://zuul.opendev.org/t/openstack/build/f4bc4020eb414ca2883c9e21cc78251e : SUCCESS in 1h 28m 10s (non-voting)\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/00e7be06f90e423d8bc2ea113f751df5 : SUCCESS in 39m 32s (non-voting)\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/4fe7c85252c449b59a01b5ec6c20ed06 : SUCCESS in 1h 11m 25s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/9bd3f5897de44058a6dd8309e7129acd : SUCCESS in 33m 34s\n- metal3-ironic-standalone-operator-functional https://zuul.opendev.org/t/openstack/build/1ac5c6deb56e4faaa01cb336fe093158 : SUCCESS in 39m 51s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/ccd503c9fba94fe0a5db513a304f3735 : SUCCESS in 20m 03s (non-voting)\n- ironic-inspector-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/40dd06c35548488c8de266caabd7e36f : SUCCESS in 43m 24s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/fb74c6a1a2504f0cbcd267329b06da9d : SUCCESS in 45m 16s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/fab05f73e2c547f3ac622a900cbb01a3 : SUCCESS in 47m 28s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/55b6a46e7f284815bccc2cb9f85bb7b1 : SUCCESS in 53m 42s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dd1bc0c7df91490fa9891700ca6694cb : SUCCESS in 38m 25s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/27f8135f9853475e9e0bd818024ce7cc : SUCCESS in 37m 25s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6df09d17f105d24b0e8d6a102e0a7da8e1d489c0","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-09-18 22:48:00.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"baec1e997b66ffbd4dbe80d020baad3288837153","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2025-09-19 12:05:42.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bf4c03f3649885cad62c18893b98d437b116f320","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2025-09-20 12:34:47.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"a2200fba390da1b519c39c0d12f9eccd68d0e827","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2025-09-20 12:34:59.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"59e8d85f063a71dd60fbb824d4fd160e58847e81","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-20 12:54:21.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9332bc4cc4be48c5a77f3783461ecd33\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/56a0439c8e1c46c1be98a80d10faddde : SUCCESS in 13m 38s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/286d4efd8e424b7c9fd54314dfdaa8fb : SUCCESS in 15m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e6d687ad10f145af94799bed2828997e : SUCCESS in 17m 46s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"092a73c54c245d6cf9cc527161dc16bf0998228a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-20 15:13:22.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2530031c029345d5aa87a00b116b57bf\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da4cbf2c6732455fafccf3c1d2f8eeea : SUCCESS in 11m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/429ab2e08aff4b7c9e9d1295cfd56a93 : SUCCESS in 8m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dad86a472d3f43d4afa11d25f9d664de : SUCCESS in 7m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0757b92ae87f44bdb5875a42170bc450 : SUCCESS in 9m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8ce4d95411da4e6a80d087fd9d5bd9d3 : SUCCESS in 13m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95669b1e1f9e4d439fd8f582c17b00ef : SUCCESS in 13m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c9023058a1e24bdfb12cebfb1047baf8 : SUCCESS in 19m 59s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/b359e3f11ab64aee8e9f49a46a993967 : SUCCESS in 5m 04s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/3132b5e2d00c4aa4a0c13bc2f7709718 : SUCCESS in 10m 01s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/f26b8cee3d1d4ba1842f4b09395d2b09 : SUCCESS in 9m 21s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/e106336c6fc24fb3a79a8a3737917a8f : SUCCESS in 25m 15s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/52cbc4a521cc4341a2c09f0ec84fe429 : SUCCESS in 1h 10m 57s (non-voting)\n- ironic-tempest-standalone-advanced https://zuul.opendev.org/t/openstack/build/eaf0347424354f4799d92f26ffa69de7 : SUCCESS in 56m 58s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/d3530b2f0ad546d5a25ca144a804ee2e : SUCCESS in 45m 50s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/578d8f7d85e04316bcc7d53b31ca7c2b : SUCCESS in 45m 25s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/530d2211aac640d59f9b0cd01fc1d649 : FAILURE in 25m 30s (non-voting)\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/125267518f97402b9e1d06aacc88471b : SUCCESS in 30m 42s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/bd4660b88b6d4357959352de1938a7ab : SUCCESS in 43m 17s\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/5382689d73f34ef8aac9a08bc191c9dc : SUCCESS in 1h 01m 54s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/470beaacdaaa445a842ff3cabc544536 : FAILURE in 2h 33m 24s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-shard https://zuul.opendev.org/t/openstack/build/36c6e7f7a99946bcbd7ba38728650b86 : SUCCESS in 58m 59s (non-voting)\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/e84e08ecbc3c4658be573098fc6b4c22 : SUCCESS in 49m 46s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/7dbd0d33776b4f1f95dfb9cfaeb5f501 : SUCCESS in 40m 13s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/7d47e25ef5c647d188be5365a0347e47 : SUCCESS in 35m 54s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/0190fb4c387a4e1b9fa8d6c40738c5fa : SUCCESS in 1h 06m 50s\n- metalsmith-integration-glance-centos9-uefi https://zuul.opendev.org/t/openstack/build/15bab059ff4e42c5874c100476bd99a2 : SUCCESS in 1h 36m 22s (non-voting)\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/3660ae99161642308cac1388c85f4d64 : SUCCESS in 1h 09m 58s (non-voting)\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/544a889d7dcf4033813e44862e736869 : SUCCESS in 1h 08m 14s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/89e58799a6724f128c90e3987e65a5f8 : SUCCESS in 38m 17s\n- metal3-ironic-standalone-operator-functional https://zuul.opendev.org/t/openstack/build/95e86c96d875433b94fa2ad5e055bfdf : SUCCESS in 39m 26s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/ee2522e0936f4d25ae76688b590391af : SUCCESS in 47m 27s (non-voting)\n- ironic-inspector-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/52b15c15b9214e7384e10642938e82e2 : SUCCESS in 32m 41s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/569dc1e0d8794e0db3bee8d8424f6610 : SUCCESS in 44m 14s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/ce78605701624945a0a559821eb4bf15 : SUCCESS in 46m 53s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/6c0994ebdcb14f75867dfbb2cebe9f8e : SUCCESS in 38m 16s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ee22169f9f5a4d0b967682933c0dd9b2 : SUCCESS in 32m 34s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b423c74d8a7a46b6835098bdfa7a438b : SUCCESS in 42m 12s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"cffafb6cd2f675cccaea2334248e299eb81bbda4","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2025-09-23 18:06:50.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":6},{"id":"5f074a2b8f3ae6adad73e2a85d2b0e789ef41f87","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2025-09-23 18:13:04.000000000","message":"Patch Set 6: Backport-Candidate+1 Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"455b7c1f4910c869eaaf6239959202d0ce9dcf65","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-09-24 07:21:40.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"8e3d2dc8952468e00b1be7f07e3cb0975ba52c13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 07:22:12.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"0e9a6b3a2ca456f6cead8a80d7b944312ab8eee2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 08:36:44.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7935451ef2a04b9d9733eac72eaeaef5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15d005bf215546278dd33a35b7edd26b : SUCCESS in 8m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ccc19dce8c4a41d5a888bfd40f1c7e27 : SUCCESS in 10m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a860850d8e9349bb97a151c85b9cb37b : SUCCESS in 7m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/980042149bab4132b804592d13590e6c : SUCCESS in 18m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8531507a7db43f29992d052cb767bd5 : SUCCESS in 20m 23s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/a42fb6434a9849258b5ea4ac97bdbf91 : SUCCESS in 5m 28s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/7c587aec0c7f4155a9047b933707a1ed : SUCCESS in 10m 51s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/6ba07b3fc263495a9011b4b24a843003 : SUCCESS in 36m 14s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/8c3087ea419443a19c274bdda0baf51b : SUCCESS in 32m 00s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/04dd06cafde54cd29de93b545a79cf79 : SUCCESS in 44m 27s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/3d5f039bd22e4464ad0212924d12064c : SUCCESS in 43m 38s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/10406f88bc3b4d1797f81df4b76bce59 : SUCCESS in 45m 13s\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/beef9d0a3c7f4177ace0a59268daa8bb : SUCCESS in 34m 40s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/9a3333f2845f4bab85e07c60c133c890 : SUCCESS in 42m 45s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/b1256c3fae344898a3c8faf75a8e5bce : SUCCESS in 51m 33s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/8f6e7176c622403ba0e78f92ae136820 : SUCCESS in 1h 10m 16s\n- metal3-integration https://zuul.opendev.org/t/openstack/build/24fabdcc3e8e4bb2946ed2b24ebc48c3 : SUCCESS in 33m 01s","accounts_in_message":[],"_revision_number":6},{"id":"b67dc043143c4682ffe9a25adda49d5f28030355","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 08:36:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"ee5c476a01ce7e33e694d0492c78162a3a69b220","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 08:37:46.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cfaac908a8a4fd2b7ff86e575560b40\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d3aeafaefeb4e9ab5c36e71a1d898e1 : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/114371e0602348a386b7e0a4eca50803 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"4b8a3733f6f179255ea98f404284b6b6c9372c28","revisions":{"8ef292656bc030eaa3ae04795ced8fc87cad6f12":{"kind":"REWORK","_number":1,"created":"2025-09-09 15:43:33.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/83/960283/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/83/960283/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/83/960283/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/83/960283/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/83/960283/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/83/960283/1"}}},"commit":{"parents":[{"commit":"7681e2216de0c92e4a857773263f19396016be20","subject":"Update release mappings for 32.0.x","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/7681e2216de0c92e4a857773263f19396016be20"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2025-09-09 15:41:39.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2025-09-09 15:42:55.000000000","tz":120},"subject":"OCI: accept both content types when requesting a manifest","message":"OCI: accept both content types when requesting a manifest\n\nQuay does not accept application/vnd.oci.image.index.v1+json, only\napplication/vnd.oci.image.manifest.v1+json, so indicate that we\nunderstand both.\n\nAlso fix get_blob_url to work with image URLs without oci://. This is\nnot possible to trigger through the API but ensures internal\nconsistency.\n\nChange-Id: Iba84bbe5da541700d20a445818a4a0d584f1eca8\nSigned-off-by: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/8ef292656bc030eaa3ae04795ced8fc87cad6f12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/8ef292656bc030eaa3ae04795ced8fc87cad6f12"}]},"branch":"refs/heads/master"},"c8c83382742ca789a4df3f327a6be10b27cf492c":{"kind":"REWORK","_number":2,"created":"2025-09-17 12:12:20.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/83/960283/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/83/960283/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/83/960283/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/83/960283/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/83/960283/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/83/960283/2"}}},"commit":{"parents":[{"commit":"7681e2216de0c92e4a857773263f19396016be20","subject":"Update release mappings for 32.0.x","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/7681e2216de0c92e4a857773263f19396016be20"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2025-09-09 15:41:39.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2025-09-17 12:12:13.000000000","tz":120},"subject":"OCI: accept both content types when requesting a manifest","message":"OCI: accept both content types when requesting a manifest\n\nQuay does not accept application/vnd.oci.image.index.v1+json, only\napplication/vnd.oci.image.manifest.v1+json, so indicate that we\nunderstand both.\n\nAlso fix get_blob_url to work with image URLs without oci://. This is\nnot possible to trigger through the API but ensures internal\nconsistency.\n\nAlso provide more specific error messages instead of piling everything\nunder the very generic ImageNotFound and provide more logging.\n\nChange-Id: Iba84bbe5da541700d20a445818a4a0d584f1eca8\nSigned-off-by: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/c8c83382742ca789a4df3f327a6be10b27cf492c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/c8c83382742ca789a4df3f327a6be10b27cf492c"}]},"branch":"refs/heads/master"},"03c17d6a42a27e59ed8c72a45d7331388065c0fa":{"kind":"REWORK","_number":3,"created":"2025-09-17 14:35:13.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/83/960283/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/83/960283/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/83/960283/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/83/960283/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/83/960283/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/83/960283/3"}}},"commit":{"parents":[{"commit":"7681e2216de0c92e4a857773263f19396016be20","subject":"Update release mappings for 32.0.x","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/7681e2216de0c92e4a857773263f19396016be20"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2025-09-09 15:41:39.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2025-09-17 14:35:03.000000000","tz":120},"subject":"OCI: accept both content types when requesting a manifest","message":"OCI: accept both content types when requesting a manifest\n\nQuay does not accept application/vnd.oci.image.index.v1+json, only\napplication/vnd.oci.image.manifest.v1+json, so indicate that we\nunderstand both.\n\nAlso fix get_blob_url to work with image URLs without oci://. This is\nnot possible to trigger through the API but ensures internal\nconsistency.\n\nAlso provide more specific error messages instead of piling everything\nunder the very generic ImageNotFound and provide more logging.\n\nChange-Id: Iba84bbe5da541700d20a445818a4a0d584f1eca8\nSigned-off-by: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/03c17d6a42a27e59ed8c72a45d7331388065c0fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/03c17d6a42a27e59ed8c72a45d7331388065c0fa"}]},"branch":"refs/heads/master"},"b89b98df9a6a88cd1536234a1fc348897395fb02":{"kind":"REWORK","_number":4,"created":"2025-09-17 17:19:49.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/83/960283/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/83/960283/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/83/960283/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/83/960283/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/83/960283/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/83/960283/4"}}},"commit":{"parents":[{"commit":"7681e2216de0c92e4a857773263f19396016be20","subject":"Update release mappings for 32.0.x","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/7681e2216de0c92e4a857773263f19396016be20"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2025-09-09 15:41:39.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2025-09-17 17:19:41.000000000","tz":120},"subject":"Fix OCI artifacts pointing to a single manifest","message":"Fix OCI artifacts pointing to a single manifest\n\nWhen using ORAS to upload a file to Quay, the tag points directly to\nthe manifest, not to the index of manifests. Currently, Ironic is not\ncapable of handling the former.\n\nFirst, GET API to a manifest does not accept\napplication/vnd.oci.image.index.v1+json, only\napplication/vnd.oci.image.manifest.v1+json, so indicate that we\nunderstand both.\n\nSecond, the logic in the OCI image service needs to be adjusted to this\ncase. If the index is a manifest, it is now treated the same as an index\nwith only that manifest.\n\nThird, the manifest\u0027s body does not contain its own digest. Instead, it\ncan be fetched from the docker-content-digest header, so get it and\nstore as a virtual field dockerContentDigest.\n\nAs part of the change, I had to refactor identify_specific_image since\nit exceeded the allowed complexity with my changes.\n\nAlso fix get_blob_url to work with image URLs without oci://. This is\nnot possible to trigger through the API but ensures internal\nconsistency.\n\nAlso provide more specific error messages instead of piling everything\nunder the very generic ImageNotFound and provide more logging.\n\nChange-Id: Iba84bbe5da541700d20a445818a4a0d584f1eca8\nSigned-off-by: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/b89b98df9a6a88cd1536234a1fc348897395fb02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/b89b98df9a6a88cd1536234a1fc348897395fb02"}]},"branch":"refs/heads/master"},"a609c3dd08786a25be3eab05e94ab8adb6c3822e":{"kind":"REWORK","_number":5,"created":"2025-09-18 11:19:51.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/83/960283/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/83/960283/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/83/960283/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/83/960283/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/83/960283/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/83/960283/5"}}},"commit":{"parents":[{"commit":"7681e2216de0c92e4a857773263f19396016be20","subject":"Update release mappings for 32.0.x","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/7681e2216de0c92e4a857773263f19396016be20"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2025-09-09 15:41:39.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2025-09-18 11:19:41.000000000","tz":120},"subject":"Fix OCI artifacts pointing to a single manifest","message":"Fix OCI artifacts pointing to a single manifest\n\nWhen using ORAS to upload a file to Quay, the tag points directly to\nthe manifest, not to the index of manifests. Currently, Ironic is not\ncapable of handling the former.\n\nFirst, GET API to a manifest does not accept\napplication/vnd.oci.image.index.v1+json, only\napplication/vnd.oci.image.manifest.v1+json, so indicate that we\nunderstand both.\n\nSecond, the logic in the OCI image service needs to be adjusted to this\ncase. If the index is a manifest, it is now treated the same as an index\nwith only that manifest.\n\nThird, the manifest\u0027s body does not contain its own digest. Instead, it\ncan be fetched from the docker-content-digest header, so get it and\nstore as a virtual field dockerContentDigest.\n\nAs part of the change, I had to refactor identify_specific_image since\nit exceeded the allowed complexity with my changes.\n\nAlso fix get_blob_url to work with image URLs without oci://. This is\nnot possible to trigger through the API but ensures internal\nconsistency.\n\nAlso provide more specific error messages instead of piling everything\nunder the very generic ImageNotFound and provide more logging.\n\nChange-Id: Iba84bbe5da541700d20a445818a4a0d584f1eca8\nSigned-off-by: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/a609c3dd08786a25be3eab05e94ab8adb6c3822e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/a609c3dd08786a25be3eab05e94ab8adb6c3822e"}]},"branch":"refs/heads/master"},"4b8a3733f6f179255ea98f404284b6b6c9372c28":{"kind":"REWORK","_number":6,"created":"2025-09-20 12:34:47.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/83/960283/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/83/960283/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/83/960283/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/83/960283/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/83/960283/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/83/960283/6"}}},"commit":{"parents":[{"commit":"7681e2216de0c92e4a857773263f19396016be20","subject":"Update release mappings for 32.0.x","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/7681e2216de0c92e4a857773263f19396016be20"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2025-09-09 15:41:39.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2025-09-20 12:34:37.000000000","tz":120},"subject":"Fix OCI artifacts pointing to a single manifest","message":"Fix OCI artifacts pointing to a single manifest\n\nWhen using ORAS to upload a file to Quay, the tag points directly to\nthe manifest, not to the index of manifests. Currently, Ironic is not\ncapable of handling the former.\n\nFirst, GET API to a manifest does not accept\napplication/vnd.oci.image.index.v1+json, only\napplication/vnd.oci.image.manifest.v1+json, so indicate that we\nunderstand both.\n\nSecond, the logic in the OCI image service needs to be adjusted to this\ncase. If the index is a manifest, it is now treated the same as an index\nwith only that manifest.\n\nThird, the manifest\u0027s body does not contain its own digest. Instead, it\ncan be fetched from the docker-content-digest header, so get it and\nstore as a virtual field dockerContentDigest.\n\nAs part of the change, I had to refactor identify_specific_image since\nit exceeded the allowed complexity with my changes.\n\nAlso fix get_blob_url to work with image URLs without oci://. This is\nnot possible to trigger through the API but ensures internal\nconsistency.\n\nAlso provide more specific error messages instead of piling everything\nunder the very generic ImageNotFound and provide more logging.\n\nChange-Id: Iba84bbe5da541700d20a445818a4a0d584f1eca8\nSigned-off-by: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4b8a3733f6f179255ea98f404284b6b6c9372c28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4b8a3733f6f179255ea98f404284b6b6c9372c28"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
