)]}'
{"id":"openstack%2Fironic~962172","triplet_id":"openstack%2Fironic~stable%2F2025.2~Iba84bbe5da541700d20a445818a4a0d584f1eca8","project":"openstack/ironic","branch":"stable/2025.2","topic":"oci-fixes-stable/2025.2","attention_set":{},"removed_from_attention_set":{"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2025-09-26 10:42:55.000000000","reason":"\u003cGERRIT_ACCOUNT_10239\u003e replied on the change","reason_account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}}},"hashtags":[],"change_id":"Iba84bbe5da541700d20a445818a4a0d584f1eca8","subject":"Fix OCI artifacts pointing to a single manifest","status":"MERGED","created":"2025-09-24 10:49:35.000000000","updated":"2025-09-26 14:22:21.000000000","submitted":"2025-09-26 14:20:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962172-oci-fixes-stable/2025.2","cherry_pick_of_change":960283,"cherry_pick_of_patch_set":6,"meta_rev_id":"cbdf703f807158193cdc69c87d747c1287180d72","_number":962172,"virtual_id_number":962172,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-26 14:20:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2025-09-25 12:44:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":1,"date":"2025-09-25 12:44:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-24 12:12:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-25 12:44:04.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"}],"messages":[{"id":"fe14c2bf294cc8ef112ba1478cfa9bbc3a9c6f4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2025-09-24 10:49:35.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a55a3c6202895a41d6fc5207d9e5a050ffd34012","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 12:12:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6ab510f2ce0344b99a71557d18846fc6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c70d7108bd89403083572ead6ab0ae3b : SUCCESS in 12m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/236e442216954555a620ad50693a2813 : SUCCESS in 7m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/171b078c38484178abb6ee635d6a2f2a : SUCCESS in 9m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/010efcfcc3c6457db1bc463072dc43ea : SUCCESS in 10m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f45b40e800a540898095092e96eca6c2 : SUCCESS in 9m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a69ea1231924570a13b3e8f4b2912ae : SUCCESS in 18m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2fa10b029464bf895eaaa34dc55b4e5 : SUCCESS in 19m 32s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/604524fd777b473c95d5025d855c46e0 : SUCCESS in 7m 09s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/01ac402732b14c5586ff7479b9983151 : SUCCESS in 10m 33s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/f83092927dd84a6b951d908dac853a55 : SUCCESS in 10m 53s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/ad590c61e5fd414ba0f4fa9f5336f9d9 : SUCCESS in 15m 33s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/762d6781ccb8407da5a37d8f91f876f1 : SUCCESS in 59m 10s (non-voting)\n- ironic-tempest-standalone-advanced https://zuul.opendev.org/t/openstack/build/e81a15cd19d8477882d9e5fb0ab1598c : SUCCESS in 32m 37s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/dd1ef554303e424fac6dec375baa76c2 : FAILURE in 29m 13s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/576bc0b54a4047d79618d9b3d96b98c3 : FAILURE in 13m 54s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/355a725ca2fb4873b0125483730e9245 : FAILURE in 28m 33s (non-voting)\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/bcd80338e9664ed78e385245fbb8026a : FAILURE in 12m 45s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/ee3a7e535f80410282173b81081c7f8e : FAILURE in 20m 18s\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/bb1423c054b14c938be184123ff70ca5 : FAILURE in 31m 27s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/940d0c3d09b84d09a607b8d92aaf9adb : FAILURE in 22m 40s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-shard https://zuul.opendev.org/t/openstack/build/0d62487b94754ec99f9fb7c800409e63 : FAILURE in 46m 24s (non-voting)\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/dc4f5131bf0944efbaff7bdeedea93c9 : FAILURE in 27m 52s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/1c34c5f895794ce8a8087e209eeeb952 : FAILURE in 29m 40s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/6b10c931e6cb4d01bf90c123c5961dc6 : FAILURE in 23m 02s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/6cc37a5744e04c12b084a4dc6664008a : FAILURE in 28m 41s\n- metalsmith-integration-glance-centos9-uefi https://zuul.opendev.org/t/openstack/build/906a13572ded40cf8efff928ba35cdb5 : FAILURE in 16m 16s (non-voting)\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/870d45394163451080839fccb8f21ce6 : SUCCESS in 1h 20m 19s (non-voting)\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/e6a8884fa41048b4901c617c325e201c : SUCCESS in 58m 08s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/4e918c86a937427c944b0bd09065e216 : FAILURE in 29m 14s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/ab997399c1714015bb63e6b711dcd96a : FAILURE in 20m 06s (non-voting)\n- ironic-inspector-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/64e9288be3b14462a97be8aaa4ddcacf : FAILURE in 22m 24s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/b7fed6235ab64e2c94c987103113fe59 : FAILURE in 27m 23s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/f6a9e88050694cc7b1dd3a5a2ddb6024 : FAILURE in 33m 19s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/ab8fc7ce0d7a4129b86f301dcab0ffce : SUCCESS in 53m 04s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/100bc852f20d4decb4cde51e5fae0bb9 : SUCCESS in 43m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3143fd393e4992831e60c00c46e422297f819573","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-09-25 12:44:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"987fffd46fbf12ff23559cb080fc5a765a5b0f94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 14:17:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/030e82317b694437bdf24b831ad35c09\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22a07d8503a34beab8d7c4544098278f : SUCCESS in 12m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f971ae3280e46c28dbbfcc16edc7f07 : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b20c85801af24a2bbb9af3ec148a7a0a : SUCCESS in 6m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/39b260fa2aab45e984755485d598c7e3 : SUCCESS in 6m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9b081abfbcee467dac2058d8b057da58 : SUCCESS in 7m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4c2b349e6fe4f79a88d9d463f4e1bcf : SUCCESS in 8m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d57cb009b93942819e05284a570dfe78 : SUCCESS in 10m 04s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/555220ef4be847fbad712ea79f4ce9fa : SUCCESS in 5m 20s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/b72ba3ed0e4d4365937e093ad627e5e3 : SUCCESS in 8m 36s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/ee5ef018a06a4453a0bc506c85c92ecc : SUCCESS in 9m 24s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/886dd637344943c8b6685766b741c361 : SUCCESS in 34m 18s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/17c5c9d772e647f983339ab15ca01aba : SUCCESS in 1h 11m 33s (non-voting)\n- ironic-tempest-standalone-advanced https://zuul.opendev.org/t/openstack/build/839fbb2e0de04db0ada60c2e2cb633b6 : SUCCESS in 1h 22m 34s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/570e0a5ad9c14d478dd9fb0158d383ee : FAILURE in 27m 24s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/9feb1521511f43f786015232b261af5b : FAILURE in 30m 26s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/e7c20770a6a84718bed48604a26b1836 : FAILURE in 25m 19s (non-voting)\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/bd78501262ae41dea07e4cd147fc6e06 : FAILURE in 28m 48s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/f391b94a77194ddbac0fb561d4b40438 : FAILURE in 28m 54s\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/09d0d901fb9047d093b4eaeef5d00f49 : FAILURE in 28m 21s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/21b9f9ca92e1454998924cda49984483 : FAILURE in 35m 06s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-shard https://zuul.opendev.org/t/openstack/build/09dadc8edca941d5bc355a448e9e13cb : FAILURE in 35m 34s (non-voting)\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/63285e3977144dc7a2cb32ac6e916d5f : FAILURE in 27m 12s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/8aceb188e0964df8a6a074947730674d : FAILURE in 28m 02s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/4c8e714f4736420d852a22dc2ae92eda : FAILURE in 29m 38s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/d8cd5b126abb46f981c914d1ebb3ad76 : FAILURE in 26m 11s\n- metalsmith-integration-glance-centos9-uefi https://zuul.opendev.org/t/openstack/build/4520e6fd7e084a489021a99029fc97bd : FAILURE in 11m 48s (non-voting)\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/524de9eb883548939766192d05262ce5 : SUCCESS in 51m 29s (non-voting)\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/6ebf870f56ac409dab6ccac92f5b12df : SUCCESS in 1h 26m 03s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/e1d967c9d01d4000a6f9db0fd2782233 : SUCCESS in 36m 03s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/6a974d7002884c7a85666c1c0b9a6ec9 : FAILURE in 22m 50s (non-voting)\n- ironic-inspector-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/512ae664f9f24ed5bd65eaa518d40b43 : FAILURE in 29m 45s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/f630406ee3a74cec860942739dd98922 : FAILURE in 21m 50s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/a37e98a39ed844afbece9fb6ab45aa1b : FAILURE in 13m 29s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/bff4099172094ebbbcf4b1606a929469 : SUCCESS in 47m 01s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69d7c04a2c5a4dfe92b70d7515cb171e : SUCCESS in 33m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"459252bf69755eb6698dab1e0553d1e4f5b67363","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2025-09-26 10:42:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f2946691fc72b31371e028bf9b7a9158df426e6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 13:00:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8dd313fc03a94619b61ee673e17a305d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3dd8f2c5817d46d490f64245f04ba491 : SUCCESS in 16m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb61b9c314074d8a8c3d91e0a9c54ef7 : SUCCESS in 8m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c94d3461e635474facbfbac472351a56 : SUCCESS in 10m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/948e564693494fef8e39890ae301cc21 : SUCCESS in 8m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c28bddec42ab419da0b84a7a6fcc7b90 : SUCCESS in 14m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56856953d09843c1b05ae7096dbef1d5 : SUCCESS in 17m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f14f103fc1344098b583884b69d6586 : SUCCESS in 17m 51s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/943578751ed941328f5a6056457ade6d : SUCCESS in 5m 53s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/ca0fed0203324ab88401072a4166f748 : SUCCESS in 9m 11s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/97d0d7cc955946e3b3c97c4916b089c4 : SUCCESS in 10m 23s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/140ec3ac3ce74f0386d12f64360a407e : SUCCESS in 37m 28s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/6c91e94d4dea4157add893e1e745cd22 : SUCCESS in 59m 51s (non-voting)\n- ironic-tempest-standalone-advanced https://zuul.opendev.org/t/openstack/build/315d580336bc4631a03d190be3f926d0 : SUCCESS in 1h 22m 23s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/fb77bc24380f4f89b9ca0f3e4a1c8dd7 : SUCCESS in 44m 14s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/a73da295a796493e9fa1eea324d878dd : SUCCESS in 47m 47s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/696f02cdb1f842a9b1f6210bf542447c : FAILURE in 26m 30s (non-voting)\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/3f155f2c201a4c398d244ff5382d8d1b : SUCCESS in 45m 41s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/497cc3e1ec864acea2ef84cee1146ae7 : SUCCESS in 49m 55s\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/236e6f516b7e4378b1f82f17baf8cf7e : SUCCESS in 1h 01m 02s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/736b755b558d4e1c8ff89a061ee8b674 : SUCCESS in 2h 12m 34s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-shard https://zuul.opendev.org/t/openstack/build/fa8dac9765b24a08950d8471e6bfe7ef : FAILURE in 2h 01m 16s (non-voting)\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/e2939475a34e4f26bf66ecb84fc9f48e : SUCCESS in 1h 13m 02s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/6816f1504a6c47ec9cb7e0ed07a41dc6 : SUCCESS in 43m 29s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/caeb8f0d932146099ae12268e636a5ce : SUCCESS in 44m 49s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/8af22b79517a4393bad277190ddeb7b3 : SUCCESS in 25m 17s\n- metalsmith-integration-glance-centos9-uefi https://zuul.opendev.org/t/openstack/build/7cdd62aa2d9d4dc2854b2fdc896af30d : FAILURE in 13m 29s (non-voting)\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/476508554bf14e74afa5f0e74cd85095 : SUCCESS in 1h 09m 04s (non-voting)\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/a8865a52256c4dd2a4b2d81a2f80d573 : SUCCESS in 58m 44s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/a84ab7f7982e4f4ea2cb1c0be50ab7c7 : SUCCESS in 35m 06s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/d40a738a4c7c4d37aa12c34ae08696dc : SUCCESS in 47m 53s (non-voting)\n- ironic-inspector-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/13d8f8b730de4918bfb806752aca888e : SUCCESS in 17m 50s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/94fbc0d45ce946cb94bc4124d5d149d5 : FAILURE in 14m 33s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/9b5bef0836f14c62aab46e5de7753f7c : SUCCESS in 46m 42s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/caae51e45b4644cb84fc6d2b25cf6d6b : SUCCESS in 56m 42s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8954ec2e3c7b4a7486f8a525cb9c8831 : SUCCESS in 40m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c9799b5b98d4a4314d6b5fb016103f7148a94644","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 13:01:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"20013ca17119ee51ecfcd18354f76f62e3a04441","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 14:20:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b213401483254c89a23fe2e7f6193b58\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/daea63d9d59447e5aab324ee182020dc : SUCCESS in 6m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2320884fb2964c498cd7748215488dde : SUCCESS in 9m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ae8ef214a0a2493089771146c7afc605 : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2995b3ce2eb64adda8e45ac678ddf950 : SUCCESS in 20m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1eaba86ab4cd47f5baa12f2f907c5ad5 : SUCCESS in 19m 32s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/adf72aca7f5a4902aac879d055a14ed4 : SUCCESS in 5m 24s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/38834a43646e4a79af31a4c4a9057fff : SUCCESS in 9m 23s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/90c11c7bc6d54827864ef6a480fa4f25 : SUCCESS in 42m 09s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/12006db233cb4080ac28a56ab0e8b46b : SUCCESS in 44m 17s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/1d4f1fc1cf3a4f409338791ba0af911f : SUCCESS in 47m 44s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/8bac71f750dd4500899efd3f6429baa7 : SUCCESS in 45m 18s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/2521e9544db84163882aa06b5a4376f3 : SUCCESS in 18m 34s\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/dc75503aa41b48f0b316a7bf585c8342 : SUCCESS in 1h 02m 58s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/94269a0c15f646598fb6d996f8605707 : SUCCESS in 27m 34s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/7d78b2b9d55e4b29b1023026ea665951 : SUCCESS in 47m 43s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/6ff02738de0f45db98a82c3967fc46d7 : SUCCESS in 1h 14m 51s\n- metal3-integration https://zuul.opendev.org/t/openstack/build/11ae0e70674648e8822385c4d975286d : SUCCESS in 36m 09s","accounts_in_message":[],"_revision_number":1},{"id":"5f8e71e725a7ad78fe811f9a2f501e547a6eaf44","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 14:20:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cbdf703f807158193cdc69c87d747c1287180d72","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 14:22:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92619387e23c4cc798bb4d5af45d0e29\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94be6759542c46f581e27c1215ffab4f : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d9c2ed36ea6492792cabea01abad022 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f97d77af6b8190ca2657518f9363e9ab5fa9a365","revisions":{"f97d77af6b8190ca2657518f9363e9ab5fa9a365":{"kind":"REWORK","_number":1,"created":"2025-09-24 10:49:35.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/72/962172/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/72/962172/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/72/962172/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/72/962172/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/72/962172/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/72/962172/1"}}},"commit":{"parents":[{"commit":"0c7660d126905d92c2b825076017abae7c02e961","subject":"fix: remove wait logic in pxe filter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/0c7660d126905d92c2b825076017abae7c02e961"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2025-09-09 15:41:39.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2025-09-24 10:49:35.000000000","tz":0},"subject":"Fix OCI artifacts pointing to a single manifest","message":"Fix OCI artifacts pointing to a single manifest\n\nWhen using ORAS to upload a file to Quay, the tag points directly to\nthe manifest, not to the index of manifests. Currently, Ironic is not\ncapable of handling the former.\n\nFirst, GET API to a manifest does not accept\napplication/vnd.oci.image.index.v1+json, only\napplication/vnd.oci.image.manifest.v1+json, so indicate that we\nunderstand both.\n\nSecond, the logic in the OCI image service needs to be adjusted to this\ncase. If the index is a manifest, it is now treated the same as an index\nwith only that manifest.\n\nThird, the manifest\u0027s body does not contain its own digest. Instead, it\ncan be fetched from the docker-content-digest header, so get it and\nstore as a virtual field dockerContentDigest.\n\nAs part of the change, I had to refactor identify_specific_image since\nit exceeded the allowed complexity with my changes.\n\nAlso fix get_blob_url to work with image URLs without oci://. This is\nnot possible to trigger through the API but ensures internal\nconsistency.\n\nAlso provide more specific error messages instead of piling everything\nunder the very generic ImageNotFound and provide more logging.\n\nChange-Id: Iba84bbe5da541700d20a445818a4a0d584f1eca8\nSigned-off-by: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\n(cherry picked from commit 4b8a3733f6f179255ea98f404284b6b6c9372c28)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/f97d77af6b8190ca2657518f9363e9ab5fa9a365"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/f97d77af6b8190ca2657518f9363e9ab5fa9a365"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
