)]}'
{"id":"openstack%2Fironic~962173","triplet_id":"openstack%2Fironic~bugfix%2F31.0~Iba84bbe5da541700d20a445818a4a0d584f1eca8","project":"openstack/ironic","branch":"bugfix/31.0","topic":"oci-fixes-bugfix/31.0","attention_set":{},"removed_from_attention_set":{"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2025-09-25 13:58:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iba84bbe5da541700d20a445818a4a0d584f1eca8","subject":"Fix OCI artifacts pointing to a single manifest","status":"MERGED","created":"2025-09-24 10:50:22.000000000","updated":"2025-09-25 13:59:59.000000000","submitted":"2025-09-25 13:58:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962173-oci-fixes-bugfix/31.0","cherry_pick_of_change":960283,"cherry_pick_of_patch_set":6,"meta_rev_id":"98ccca2ea3ff9b69145245280074e9469311522f","_number":962173,"virtual_id_number":962173,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-25 13:58:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2025-09-25 12:44:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":1,"date":"2025-09-25 12:44:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-24 11:02:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-25 12:44:17.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"}],"messages":[{"id":"ace85e4fb2fb2850def2c7835a07a0dcbb7a45cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2025-09-24 10:50:22.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8ba39d83cd62967faef3c94c57598bf6f194d39e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 11:02:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71df34409a224786a55855b60ec91d0f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c589d04b60d4ed7b887ac1c6792f263 : SUCCESS in 7m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02623c0a35a1459a88babc652b78b585 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fe01eee5d7c4238a664514c320059bc : SUCCESS in 9m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc57020d52d0477ebd92416eb40e58b8 : SUCCESS in 10m 20s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/81303b0b678c4e2ba6f960ace27973ad : SUCCESS in 6m 59s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/111bc1cf5e7d4d8ea3578f1097803940 : SUCCESS in 8m 33s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/4552447c5e1a4b4ab2ebc55c342adf99 : SUCCESS in 5m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"67dfbf0f17884fae8e44963e76480a07a4871709","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-09-25 12:44:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"042502f2ee26dd7cab8c2fa30120d986a14f8b67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 12:44:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"585e45f985dbcef142bd762b793192e53a9eef4c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 13:58:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4af68db4e56414eb189def1ba666da5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/972707949b954e08a0630a0988f2676e : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed0bae92eb9e438dbc523d1a9f37092d : SUCCESS in 16m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06acec94bd9f480ea69ac35bc9242d9a : SUCCESS in 21m 17s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/83c712fe2bb34daa8d2c70ec599871e8 : SUCCESS in 4m 54s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/da364836f038472288516f4fbf791821 : SUCCESS in 9m 43s","accounts_in_message":[],"_revision_number":1},{"id":"e91140a92b1f18bdac644ef9f1d263c4e77604f9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 13:58:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"98ccca2ea3ff9b69145245280074e9469311522f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 13:59:59.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fdaf4d0e452941dfbdbc3fd7310088e0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6c093c2b5fc4851bf5a2f60b6f70d7e : SUCCESS in 1m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b9a7c79246742da9831304d0049cd50 : FAILURE in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f165bfe306d229120d4f250772e2934ca194b3aa","revisions":{"f165bfe306d229120d4f250772e2934ca194b3aa":{"kind":"REWORK","_number":1,"created":"2025-09-24 10:50:22.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/73/962173/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/73/962173/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/73/962173/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/73/962173/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/73/962173/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/73/962173/1"}}},"commit":{"parents":[{"commit":"0e2600bf62cde471c5d4dabf7cfb50df60074ad1","subject":"Memoize calls to bcrypt.checkpw","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/0e2600bf62cde471c5d4dabf7cfb50df60074ad1"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2025-09-09 15:41:39.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2025-09-24 10:50:22.000000000","tz":0},"subject":"Fix OCI artifacts pointing to a single manifest","message":"Fix OCI artifacts pointing to a single manifest\n\nWhen using ORAS to upload a file to Quay, the tag points directly to\nthe manifest, not to the index of manifests. Currently, Ironic is not\ncapable of handling the former.\n\nFirst, GET API to a manifest does not accept\napplication/vnd.oci.image.index.v1+json, only\napplication/vnd.oci.image.manifest.v1+json, so indicate that we\nunderstand both.\n\nSecond, the logic in the OCI image service needs to be adjusted to this\ncase. If the index is a manifest, it is now treated the same as an index\nwith only that manifest.\n\nThird, the manifest\u0027s body does not contain its own digest. Instead, it\ncan be fetched from the docker-content-digest header, so get it and\nstore as a virtual field dockerContentDigest.\n\nAs part of the change, I had to refactor identify_specific_image since\nit exceeded the allowed complexity with my changes.\n\nAlso fix get_blob_url to work with image URLs without oci://. This is\nnot possible to trigger through the API but ensures internal\nconsistency.\n\nAlso provide more specific error messages instead of piling everything\nunder the very generic ImageNotFound and provide more logging.\n\nChange-Id: Iba84bbe5da541700d20a445818a4a0d584f1eca8\nSigned-off-by: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\n(cherry picked from commit 4b8a3733f6f179255ea98f404284b6b6c9372c28)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/f165bfe306d229120d4f250772e2934ca194b3aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/f165bfe306d229120d4f250772e2934ca194b3aa"}]},"branch":"refs/heads/bugfix/31.0"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
