)]}'
{"id":"openstack%2Fironic~962175","triplet_id":"openstack%2Fironic~stable%2F2025.1~Iba84bbe5da541700d20a445818a4a0d584f1eca8","project":"openstack/ironic","branch":"stable/2025.1","topic":"oci-fixes-stable/2025.1","attention_set":{},"removed_from_attention_set":{"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2025-09-25 13:58:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iba84bbe5da541700d20a445818a4a0d584f1eca8","subject":"Fix OCI artifacts pointing to a single manifest","status":"MERGED","created":"2025-09-24 10:51:27.000000000","updated":"2025-09-25 13:59:58.000000000","submitted":"2025-09-25 13:58:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962175-oci-fixes-stable/2025.1","cherry_pick_of_change":960283,"cherry_pick_of_patch_set":6,"meta_rev_id":"e73fb3184d4839c3b42a06f2a459cc1398274924","_number":962175,"virtual_id_number":962175,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-25 13:58:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2025-09-25 12:43:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":1,"date":"2025-09-25 12:43:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-24 11:17:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-24 13:13:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-25 12:43:44.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"}],"messages":[{"id":"084f4728ca49a26bfed809df4d9560f115e6283a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2025-09-24 10:51:27.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"fcce0dc88c7812c4ed0bfb706809a94895bccba7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 11:17:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec13c2bf665444469962d1c94928ea0a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4810ba3b55ea4ad6a764854ec6ec122a : SUCCESS in 16m 19s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/efdebaa11c644ba4a2aab7c889242fee : SUCCESS in 15m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e559d582df248e507a2acca7dc5a992114b2fe30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 13:13:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6340149142dd4d6e91609069fee6e4c7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e462d99b0644cb585c8e417d0500b0b : SUCCESS in 11m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff8ad6007c5f4c6a80aaa4ca43832997 : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d66cc9ef9fad4c728dc369ac14e0588a : SUCCESS in 5m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/54bc8950de8e482f8b5f23aa20e56bc4 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a14a3cd9fb645ea82ed439908a9190f : SUCCESS in 16m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4748a03d6719461faadbe9e94432a33e : SUCCESS in 17m 10s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/559be0ee260c449099235ff363c3df3f : SUCCESS in 6m 46s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/ffdddde6f591475baf817341879855c2 : SUCCESS in 10m 39s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/b36218263621494884798906b7e80f8a : SUCCESS in 9m 01s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/58a9b146134d4413bec8a63ce15c4ca0 : SUCCESS in 30m 47s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/b30fa53701084e30a8a87de87c0f8d8c : SUCCESS in 48m 59s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/5658fb077c7e48698b46ac8e8ad9a52b : SUCCESS in 39m 46s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/b298b918bda44d66a43506381d4f3e59 : SUCCESS in 44m 01s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/321e5cf4f04e4615a3c23bcb322c2b86 : SUCCESS in 22m 39s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/8f61d119aa2043c09e7069558b5bb422 : SUCCESS in 57m 57s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/538ec616178043bca428c5849aea0f03 : SUCCESS in 44m 50s\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/1c9fddfbbb4c4b7eaa00a714e2beca3a : SUCCESS in 37m 26s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/e6804952fc8a47af8cd5483516155d49 : SUCCESS in 2h 16m 04s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-shard https://zuul.opendev.org/t/openstack/build/a978b61812904098adb987ad1aaa1613 : SUCCESS in 1h 03m 12s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/cb6e7f3d0f7e4768bc0724e5ed8fdd5b : SUCCESS in 59m 47s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/4754add878aa49c8b8bd5175bea1bf7f : SUCCESS in 39m 40s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/f296aa73e6694ff4b3403b355103b1bb : SUCCESS in 42m 55s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/2ff8422e5bc34cf2803ddd14c2491ebd : SUCCESS in 1h 07m 44s\n- metalsmith-integration-glance-centos9-uefi https://zuul.opendev.org/t/openstack/build/deb502c72e9b4e34b1042683535f4877 : SUCCESS in 1h 32m 04s (non-voting)\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/00a5a57a02ce49c2a2d76edc202a5f20 : SUCCESS in 1h 02m 44s\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/ee23553fdad4475292b5b8d44be72353 : FAILURE in 36m 50s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/e28d3f908443406bb5b1e3745fc710a1 : SUCCESS in 43m 33s (non-voting)\n- ironic-inspector-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/fc8af40fdd3d48f0bd1174c7f7324abb : SUCCESS in 41m 36s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/06c023d470004fd2bfdf28ce331ca507 : FAILURE in 12m 23s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/99298efc0d6047f8babf02078a27a731 : SUCCESS in 58m 09s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/551f77fa7c10473395bdbfb225fe1511 : SUCCESS in 44m 34s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/247412dad74240c092e4fd6a0d3cdf9b : SUCCESS in 24m 29s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/807eb41aa2504c61b3e86864bb145244 : SUCCESS in 36m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cb3f1e8785960f05b71f7c399ee9def9c16cd0e5","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-09-25 12:43:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cc2b1856b357c7a7e70227da37ed3025f1bae4c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 12:44:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"009e600523e490aeb5850d32d4af296228cd1e09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 13:58:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7201e5e02ab74440b98f56357d6ac8f4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c552430824a47dcab5e4490cf929cd9 : SUCCESS in 7m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff88b34f07b94f949588ccaac6621799 : SUCCESS in 9m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b9d41cb90db44991a91f360664db18e5 : SUCCESS in 9m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0736f74b455430893948944b91759ad : SUCCESS in 15m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c93efd9843b417faec788118345fc2c : SUCCESS in 19m 39s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/1a0ebadc321744699bc1f0a011664e2f : SUCCESS in 5m 23s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/c5c38f22a08f4f589647dc8b71958204 : SUCCESS in 9m 45s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/f08ba602cfb648e8a72c954ee889707c : SUCCESS in 40m 18s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/6b94a1529b594df5970fac103d644417 : SUCCESS in 1h 08m 06s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/05678b079f8a47e28132dbc8a702b02f : SUCCESS in 41m 22s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/b07389a02627421aa52b9a4887187ab0 : SUCCESS in 42m 14s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/3b42eca764ca4dbfab3c8b50146240e9 : SUCCESS in 31m 39s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/b2bdb63736d14df587b96335b3e14a23 : SUCCESS in 44m 34s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/3578d78167f144f6b876c913848ef106 : SUCCESS in 36m 32s\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/23675ba77915485998781a1edfba67a8 : SUCCESS in 29m 38s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/5cad2b0d3b7c4f27b550ecd0a301035c : SUCCESS in 54m 25s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/16d3e64bda7e4a7d8ae2994dc26d7313 : SUCCESS in 35m 03s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/5a76830df7ff4f338312a2cd6128a513 : SUCCESS in 46m 08s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/4b680c964eac42e28f154d34a9147004 : SUCCESS in 1h 04m 29s","accounts_in_message":[],"_revision_number":1},{"id":"afd3ed5f07bc62a22bd68288aa296c74f5bd7bdf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 13:58:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e73fb3184d4839c3b42a06f2a459cc1398274924","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 13:59:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/339021da731943af98bfcfcb66e86945\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06b1e54caf4043dc860cb1ac4e53b323 : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1785b9ff523d49e7a285666806e2909a : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1394a234b1cd09afc82465a382e52f49dfd00c32","revisions":{"1394a234b1cd09afc82465a382e52f49dfd00c32":{"kind":"REWORK","_number":1,"created":"2025-09-24 10:51:27.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/75/962175/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/75/962175/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/75/962175/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/75/962175/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/75/962175/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/75/962175/1"}}},"commit":{"parents":[{"commit":"88245b5ec392c9acee3ef1c3b9772153aae6a81e","subject":"Fix inspection rules validation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/88245b5ec392c9acee3ef1c3b9772153aae6a81e"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2025-09-09 15:41:39.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2025-09-24 10:51:27.000000000","tz":0},"subject":"Fix OCI artifacts pointing to a single manifest","message":"Fix OCI artifacts pointing to a single manifest\n\nWhen using ORAS to upload a file to Quay, the tag points directly to\nthe manifest, not to the index of manifests. Currently, Ironic is not\ncapable of handling the former.\n\nFirst, GET API to a manifest does not accept\napplication/vnd.oci.image.index.v1+json, only\napplication/vnd.oci.image.manifest.v1+json, so indicate that we\nunderstand both.\n\nSecond, the logic in the OCI image service needs to be adjusted to this\ncase. If the index is a manifest, it is now treated the same as an index\nwith only that manifest.\n\nThird, the manifest\u0027s body does not contain its own digest. Instead, it\ncan be fetched from the docker-content-digest header, so get it and\nstore as a virtual field dockerContentDigest.\n\nAs part of the change, I had to refactor identify_specific_image since\nit exceeded the allowed complexity with my changes.\n\nAlso fix get_blob_url to work with image URLs without oci://. This is\nnot possible to trigger through the API but ensures internal\nconsistency.\n\nAlso provide more specific error messages instead of piling everything\nunder the very generic ImageNotFound and provide more logging.\n\nChange-Id: Iba84bbe5da541700d20a445818a4a0d584f1eca8\nSigned-off-by: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\n(cherry picked from commit 4b8a3733f6f179255ea98f404284b6b6c9372c28)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1394a234b1cd09afc82465a382e52f49dfd00c32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1394a234b1cd09afc82465a382e52f49dfd00c32"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
