)]}'
{"id":"openstack%2Fironic~962329","triplet_id":"openstack%2Fironic~bugfix%2F31.0~Id1eced2e0a30044b0da7dd5f4f2dedc50a5297b6","project":"openstack/ironic","branch":"bugfix/31.0","topic":"fix-image-cache-copy-bugfix/31.0","attention_set":{},"removed_from_attention_set":{"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2025-09-26 10:06:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id1eced2e0a30044b0da7dd5f4f2dedc50a5297b6","subject":"Fix cache image hard link between different file systems","status":"MERGED","created":"2025-09-26 09:18:24.000000000","updated":"2025-09-26 10:07:46.000000000","submitted":"2025-09-26 10:06:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962329-fix-image-cache-copy-bugfix/31.0","cherry_pick_of_change":962185,"cherry_pick_of_patch_set":3,"meta_rev_id":"571eef6b254f4b767e6a575f78a4586926f85457","_number":962329,"virtual_id_number":962329,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-26 10:06:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-09-26 09:21:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-09-26 09:21:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-26 09:21:01.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2025-09-26 09:21:59.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2025-09-26 09:47:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5e71505911ad5a610c01676b6fcd4ba9442c7ba1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-09-26 09:18:24.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"49ada0f2a924d14a0d36386da30587d0c8af7360","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2025-09-26 09:21:01.000000000","message":"Patch Set 1:\n\nhitachi fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-29-962329-1-fc.tar.gz: SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"ac2cb1c699c06c23a812cae3cd1416fc3acca55e","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2025-09-26 09:21:06.000000000","message":"Patch Set 1:\n\nhitachi iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-29-962329-1-iscsi.tar.gz: SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"dd6488ad34fade5fe79466c7ee6f31f6d723a2a3","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2025-09-26 09:21:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8ed8c313188a24ab746ee4d2875851efd593cdb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 09:47:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8679b6cc283b464eb792738bc756d506\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12aa758291734915b82179aefdc1298a : SUCCESS in 13m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e2eb17b4ee5492d96198ee0d6bda18d : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/818a54929d584721b1340fdd7d306d7a : SUCCESS in 16m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7b7cc4e62b04224aea3272f8fae0759 : SUCCESS in 20m 23s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/d69723eb00ed4334aaab58558dfa16c3 : SUCCESS in 3m 00s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/95f861aab5e544968ca36016c617dd43 : SUCCESS in 9m 22s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/09caeae1f32c44edb5f9997ddba88ed5 : SUCCESS in 9m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aa3f5d4a40bd2f2e7fcd8bc29d69eedb6edfa51b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 09:47:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1c6b6bf4770427cfa6b777276fc408e9e2f05195","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 10:06:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6ac99c53c9a495284fbb2ed4b518c90\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb62e201da084655aec8c13f11c30fb6 : SUCCESS in 7m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56a6755bae2b4d6482645f44f2b287df : SUCCESS in 14m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aea7eb95e079489bbc4563b02fb5f96b : SUCCESS in 16m 41s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/093feebee7254adcb6a2492380246861 : SUCCESS in 5m 53s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/db183f8dd7874e759c900a9bb6038a52 : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":1},{"id":"70fd20b28dd3b0df47f86494b5a282162b6f6096","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 10:06:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"571eef6b254f4b767e6a575f78a4586926f85457","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 10:07:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e92146fcbfd54a35a54d211c058841b0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/086ee35d589f41dd9bd0ef58525dc714 : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10fd36f435c54d8eb6902d997e0b4147 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b1f88d39ea158f975f6fe1f4a567adcc201b0e35","revisions":{"b1f88d39ea158f975f6fe1f4a567adcc201b0e35":{"kind":"REWORK","_number":1,"created":"2025-09-26 09:18:24.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/29/962329/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/29/962329/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/29/962329/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/29/962329/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/29/962329/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/29/962329/1"}}},"commit":{"parents":[{"commit":"f165bfe306d229120d4f250772e2934ca194b3aa","subject":"Fix OCI artifacts pointing to a single manifest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/f165bfe306d229120d4f250772e2934ca194b3aa"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-09-24 13:09:33.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-09-26 09:18:24.000000000","tz":0},"subject":"Fix cache image hard link between different file systems","message":"Fix cache image hard link between different file systems\n\nWhen caching an image between different file systems, the hard link\noperation would fail. This is fixed by falling back to a copy\noperation.\n\nAssisted-By: claude-4-sonnet\nChange-Id: Id1eced2e0a30044b0da7dd5f4f2dedc50a5297b6\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\n(cherry picked from commit 685dbeb356ad09f0e757e7a0fff32a0507d97f34)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/b1f88d39ea158f975f6fe1f4a567adcc201b0e35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/b1f88d39ea158f975f6fe1f4a567adcc201b0e35"}]},"branch":"refs/heads/bugfix/31.0"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
