)]}'
{"id":"openstack%2Fironic~965039","triplet_id":"openstack%2Fironic~master~I890d551122489e5a0b3162f08dbc10270968fb00","project":"openstack/ironic","branch":"master","attention_set":{},"removed_from_attention_set":{"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2025-10-28 07:26:39.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"I890d551122489e5a0b3162f08dbc10270968fb00","subject":"CI: make multinode runtime check sane","status":"MERGED","created":"2025-10-27 23:55:59.000000000","updated":"2025-10-28 07:29:27.000000000","submitted":"2025-10-28 07:26:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"965039","meta_rev_id":"dd5d88fd16e6265d1f11c2a97bb508e63fd85cdb","_number":965039,"virtual_id_number":965039,"owner":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-28 07:26:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":2,"date":"2025-10-28 03:35:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":2,"date":"2025-10-28 06:07:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":1,"date":"2025-10-28 06:07:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":1,"date":"2025-10-28 03:35:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-28 00:13:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-28 03:01:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-28 03:35:35.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"},{"updated":"2025-10-28 06:07:22.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"}],"messages":[{"id":"56db7036cba274bbbca6d5df126eb80ac4e47503","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2025-10-27 23:55:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d0a8fc46266242ffd5af2bbc0ce82dec73467083","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2025-10-27 23:56:09.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"16ffef70ba20930181b2fd9e016745ab557b8fda","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2025-10-27 23:56:18.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":2},{"id":"a8a8ca1b266d9c2b2e5f8493729c56df4857023e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 00:13:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84da14acec7f47b7bf231cff681d6128\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7c933296e9004ac7a52f7130d9252b05 : SUCCESS in 14m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/96ddd769838642419c84a114f6262907 : SUCCESS in 13m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0fa1e24317874e5abb1b9749d57bbf09 : SUCCESS in 12m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1d011bef78f211c81b36807775cffe65025fe03a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 03:01:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1cfb7f771091461880d53a81c47e307f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db10fea3997d4f4ea63c8cdc7c7f473b : SUCCESS in 12m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d55a15f6ae28462cae9f1610861803a9 : SUCCESS in 6m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5ae2d81a466a4c558b472a83f434dc67 : SUCCESS in 8m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/43ff4fd28ca243fcb086676cb5145eda : SUCCESS in 7m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98b533c7a28143299d7706f8940bd39c : SUCCESS in 16m 22s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/92010d9d4a2342e78a2a6d5b7fc92c1f : SUCCESS in 5m 11s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/fc04646765494ca7ae78693354aae56c : SUCCESS in 10m 52s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/b87623b608fe48bb9dea2ed7687444ec : SUCCESS in 9m 04s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/9fdd5b93ecf344dfbd6c32a9e2a577ee : SUCCESS in 38m 23s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/8f2d89b5562a4c1388cb96f98b65a687 : SUCCESS in 1h 00m 37s (non-voting)\n- ironic-tempest-standalone-advanced https://zuul.opendev.org/t/openstack/build/900b4e6efabd43d2a071c7c28dedef20 : SUCCESS in 51m 52s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/c7ceeb01f5b548979413b8facff256a7 : SUCCESS in 41m 24s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/51d35ff0bca34c2b914e88969a71ef97 : SUCCESS in 47m 54s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/6668344fb8f14dd797fd01962b466da7 : FAILURE in 16m 38s (non-voting)\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/3a17d7aa8452493094e5e71aa22034b4 : SUCCESS in 30m 55s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/1a0f011ceeac4ecc998f4fd2b5d5d1a4 : SUCCESS in 28m 24s (non-voting)\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/99dd50a0c52044f693572da29dfcabba : SUCCESS in 40m 16s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-multinode https://zuul.opendev.org/t/openstack/build/34a815546ed34f2489b35a9dcbf5703f : SUCCESS in 2h 51m 40s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-multinode-shard https://zuul.opendev.org/t/openstack/build/f3b6a693f6f24324816da31e52d3cad2 : SUCCESS in 1h 10m 53s (non-voting)\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/2d159705da94439bbe6654da6ec6fcba : SUCCESS in 45m 58s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/a16e43b6d7504166a35d9c03f8ebd4cf : SUCCESS in 19m 10s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/8745ed82129d4ea0b44bf4de9784627b : SUCCESS in 44m 24s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/326b237c84d6498f9854b9b7e570b96e : SUCCESS in 1h 08m 47s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/5abf5859e4914d23870b1f636e3594bd : SUCCESS in 1h 31m 18s (non-voting)\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/f3fec1946dee4feb94f238202dda1c33 : FAILURE in 48m 42s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/22a4fd0b01c746359c42478ad1fbbb01 : SUCCESS in 32m 58s\n- metal3-ironic-standalone-operator-functional https://zuul.opendev.org/t/openstack/build/140f1b1dcf0f402b9156c79903bc5d4f : SUCCESS in 35m 08s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/44761b084d1840ebb0c0c11c02698102 : FAILURE in 27m 04s (non-voting)\n- ironic-inspector-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/2329be92eed5417896e0bf862c74df02 : FAILURE in 14m 46s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/bbb65e497d344ae488d8ebe436d84724 : SUCCESS in 26m 53s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/308a7b6f346249a9bf5923098e8b9b0e : SUCCESS in 46m 46s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/a5149014353841539633ef8eceba25f9 : SUCCESS in 51m 57s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b844ee2180b449bcaa17705164fb0c3c : SUCCESS in 34m 37s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/224c634bb83149778e275671e001a44c : SUCCESS in 38m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"51bae77830903646d7d3f7a4dfba25e5fa3779fe","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2025-10-28 03:35:35.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e40a410c1c6688f1fe760288238011f8d51b1d92","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-10-28 06:07:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9cb2e60f0344def2604708e1989ff9a0bbc66cc8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 06:08:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5a6d5380bcc425d53dbe2e882dd3564a23bc6658","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 07:26:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc31249d8c8e49b9b3f30ad0e2960d83\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/527b39317dff4d409e541f24e95b9da0 : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4bb4add5e3074b36811263f77b9f25d3 : SUCCESS in 9m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b9911974fa63401a90ee24cbde5111c4 : SUCCESS in 9m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86fade9ed5dd497e868b941128aacb31 : SUCCESS in 17m 23s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/eb7d1623cd994b08a93b95145050e06e : SUCCESS in 3m 00s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/4a73ebd9d6114d9ab011fef361b6ad2b : SUCCESS in 11m 26s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/4e3e67dfeb3e4fccb6bd3f208cde009e : SUCCESS in 35m 45s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/f8af5cedbcaa46beae1ef49889bd5ae4 : SUCCESS in 47m 02s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/0d146c2fd664472aa862e6cc5b71064b : SUCCESS in 18m 12s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/9e1c31087b884bb480d8a126b991b230 : SUCCESS in 48m 43s\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/e326a01610224ef1a171466abdd331c1 : SUCCESS in 1h 13m 55s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/67036c70df684a08aa874985cf79a64d : SUCCESS in 20m 23s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/9b7d528b0b2b4d67bdbc55acc5455102 : SUCCESS in 52m 30s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/06b4a0aaf21d42b3be497d196b8e0de2 : SUCCESS in 1h 10m 38s\n- metal3-integration https://zuul.opendev.org/t/openstack/build/05b7d6d6caa4477799bff9758c2260e3 : SUCCESS in 31m 47s","accounts_in_message":[],"_revision_number":2},{"id":"be20eece9cfd1c42637c4e144df9d206adb99a7b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 07:26:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"dd5d88fd16e6265d1f11c2a97bb508e63fd85cdb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 07:29:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9800e618019d4a2bb6b1b8faed5c7b98\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/960d447aee7140d3b86c083539e2dcd3 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"571e1e86600b7741a9ec52efcded85b976d6787c","revisions":{"c322ae479b48cdece8da2e486c1076e567eaeca3":{"kind":"REWORK","_number":1,"created":"2025-10-27 23:55:59.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/39/965039/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/39/965039/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/965039/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/965039/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/965039/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/39/965039/1"}}},"commit":{"parents":[{"commit":"f11533e2fda656491b2e0a11cffa2c70a4db17f4","subject":"docs: add a \"i want a pony\" bit of text","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/f11533e2fda656491b2e0a11cffa2c70a4db17f4"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2025-10-27 23:54:15.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2025-10-27 23:54:15.000000000","tz":-420},"subject":"CI: make multinode runtime check sane","message":"CI: make multinode runtime check sane\n\nIts complicated, but basically because we run full size VMs which\ntake a while to boot, the multinode tests need a bit more than 600\nseconds to deploy a node. They can get there in just about that time\nand even sometimes beat the time window, but sometimes the job\ntimes out internally and kills the test run.\n\nThis changes the time to be 2000 seconds, which is more consistent\nacross other jobs. Independently, the defaults in the tempest plugin\nwill need to be made sane.\n\nChange-Id: I890d551122489e5a0b3162f08dbc10270968fb00\nSigned-off-by: Julia Kreger \u003cjuliaashleykreger@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/c322ae479b48cdece8da2e486c1076e567eaeca3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/c322ae479b48cdece8da2e486c1076e567eaeca3"}]},"branch":"refs/heads/master"},"571e1e86600b7741a9ec52efcded85b976d6787c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-10-27 23:56:09.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/39/965039/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/39/965039/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/965039/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/965039/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/965039/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/39/965039/2"}}},"commit":{"parents":[{"commit":"f5682839343bd6465a8e4a3404427cd10b9837d1","subject":"Merge \"Document instance_uuid JSON PATCH non-compliance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/f5682839343bd6465a8e4a3404427cd10b9837d1"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2025-10-27 23:54:15.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2025-10-27 23:56:09.000000000","tz":0},"subject":"CI: make multinode runtime check sane","message":"CI: make multinode runtime check sane\n\nIts complicated, but basically because we run full size VMs which\ntake a while to boot, the multinode tests need a bit more than 600\nseconds to deploy a node. They can get there in just about that time\nand even sometimes beat the time window, but sometimes the job\ntimes out internally and kills the test run.\n\nThis changes the time to be 2000 seconds, which is more consistent\nacross other jobs. Independently, the defaults in the tempest plugin\nwill need to be made sane.\n\nChange-Id: I890d551122489e5a0b3162f08dbc10270968fb00\nSigned-off-by: Julia Kreger \u003cjuliaashleykreger@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/571e1e86600b7741a9ec52efcded85b976d6787c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/571e1e86600b7741a9ec52efcded85b976d6787c"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
