)]}'
{"id":"openstack%2Fironic~968151","triplet_id":"openstack%2Fironic~master~I342ef72e0ad007fa6f5b72f634ee90ef30137446","project":"openstack/ironic","branch":"master","attention_set":{},"removed_from_attention_set":{"5890":{"account":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"last_update":"2025-11-25 01:02:52.000000000","reason":"\u003cGERRIT_ACCOUNT_5890\u003e replied on the change","reason_account":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}}},"hashtags":[],"change_id":"I342ef72e0ad007fa6f5b72f634ee90ef30137446","subject":"chore: disable pycodestyle and pyflakes double runs","status":"MERGED","created":"2025-11-23 20:06:43.000000000","updated":"2025-11-25 01:55:33.000000000","submitted":"2025-11-25 01:54:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"968151","meta_rev_id":"0dd213ad2a3c7209e4a0e150630e35db3a62b151","_number":968151,"virtual_id_number":968151,"owner":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-25 01:54:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":2,"date":"2025-11-24 15:55:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":1,"date":"2025-11-24 15:55:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-23 20:33:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-24 15:55:57.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"}],"messages":[{"id":"4400251f9bd8265f278bf73997132275366b82e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2025-11-23 20:06:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"33ccbf410838b15fc7979fd534ba55194a6536e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-23 20:33:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb9aa98e1cdf4cabb9f506505bebfe62\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c1049f5401b949b9a77e12f925533a71 : SUCCESS in 8m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4993444ba54405f96ce1ed2fe4dc5cc : SUCCESS in 8m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c99e19199ba4c02be09bcba591b6471 : SUCCESS in 5m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5e1b5e3c051b4e53ba4224275ebb2c69 : SUCCESS in 10m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e5de6e434d39495291590616b41f00b3 : SUCCESS in 12m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fc493b51c214a17accdae9fe456fb89 : SUCCESS in 13m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d201a03d687c4b69826292265644d125 : SUCCESS in 21m 14s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/f669cba14f04454d875977963891cf5f : SUCCESS in 5m 08s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/91c1baa70f674d66bfdf903539e0262c : SUCCESS in 11m 59s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/40e8e230bfcc47a4be9f10be97973502 : SUCCESS in 9m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c0dc07911a391a42b30d2a05e9f78a8c8d3d8dfc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-23 20:38:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c1de6737acc422a858f9bf577ac739a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9efe7596b96747d8b87786416a8b3668 : SUCCESS in 17m 09s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cea2fbf33e28412f9cd77cfee5378d4b : SUCCESS in 13m 34s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0959c6b13231448c95050280d4ec0f15 : SUCCESS in 11m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"914b6b562beb42d8e151db52a65771d875ab148e","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-11-24 15:55:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e04021e8556d74ad03b81ddaeece6c62d65aef2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 16:23:55.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/ironic/+/968150 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"2c8fd879a35cd312bfa0123599e499eabb7a14dc","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2025-11-25 01:02:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"906b40decce9eaa0be97b9ba56de81c2743c26e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 01:25:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ee5c3d2d77a43a79201f7b14971ae65\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a288c9e8f1ae45f2854f863f184be1f8 : SUCCESS in 7m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b068ade563234d0ca1aeb77c56ba83a1 : SUCCESS in 12m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d99213c0fbd743f6a262185017b0736d : SUCCESS in 6m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fa85c878363c4a66833d3549739ce207 : SUCCESS in 9m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b665c7aad3b248bd92cedef803176429 : SUCCESS in 11m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8fe41df53804a0d820b06a058bdfae0 : SUCCESS in 16m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1013162794245a8b6e366fefd205129 : SUCCESS in 15m 54s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/73d9500ce5e34245bddc40effe91a426 : SUCCESS in 3m 00s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/327ecdd3f9b146a99774387766c2479a : SUCCESS in 9m 21s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/506a6e091eb640dd9bdf05b362fe796e : SUCCESS in 9m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8a5007b9b41956db078b98eb27a57e703c6587b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 01:25:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"99d42a6a89da11247bc957cdba4b9f9761263017","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 01:42:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36d58648b82443f2ae4cfe16830e6fc6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7cb74271220a45449f02e35fde2982b6 : SUCCESS in 9m 19s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a4ee400eef2a4db8ba504cf9e6d52765 : SUCCESS in 9m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1f655625bb0e41df8bcb86e5ddbcc2cc : FAILURE in 35m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"07d7a682bbe13b088dff5d69e815616e82a6a356","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 01:54:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc12a9e9b3e340fc84e880ad860190ec\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/795964bf30704c00bf68f5c3b46d473a : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/424ec9b30ff54cb7847e70aa3a436e9f : SUCCESS in 6m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5402b02cd98a401e951a336eb54f6124 : SUCCESS in 9m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6a5a5316388f47d1a357cd81731b0e05 : SUCCESS in 11m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/badd2299e9a646559c0d5656c608f94e : SUCCESS in 18m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ad7d8314e0b43cfab57b9d4c54418c7 : SUCCESS in 21m 06s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/07c02e79f00248479fc44040326062bb : SUCCESS in 5m 27s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/62260720c0944f8ebed59a4b28a93109 : SUCCESS in 11m 10s","accounts_in_message":[],"_revision_number":1},{"id":"92f37220a518dd99606834e114c0fcf2e5d20191","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 01:54:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0dd213ad2a3c7209e4a0e150630e35db3a62b151","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 01:55:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/064f00608ded4ec9b8f96440d514bee9\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/a737092b57074e798ca3b06ca5b37cbf : SUCCESS in 38s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e343b8d1dd1d4335a821078f78da35b9 : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5da18f95a1fd489db20fea77026332ed : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"32bb6e3dff358fc73a43c1689f06392909481d30","revisions":{"32bb6e3dff358fc73a43c1689f06392909481d30":{"kind":"REWORK","_number":1,"created":"2025-11-23 20:06:43.000000000","uploader":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"ref":"refs/changes/51/968151/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/51/968151/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/51/968151/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/51/968151/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/51/968151/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/51/968151/1"}}},"commit":{"parents":[{"commit":"9ce7045970a69a1ac99ff668be8711667d1fbd5b","subject":"fix: remove unnecessary trailing slash","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/9ce7045970a69a1ac99ff668be8711667d1fbd5b"}]}],"author":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2025-11-23 19:36:49.000000000","tz":-360},"committer":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2025-11-23 20:06:21.000000000","tz":-360},"subject":"chore: disable pycodestyle and pyflakes double runs","message":"chore: disable pycodestyle and pyflakes double runs\n\nWe are running pycodestyle and pyflakes checks via ruff so we can\ndisable to double run via hacking using flake8. Remove ignores that were\nnot used or covered files that did not exist.\n\nChange-Id: I342ef72e0ad007fa6f5b72f634ee90ef30137446\nSigned-off-by: Doug Goldstein \u003ccardoe@cardoe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/32bb6e3dff358fc73a43c1689f06392909481d30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/32bb6e3dff358fc73a43c1689f06392909481d30"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
