)]}'
{"id":"openstack%2Fironic~968504","triplet_id":"openstack%2Fironic~stable%2F2025.2~Ibaf1486da9510fdad479523159797815e783e5f6","project":"openstack/ironic","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"5890":{"account":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"last_update":"2025-11-26 18:37:54.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"Ibaf1486da9510fdad479523159797815e783e5f6","subject":"fix: use the correct path to the image when deep image inspection is off","status":"MERGED","created":"2025-11-26 15:50:47.000000000","updated":"2025-11-26 18:39:05.000000000","submitted":"2025-11-26 18:37:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"968504","cherry_pick_of_change":968258,"cherry_pick_of_patch_set":2,"meta_rev_id":"d29c4cc6ce1ce25cdd06ee696aac4c7a61793384","_number":968504,"virtual_id_number":968504,"owner":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-26 18:37:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-11-26 17:35:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-11-26 17:35:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-26 17:25:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-26 17:35:48.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"1b5c239590c822ceb38eca3cecd7fc3affda04af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2025-11-26 15:50:47.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"3dc22eac63c35503d9143f9db44ee5262a780f25","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2025-11-26 15:51:13.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":1},{"id":"8e718360e360ea5b524298142cce8453274173de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 17:25:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c7082155dd9456c8e59395fb40a72cc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9916563f61cd4ec58baf0e88125ee317 : SUCCESS in 13m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd1eb8ddbb584d31b26387bdae420455 : SUCCESS in 6m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/83107a1691194b8e8216c22009e63cd8 : SUCCESS in 8m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dee50de9f7cf4d6b8c0e61b8eb1a3270 : SUCCESS in 7m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eb65ae6438c341ec9f15e1a3b71b232e : SUCCESS in 12m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b3855ad44774c169a24c18b1b442f65 : SUCCESS in 16m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef4c19a0910d41fe8a4d188134045342 : SUCCESS in 22m 55s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/9bcc106577e749cd9642eec337ff0006 : SUCCESS in 6m 59s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/aa97da25021d49e6b8de0a0a3b82d656 : SUCCESS in 12m 03s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/885d8e1c5b3247c7946dc898821d4c05 : SUCCESS in 10m 15s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/c18ef4c7670842d8919a1e966ba9aab9 : SUCCESS in 33m 13s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/0af780e9521d4f6eb1d77ce29020b9f7 : SUCCESS in 57m 02s (non-voting)\n- ironic-tempest-standalone-advanced https://zuul.opendev.org/t/openstack/build/8623314dd8784d8dacd563e2ab0fa8e2 : FAILURE in 35m 17s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/2ea27ad77d3c4e72aedb95668ee17174 : SUCCESS in 22m 16s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/059a69e88141430a8c2a34f7c72247ad : SUCCESS in 48m 58s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/22afac84268b4fcf9e8460692aa8070c : FAILURE in 27m 20s (non-voting)\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/452d9c0fe4a144f1a73bf18fb05d246a : SUCCESS in 37m 46s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/fb1e1e413ed145758433986cc4ddc21e : SUCCESS in 44m 43s\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/c9fb905ccfc847b298bf237960f9068d : SUCCESS in 1h 01m 33s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/188e736b461646bd86628fc1a72e8600 : FAILURE in 46m 15s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-shard https://zuul.opendev.org/t/openstack/build/1f6e5e8fa7d146c4918919add0495319 : SUCCESS in 1h 31m 20s (non-voting)\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/cae770b22e0e45af852c3aae680af188 : SUCCESS in 56m 41s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/a2a34f76eedd4e709ada9ea9e0607ac7 : SUCCESS in 36m 07s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/84b2aca0b93b45bfaab0992c7d4ad6b0 : SUCCESS in 42m 16s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/5130ceb2b4dd4d55b091de4225a6dfc0 : SUCCESS in 59m 24s\n- metalsmith-integration-glance-centos9-uefi https://zuul.opendev.org/t/openstack/build/10393aff8b524d3294944f65291b0347 : FAILURE in 16m 53s (non-voting)\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/786a5307994b4da98b5dfc462b082b91 : SUCCESS in 1h 14m 20s (non-voting)\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/426149c845ce4e8a979f07c53e841967 : SUCCESS in 1h 17m 34s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/49a1fe40547b454f87be29d783af2395 : SUCCESS in 46m 55s (non-voting)\n- ironic-inspector-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/4670ae96511947909982e13ea11e110e : SUCCESS in 43m 41s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/231750f03b1549e68698b3cf14ab5516 : FAILURE in 26m 35s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/0071f8b8f0c645cbb629b751e0a87d84 : SUCCESS in 25m 45s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/45e4bf71362f4d23a52ca3e57fbfb52e : SUCCESS in 36m 32s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6b9a845e2537480fbfb15935621117aa : SUCCESS in 41m 58s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e136ade394374cd88f58c5f7fd5f5aac : SUCCESS in 39m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"776f626cd69c1485c47dde2d186e6cc5fd83f238","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2025-11-26 17:35:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7869f9cdca34da54cccedef2ad30fec110504274","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 17:36:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"119f230426b5b41ba29fe1a6a4828b4536e0d17e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 18:37:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bd5b650055e4f7e866201d28b97846f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38b39401384842c6bf8b8a5174d85184 : SUCCESS in 5m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fe2d469544b84aefb5b6bdcf889204b1 : SUCCESS in 6m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6b467f36882f442a90c580b303483a99 : SUCCESS in 7m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4e4a05aaeec499e971a52043258bd36 : SUCCESS in 13m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8b8f2d99f1d4edb8c5a554a7c36b533 : SUCCESS in 21m 50s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/0a554a8221474eaf93786bb83a305b63 : SUCCESS in 3m 47s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/9e133b18c1ce45408bb45cf0097c929b : SUCCESS in 9m 56s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/eb9b01db8c8d4b68abb8ae9fade2e758 : SUCCESS in 16m 24s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/d1ea82fb851946d889a7c54128bb1b4e : SUCCESS in 21m 26s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/eedb5a59780b41aeb22fe0e8bdfe62a1 : SUCCESS in 47m 34s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/26fa9e9f444b4cfbbf7872cfaf6b88a7 : SUCCESS in 47m 39s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/77deb4ea0fb1414ab00d7a1e665a3d01 : SUCCESS in 48m 40s\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/369897d8f3fa480180952deaf041d61b : SUCCESS in 1h 00m 17s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/fbf6ee7db77a4ad3bc8f1a9eea3dc9b3 : SUCCESS in 19m 29s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/a6ee11fdd84542629545d4d98ec8fc31 : SUCCESS in 45m 05s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/e178bcbba19146c183e396ffadcf1c57 : SUCCESS in 32m 09s","accounts_in_message":[],"_revision_number":1},{"id":"f175f6ed80113ddcbc995c9e874c2c516203b9c7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 18:37:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d29c4cc6ce1ce25cdd06ee696aac4c7a61793384","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 18:39:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f34337339b4b407cb1d3e719603999aa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c67fb0b0f0e14127af6b8350c7cec79e : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e694e10e3d54405195625edf60ea0a86 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b423bf51a68557b1dd55f0a4ce942e560c9b245a","revisions":{"b423bf51a68557b1dd55f0a4ce942e560c9b245a":{"kind":"REWORK","_number":1,"created":"2025-11-26 15:50:47.000000000","uploader":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"ref":"refs/changes/04/968504/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/04/968504/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/04/968504/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/04/968504/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/04/968504/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/04/968504/1"}}},"commit":{"parents":[{"commit":"6c495a9c76668687093eba4c8eccdfe57cc93655","subject":"fix: glance image member lookup resulted in an empty list always","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/6c495a9c76668687093eba4c8eccdfe57cc93655"}]}],"author":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2025-11-24 21:04:57.000000000","tz":-360},"committer":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2025-11-26 15:50:47.000000000","tz":0},"subject":"fix: use the correct path to the image when deep image inspection is off","message":"fix: use the correct path to the image when deep image inspection is off\n\nWhen deep image inspection is disabled, the incorrect path was used to\ndetermine the image format of the file resulting in a no such file or\ndirectory exception which bubbled up and made it appear as if the cache\nwas missing.\n\nChange-Id: Ibaf1486da9510fdad479523159797815e783e5f6\nSigned-off-by: Doug Goldstein \u003ccardoe@cardoe.com\u003e\n(cherry picked from commit dca688bead95dd602cf55d1134bf9d73b7f92094)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/b423bf51a68557b1dd55f0a4ce942e560c9b245a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/b423bf51a68557b1dd55f0a4ce942e560c9b245a"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
