)]}'
{"id":"openstack%2Fironic~970004","triplet_id":"openstack%2Fironic~master~Ie9c8eb16747a7abd21fb08cf8d34226c22157d89","project":"openstack/ironic","branch":"master","attention_set":{},"removed_from_attention_set":{"36770":{"account":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"last_update":"2025-12-09 21:33:34.000000000","reason":"Change was submitted"},"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2025-12-09 20:01:44.000000000","reason":"\u003cGERRIT_ACCOUNT_10342\u003e replied on the change","reason_account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}}},"hashtags":["ironic-week-prio"],"change_id":"Ie9c8eb16747a7abd21fb08cf8d34226c22157d89","subject":"Trivial: Fix `instance_name` policy name","status":"MERGED","created":"2025-12-06 00:11:02.000000000","updated":"2025-12-09 21:34:42.000000000","submitted":"2025-12-09 21:33:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"970004","meta_rev_id":"6a55c12a8f5318d12d9f8c0875a3b3589dac5f17","_number":970004,"virtual_id_number":970004,"owner":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-09 21:33:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":2,"date":"2025-12-09 20:01:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":2,"date":"2025-12-09 20:05:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":1,"date":"2025-12-09 20:05:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-06 00:27:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-12-06 01:55:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-06 16:55:18.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"CC"},{"updated":"2025-12-09 20:01:44.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2025-12-09 20:05:25.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"REVIEWER"}],"messages":[{"id":"b42ac976a5858a8bb2e1c440fbbfcc0229190d44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2025-12-06 00:11:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5cdd7001495176f4b280ce85de9d546368e436be","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2025-12-06 00:12:00.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":1},{"id":"dd7aacf79b3b21a6a00c1435fc1abb62a5d80895","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-06 00:27:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6de7cd696c99467ebb2d8b906c223286\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7270c68c20bc467c9067d157a5488a1f : SUCCESS in 9m 34s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/979cb9cadff348ab93f8f281c6db4795 : SUCCESS in 11m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cd6d990cf52445d085096cb80c6d0c58 : SUCCESS in 14m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3e7d55f801aa9e77de2ff9918c4d3066a054f245","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-06 01:55:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe25f314ece047b3b0bc38f49eee5b87\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/926de036d2ec4d7eadefe2c6fbffdd8a : SUCCESS in 14m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f4648523e5e47c8b9af2f505a686f2f : SUCCESS in 6m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/daf7c6e11f2d497488aded26438864f0 : SUCCESS in 7m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/323b88a639f34b7bb8bf8f64a1d76cb6 : SUCCESS in 7m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/764cd22b22694d0392931a0995ace56c : SUCCESS in 8m 27s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/790a2be031e54d7da07321cbab96cec4 : SUCCESS in 6m 01s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/d5589acc81524cc2a62145b2d9e0eaf0 : SUCCESS in 10m 11s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/c1d4585714984872b785055107741783 : SUCCESS in 7m 42s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/c8deaec4b8cd41a9ba40559211e0c56c : SUCCESS in 25m 05s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/26eee28f46004989b980cf78c66cc54a : SUCCESS in 1h 00m 07s (non-voting)\n- ironic-tempest-standalone-advanced https://zuul.opendev.org/t/openstack/build/548a949cebc842878cdfbe03e17d84b5 : FAILURE in 24m 57s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/b68e14b672634f698a1457443788590b : SUCCESS in 46m 37s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/e854b0916edd4574a90162ad60561c93 : SUCCESS in 31m 57s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/01fa29499c4240078a7b2d9b32fb03be : SUCCESS in 32m 29s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/65bcea5e732643189ade8f31b9bb46a0 : SUCCESS in 22m 26s (non-voting)\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/e43467fcc71f46628a2cc17a35ca8560 : FAILURE in 1h 08m 59s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-multinode https://zuul.opendev.org/t/openstack/build/fa4ef568b3ec4ceabf949fe7cc37c830 : SUCCESS in 1h 33m 20s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-multinode-shard https://zuul.opendev.org/t/openstack/build/5f47afbc255a48da88a5b604cc7e7e5c : SUCCESS in 1h 34m 55s (non-voting)\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/f8042e42f407449eb00ccdf671a6a35a : SUCCESS in 1h 16m 10s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/1056967a6e254cf08d8ce6688f282c12 : SUCCESS in 41m 10s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/bd7e8cc9badf4efe8db035f3eb924ef4 : SUCCESS in 51m 00s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/6d5c91127d00417b8e82065ca6aef7b3 : SUCCESS in 1h 09m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/78783b33f76b45679ff7f28897139425 : SUCCESS in 1h 06m 04s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/7ab29f0cbba94db68fccfaf289eef530 : SUCCESS in 51m 12s (non-voting)\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/73dc9865f5d543c69e68cfeb38e37a6f : SUCCESS in 1h 11m 41s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/b2c7c3d062a04ff1a382350e19151619 : SUCCESS in 27m 03s\n- metal3-ironic-standalone-operator-functional https://zuul.opendev.org/t/openstack/build/0431fe014efe4606af595ed8543b8a14 : FAILURE in 8m 07s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/bafdba30cee547dbbae69ec9df4f6751 : SUCCESS in 40m 02s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/7719341f81bd418cba6976bab3a94a40 : SUCCESS in 28m 42s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/fcb32d751a764985b00ebf382e4a7308 : SUCCESS in 51m 21s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb19b2f3119b40009b844a99cb1bc003 : SUCCESS in 1h 05m 23s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/75568f022876455ebb04e36c633851c3 : SUCCESS in 52m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"02690abf77008901d543719cae72059aa284f159","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-12-06 16:55:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"54afe96fbfd64fe21dae8a3d05bb2512b434c677","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2025-12-06 16:58:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"44b5a53ea23532a04102dd873d0363e83bb9cce7","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2025-12-06 16:59:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4e9dde21396c5ae089d02661a83769a1aaafe733","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-12-09 20:01:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"64fdfc28454c8903dbaa80d446fef3353b4c8fb6","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-12-09 20:02:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d62670e183f82137fcfadce6452d6fbd3f883ce0","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2025-12-09 20:05:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c3d896675962fda648173f9a5b0ff5fc1d8577ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-09 20:05:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"65763aa2efdf9e41eabf11fc664f994961761beb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-09 21:33:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff797de3c95f492399d11c865ee0a1ca\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d218414f46cc46e396b40abbcf5d95b5 : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/81461ae305b6403598d67aea720660ec : SUCCESS in 9m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6d1281435af746818537b80290e60cbc : SUCCESS in 11m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d284fc670737419ab3c1b61d613a5c37 : SUCCESS in 17m 34s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/890db217530743d1b35d0f37c975b7f4 : SUCCESS in 3m 10s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/973b8672032949e59df720f14f04d06a : SUCCESS in 10m 01s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/e015710df6f9468497d581ff34c2624e : SUCCESS in 39m 58s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/176740d5a32f4309b8745c8009bdca36 : SUCCESS in 42m 55s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/136aa9766c5944aea44bad9f57974b24 : SUCCESS in 32m 50s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/cbe889c1e51c40a2b830d2b59dbd4204 : SUCCESS in 21m 28s\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/42cc3d48e8cc49df99cc684025041e03 : SUCCESS in 33m 03s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/7df1b564a18c47b6970e4ef949f5cbcb : SUCCESS in 41m 05s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/5075f7d300df45bb9917100bd7ae7c4f : SUCCESS in 51m 06s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/225a381442d34af99ec617ca7789d2ac : SUCCESS in 1h 22m 30s\n- metal3-integration https://zuul.opendev.org/t/openstack/build/9d587ee223c642ce9442a01cd1cf4e17 : SUCCESS in 25m 51s","accounts_in_message":[],"_revision_number":1},{"id":"392089929b7b5755a9dab8a7c0e1ff0f4ebb81e8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-09 21:33:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6a55c12a8f5318d12d9f8c0875a3b3589dac5f17","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-09 21:34:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/661eeb5666d34f97ade4ff96160ff236\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6da45353bc84a4a9a92356e295a941d : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2756b8e81bca169a864cb5001ad8a01218389ed3","revisions":{"2756b8e81bca169a864cb5001ad8a01218389ed3":{"kind":"REWORK","_number":1,"created":"2025-12-06 00:11:02.000000000","uploader":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"ref":"refs/changes/04/970004/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/04/970004/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/04/970004/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/04/970004/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/04/970004/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/04/970004/1"}}},"commit":{"parents":[{"commit":"d2d4c57ffa82484b7b902b42d0164903af1e5eb4","subject":"Merge \"Add a kubernetes provider for console container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/d2d4c57ffa82484b7b902b42d0164903af1e5eb4"}]}],"author":{"name":"Afonne-CID","email":"afonnepaulc@gmail.com","date":"2025-12-06 00:09:44.000000000","tz":60},"committer":{"name":"Afonne-CID","email":"afonnepaulc@gmail.com","date":"2025-12-06 00:09:44.000000000","tz":60},"subject":"Trivial: Fix `instance_name` policy name","message":"Trivial: Fix `instance_name` policy name\n\nChange-Id: Ie9c8eb16747a7abd21fb08cf8d34226c22157d89\nSigned-off-by: Afonne-CID \u003cafonnepaulc@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/2756b8e81bca169a864cb5001ad8a01218389ed3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/2756b8e81bca169a864cb5001ad8a01218389ed3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
