)]}'
{"id":"openstack%2Fironic~974776","triplet_id":"openstack%2Fironic~master~I35a80f710d6f79ac90649fcc5120a25acd1ea989","project":"openstack/ironic","branch":"master","topic":"clif/tbn/main_feature_branch","attention_set":{},"removed_from_attention_set":{"14228":{"account":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"last_update":"2026-02-06 20:06:23.000000000","reason":"\u003cGERRIT_ACCOUNT_14228\u003e replied on the change","reason_account":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"}}},"hashtags":[],"change_id":"I35a80f710d6f79ac90649fcc5120a25acd1ea989","subject":"Update TBN config file to improve trait structure","status":"MERGED","created":"2026-01-26 21:17:43.000000000","updated":"2026-02-07 00:08:11.000000000","submitted":"2026-02-07 00:05:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974776-clif/tbn/main_feature_branch","meta_rev_id":"15790aba773d953cc3517d8515b12088d09918e6","_number":974776,"virtual_id_number":974776,"owner":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-07 00:05:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"all":[{"value":2,"date":"2026-02-04 02:20:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"value":2,"date":"2026-02-04 18:15:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":2,"date":"2026-01-29 18:47:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":0,"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"value":1,"date":"2026-02-04 18:15:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-26 22:49:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-28 17:55:14.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2026-01-29 18:47:59.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2026-02-04 02:20:32.000000000","updated_by":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"reviewer":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"state":"REVIEWER"}],"messages":[{"id":"8207e3402ed4fb388fb7778d08661997a782c390","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2026-01-26 21:17:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aab1806afb9e1f8a9a5f58025d6c36ef48b38184","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 22:49:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27313abf54de400d970237cec4895e8a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f79d1c2295fe41f6acc34a67adfe4602 : SUCCESS in 20m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb8df0cd76b1412c8fb79827881a0b01 : SUCCESS in 6m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1365756033b0433993f8286a57d3631e : SUCCESS in 9m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/37cab83574324c0f93e3d0d8bb047913 : SUCCESS in 7m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cf0ce2b6f1740a79a5f0b572596be81 : SUCCESS in 17m 49s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/58d6142f9f434a3d9bc7d330254c51ed : SUCCESS in 7m 04s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/2f1280ab39a447d6a72d8cef19ef938d : SUCCESS in 9m 55s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/64d12be8c38a4634b277a40e5f2f1c7f : SUCCESS in 9m 23s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/0bf83e25f0bb427a84f1c7457db44775 : SUCCESS in 29m 59s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/406098c1d5d04342b173e6453069b201 : SUCCESS in 56m 15s (non-voting)\n- ironic-tempest-standalone-advanced https://zuul.opendev.org/t/openstack/build/f48819f139724cb996a7af52b65187c4 : SUCCESS in 1h 03m 10s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/f7b78a0536984d8e8bf53d620a5ed605 : SUCCESS in 42m 24s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/291cfbbc38ed409fb22d51fcde848c01 : SUCCESS in 23m 29s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/6987f6133bac4047930639704d06d4a9 : SUCCESS in 42m 58s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/924668a0e7da4697aa10de05e4639f58 : SUCCESS in 33m 01s (non-voting)\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/96b1ed420348411e92bb6f9c7d138d00 : SUCCESS in 32m 45s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-multinode https://zuul.opendev.org/t/openstack/build/d96e2ec237f046adbd628b70a740692d : SUCCESS in 1h 06m 35s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-multinode-shard https://zuul.opendev.org/t/openstack/build/1b36d8f1772e4d73b48c1663d799d518 : FAILURE in 56m 58s (non-voting)\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/f48ee705d0604060a48a169c6f318908 : SUCCESS in 1h 12m 34s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/6c52ac222e354be5be6548b876b2da1e : SUCCESS in 22m 18s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/14d5eddc5c114c8b86b690d94c732816 : SUCCESS in 23m 31s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/68d0f6d26de1445cadc0153f08b35505 : SUCCESS in 1h 05m 28s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/ec1e4521f0884a9692de07df9f814a5a : SUCCESS in 1h 17m 45s (non-voting)\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/836a17e51f944a0b9190f4ca5c03f81a : SUCCESS in 1h 10m 17s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/0bb006452dde41cfa8ce929491517213 : SUCCESS in 31m 08s\n- metal3-ironic-standalone-operator-functional https://zuul.opendev.org/t/openstack/build/3e572f3ce0eb4877bd367d5c89514896 : FAILURE in 8m 22s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/78e5e508d67d48c999f373f0c92dd810 : SUCCESS in 38m 45s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/8e32df5b17a643f39774089618876da9 : SUCCESS in 48m 47s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/e5f5e78c915f41fbb8b274262ff3f394 : SUCCESS in 39m 20s (non-voting)\n- bifrost-integration-on-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f9a7bf07f62428883e714f15f0bea4c : SUCCESS in 37m 41s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a502f7edcd904cfcaad61073d9dc2d9c : SUCCESS in 31m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aa37013c7bf48ab37c37c2e16f69b578eea6f1c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 23:50:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bc5ff232efc4bdd93abb0b55bc86090\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d75eb8445a144cf5aecfeb74a9d0172d : SUCCESS in 15m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f7e64a8a00f14edebba71d7c28bc3e48 : SUCCESS in 13m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/729e0c4b9f06463b8d50f1278826399b : RETRY_LIMIT in 4m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c4773d76930453ecadfba4b3eaa84609c24cfd89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2026-01-28 14:54:24.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"b05d8aa557e98b1db281a66fc254005d2f243d75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2026-01-28 17:29:54.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"dfe0e74f1e8c2fc77d64b2d67eb1bf5469cc1083","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2026-01-28 17:44:42.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"1db1ddeb1393c098deaaeb7fbab04a91d4a59bbb","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2026-01-28 17:55:14.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"5fa80613d21472fd84998030c270c48dff7feb9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 19:27:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3909c0f7f884a09b3d45c0df2d8d272\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8826eba5d38848aeb8111d730073651a : SUCCESS in 9m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19bff47624a74a2890f1de0935b3ccbc : SUCCESS in 6m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0e8ab9f28d1e43408fd85cdea4f99240 : SUCCESS in 8m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0051ef2ea39e4bbfb7f5ee605852d1ee : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96b25d89a03f4994b28d51525a4bbb9a : SUCCESS in 10m 25s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/154effde15c042eeb43864f828ef2d36 : SUCCESS in 6m 04s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/811f3efe8e2d4741a4945dbe9906da6b : SUCCESS in 9m 48s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/8ef30cc697a94564b7d2668291b1ab83 : SUCCESS in 8m 27s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/02158183b7994d6ead61813bfe5ba771 : SUCCESS in 38m 27s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/d2ac2bef973e42aa8f80b908b152df5f : FAILURE in 1h 30m 34s (non-voting)\n- ironic-tempest-standalone-advanced https://zuul.opendev.org/t/openstack/build/7f29190437144364a1228d04f261d70b : SUCCESS in 1h 16m 51s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/dcd4031927cf48c2a95e3f5c09cb37a3 : SUCCESS in 38m 43s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/fe0fb4a9fe724ae7b59307ff0e941c3e : SUCCESS in 47m 08s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/f9c80a72bbba41178bcc5645b278ab24 : SUCCESS in 49m 59s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/fb928765f23240fb8e1f09dfecce7a17 : SUCCESS in 47m 28s (non-voting)\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/d5f4f0814a4442518fab1b7054c626c5 : SUCCESS in 41m 54s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-multinode https://zuul.opendev.org/t/openstack/build/878b426728fb4e488f420778bd7e3616 : SUCCESS in 1h 34m 07s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-multinode-shard https://zuul.opendev.org/t/openstack/build/552210ebfcc042bf836b3b09f65fd6e6 : SUCCESS in 40m 47s (non-voting)\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/01ba47ca10144f37a8fc64b38e32d11e : SUCCESS in 1h 09m 30s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/32ca7e45367041139ed5a2972653a276 : SUCCESS in 39m 19s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/fd79b5e1f2b146baaa31f7008619c727 : SUCCESS in 26m 42s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/1cd1e2ea96d14f34a748b1eafe0eee55 : SUCCESS in 1h 00m 04s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/3a39a4d1e2e94771bd29999bc312e086 : FAILURE in 1h 05m 18s (non-voting)\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/de56359290c44b079474be6b0063a1d1 : SUCCESS in 1h 07m 45s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/34801c1998a34b7d91a162466e7da7b5 : SUCCESS in 34m 50s\n- metal3-ironic-standalone-operator-functional https://zuul.opendev.org/t/openstack/build/ad190eac2ca049fb8ee62f431563c76a : FAILURE in 8m 31s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/98ea18779a864f0baea74ce3260b5c75 : SUCCESS in 22m 39s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/65189f7e892e43ef9e9599647651a1dd : SUCCESS in 25m 17s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/1e74e3a0d26844da8d797196e8925b58 : SUCCESS in 26m 28s (non-voting)\n- bifrost-integration-on-ubuntu-noble https://zuul.opendev.org/t/openstack/build/257942b610b0400d98e883aef18333b3 : SUCCESS in 26m 43s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b8b068bc8a7343b0a61fdff467bf717c : SUCCESS in 27m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c7ef334db79e14d507d459c8996bc965d75d3919","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 21:23:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e51e2289e724dffbe42705fffb969a7\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2706cb12aa614a1b8ea4d1ff3484b0fa : SUCCESS in 14m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/736e9b5ad0ce4649a902d2a5669efb0c : SUCCESS in 13m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3a6ebb1575e14ff3b6aa304e4e0125ae : RETRY_LIMIT in 4m 32s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f2df2afe5e2246cef487dc0b5200b6085447e0e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2026-01-29 17:58:21.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"9e99366ada9a3f3659b4952745869c2b8dc946c7","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2026-01-29 18:47:59.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"610daec0ae756065954a2ea4980bbe440717adcb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 20:30:52.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85445f540b8a40b3b3d821a682264d4b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/48348ae398774ffc8b08c8949de377f7 : SUCCESS in 16m 22s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/63d955edf2bb4679a21db625089a116c : SUCCESS in 19m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d880312f12554e17b621e81a2e2bdad1 : RETRY_LIMIT in 4m 47s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a213e38ecfc2b07ec0a896d39ce76e170daa2012","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 21:08:08.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd388d4e01ac404ab3c4b73b34cb143f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93fb2e60843e409ba6531e6cfbf7eadc : SUCCESS in 14m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edc5d079161d4e2ea5b9f818486820fc : SUCCESS in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c7afaf1d38044089a466a97ca6ac0ff9 : SUCCESS in 8m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2058a061f35948eca829cfc524b46236 : SUCCESS in 7m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/950f4a8995704203ab100d98bff17468 : SUCCESS in 19m 04s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/09e282f74b9b4f91b2721361024243eb : SUCCESS in 5m 39s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/eed91fa9482342b0975dc53024de4501 : SUCCESS in 10m 57s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/f60b93aa5cc54d78854e984989315288 : SUCCESS in 8m 54s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/0326d84e9e9942818357d959b5102c95 : SUCCESS in 23m 01s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/511aae4e06ac48949b1e0b571ac070dd : NODE_FAILURE Node(set) request 9d9ed812366c49a7980e34b94f10f0e4 failed in 0s (non-voting)\n- ironic-tempest-standalone-advanced https://zuul.opendev.org/t/openstack/build/89ce4bb2cdb64eef8501df5d9cc77a7c : SUCCESS in 1h 05m 57s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/8ae9305128e94d89b9c3b570dbe2aac6 : SUCCESS in 52m 56s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/6ac8fe942b8a4f658271fa11025e7891 : SUCCESS in 50m 02s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/7630870476f8468db7a553d33cd2d4c3 : SUCCESS in 27m 19s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/8d9082cdc6dd4d999ab26d7ca0b19267 : SUCCESS in 52m 41s (non-voting)\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/e891100fddeb42ac901382a8401a7f4e : SUCCESS in 1h 20m 26s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-multinode https://zuul.opendev.org/t/openstack/build/0db28308c4a04f35bfc35d573ea28022 : FAILURE in 3h 00m 16s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-multinode-shard https://zuul.opendev.org/t/openstack/build/17ab52c353a54cc6b6a595f7c260ca38 : SUCCESS in 1h 11m 27s (non-voting)\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/8025ae652609482c829e325990346251 : SUCCESS in 1h 16m 14s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/4f41722e90774c988b423975cd1102aa : SUCCESS in 42m 57s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/b7afda07ece54639b165401f049a3288 : SUCCESS in 1h 07m 27s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/87bbc52db30a4c36b409a329ff5fb27a : SUCCESS in 1h 18m 25s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/d5916cf81ff249789cb702b97463b546 : SUCCESS in 49m 29s (non-voting)\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/ee767770a8764e888fd4969f640d8a13 : SUCCESS in 1h 12m 30s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/2b56ed50ad0a401885c15a877a863402 : NODE_FAILURE Node(set) request 7513e6c2e5c24e8e9fe9f1fe904e0d71 failed in 0s\n- metal3-ironic-standalone-operator-functional https://zuul.opendev.org/t/openstack/build/ee4787a8bc69493298033f000c5ef574 : FAILURE in 7m 33s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/831aa9179ccc44df952bf6c862b5d6b8 : SUCCESS in 28m 50s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/e64386a3a1f5451da9f86544d635fa69 : SUCCESS in 57m 23s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/0799b508ec0348dc9003bf585dd5a585 : SUCCESS in 47m 53s (non-voting)\n- bifrost-integration-on-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bea818095d664013bc44e8cc10ba7004 : SUCCESS in 51m 51s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e9641924ffb4028a362cb9a912f9db2 : SUCCESS in 55m 19s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"af183297a0c9f9447f3eb30b913174b6ea6ff979","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2026-02-04 02:20:32.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"665cdc7e73fdabcfc91342d8a6eca61a5f079684","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2026-02-04 18:15:26.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a7323c718f672d5672d0b45f179b88b31f28e52e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 19:05:40.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1dc85b90b164104885b397369f3cad8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/734f40b85efd4074837b7ad54a1d79f7 : SUCCESS in 16m 06s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f6795eb3d21a498a89534053cb149993 : SUCCESS in 12m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/74b46f0b59984a69a2035b558dc570b1 : RETRY_LIMIT in 6m 05s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ee9a2a05ff5ac8f523ee2bc28b3d552a2fdcc38f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 20:50:35.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a8c983342574880ba093c56bf0093c2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c3a1981807954aecae1ddfc4cf160024 : SUCCESS in 11m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1b9b89d36b546ea80b75568eaffec68 : SUCCESS in 7m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/60af25b8585f470683fd990d69413f83 : SUCCESS in 8m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c43ff4b16eb84a98a89de039405fb43d : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/012e9d9bca1c4eb69f7f8d2b6bc21b3b : SUCCESS in 20m 21s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/ac6eb0d43beb4d93b1e872435940f86a : SUCCESS in 4m 09s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/0c6d571f8c3047d5acf156240c947b36 : SUCCESS in 9m 44s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/3901d8e024b6403895c06706a92fe654 : SUCCESS in 7m 57s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/7df56edd4d7b4493bd327bf6e6783365 : SUCCESS in 38m 23s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/1388d43c35974cb78bccc546ec8158b9 : SUCCESS in 1h 04m 45s (non-voting)\n- ironic-tempest-standalone-advanced https://zuul.opendev.org/t/openstack/build/c9ea30e7a0374134a36036316ce5ee0a : SUCCESS in 1h 32m 17s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/0f79985ec9674d8db209436924c015d1 : SUCCESS in 49m 28s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/05284b5d8ce540caa6310f428a49fb0e : SUCCESS in 47m 12s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/100fc17ef52342889e1d4aae2655b88b : SUCCESS in 46m 23s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/b710339e5ee24c88a47f656b3bb6f7d2 : SUCCESS in 49m 52s (non-voting)\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/12a4c1073ed042d1bee48bc67659a38e : SUCCESS in 1h 00m 02s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-multinode https://zuul.opendev.org/t/openstack/build/6cac14a9a3c441e2baddb0ba03537cdb : SUCCESS in 2h 27m 16s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-multinode-shard https://zuul.opendev.org/t/openstack/build/06a76097604c484e9756194cd1edbb67 : FAILURE in 34m 43s (non-voting)\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/a5901d1442d04b2a8710af79ccb21f2c : SUCCESS in 34m 29s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/f1d99366f60f4b1c964c794a490e412c : SUCCESS in 44m 24s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/931d76e720a340e7b1f02be0d0b75d8f : SUCCESS in 59m 32s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/5c79cb00a8f9409d83af66bb9c2ae6ba : SUCCESS in 1h 06m 39s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/0aad888ca2804b5aa30f12388362242a : SUCCESS in 1h 32m 10s (non-voting)\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/3277e49fa95c46a890a45f119e70c201 : SUCCESS in 1h 11m 51s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/e79dc4d5dc57436aa426e20ad10ab5a7 : SUCCESS in 31m 49s\n- metal3-ironic-standalone-operator-functional https://zuul.opendev.org/t/openstack/build/fc474a36678440ce8c01fdcf81407109 : FAILURE in 7m 54s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/1d96c903b1964ba1bfec05059df63a58 : SUCCESS in 46m 23s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/e8d4cc50b90142288e33d2f50b559945 : SUCCESS in 50m 24s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/1a2ff194599f40f0b4c1cb8fa7691264 : SUCCESS in 49m 10s (non-voting)\n- bifrost-integration-on-ubuntu-noble https://zuul.opendev.org/t/openstack/build/659f305d3f03428aa85af4caffecf278 : SUCCESS in 56m 00s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4edc146e6be74fb5a5e3b1ce587fdd62 : SUCCESS in 27m 56s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"652ca754cb5d8e5f704ca52f57ae73b2cf2a5daf","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2026-02-06 20:06:23.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3967d1edc27afa3a161d55f19a48b161a5e91ec9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 22:43:52.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"3e1e8fb8438982e17e9af87b90f08426f0f06ba2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-07 00:05:56.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe47dd8dd94e4a62a6eb1af81abf5414\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6804e1bc5fb244a1a5e6915a6e25baa6 : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/79625f81464d4aabb84ea68e281861c1 : SUCCESS in 7m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/25ecad900b7f4b53a38186ed2f9e475f : SUCCESS in 8m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bc53840831142eaaab7b8bff12c6742 : SUCCESS in 18m 27s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/1bdfbe92b9824db09e87e1b032bbc6c1 : SUCCESS in 4m 54s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/79aa1a05932546368088e1f02e7c1d48 : SUCCESS in 8m 58s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/6a77d5cb06df4d19a76f18a9b4c35049 : SUCCESS in 39m 17s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/38bb8cb3cca44a4e9f9c3bed7b7f54d8 : SUCCESS in 40m 28s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/42d3fb9f7314451dad02c4893c8300a3 : SUCCESS in 19m 58s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/ae349dec0c7143f08923ab7efbb08cfa : SUCCESS in 22m 04s\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/818cba6911434a4ea289f0e21f4a646a : SUCCESS in 42m 33s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/f156f9c787784c7da6930e8646a5b9ac : SUCCESS in 19m 24s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/31e52605b12e4198929845ff51583dbc : SUCCESS in 20m 45s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/a9d6fe2474934df0a65ce5ab78c20fdd : SUCCESS in 42m 19s\n- metal3-integration https://zuul.opendev.org/t/openstack/build/583ad8d3db1f4bbbb3d6188b2d59229b : SUCCESS in 31m 24s","accounts_in_message":[],"_revision_number":5},{"id":"1ce156793d20d369bdd69b12f1b10837be4d1243","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-07 00:05:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"15790aba773d953cc3517d8515b12088d09918e6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-07 00:08:11.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8382255c1d1480698fd24908f98bd6f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0eb8f4a121a4d5b80244addc91c256e : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"d1f2364091ff0f0a6a1ffac5cb51a618169c6e2a","revisions":{"384a90ce2cc52f3ce376f077461eb877e5031e38":{"kind":"REWORK","_number":1,"created":"2026-01-26 21:17:43.000000000","uploader":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"ref":"refs/changes/76/974776/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/76/974776/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/76/974776/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/76/974776/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/76/974776/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/76/974776/1"}}},"commit":{"parents":[{"commit":"71d633c6dad82ee9ed6735b45962a11aca7d42a8","subject":"Filter out NoMatch actions in _vif_attach_tbn","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/71d633c6dad82ee9ed6735b45962a11aca7d42a8"}]}],"author":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2026-01-26 15:34:13.000000000","tz":-360},"committer":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2026-01-26 21:17:19.000000000","tz":-360},"subject":"Update TBN config file to improve trait structure","message":"Update TBN config file to improve trait structure\n\nAdds an explicit \u0027actions\u0027 member which is the list of actions belonging\nto a trait.\n\nAlso add an \u0027order\u0027 member to aid in more explicit ordering of traits if\nnecessary.\n\nChange-Id: I35a80f710d6f79ac90649fcc5120a25acd1ea989\nSigned-off-by: Clif Houck \u003cme@clifhouck.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/384a90ce2cc52f3ce376f077461eb877e5031e38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/384a90ce2cc52f3ce376f077461eb877e5031e38"}]},"branch":"refs/heads/master"},"c42134c544ca209a82631f152bcf2e4f13a2fa87":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-01-28 14:54:24.000000000","uploader":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"ref":"refs/changes/76/974776/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/76/974776/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/76/974776/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/76/974776/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/76/974776/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/76/974776/2"}}},"commit":{"parents":[{"commit":"051e13dece2f08a513c955b5dcff7458a76c1c7f","subject":"Filter out NoMatch actions in _vif_attach_tbn","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/051e13dece2f08a513c955b5dcff7458a76c1c7f"}]}],"author":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2026-01-26 15:34:13.000000000","tz":-360},"committer":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2026-01-28 14:37:59.000000000","tz":-360},"subject":"Update TBN config file to improve trait structure","message":"Update TBN config file to improve trait structure\n\nAdds an explicit \u0027actions\u0027 member which is the list of actions belonging\nto a trait.\n\nAlso add an \u0027order\u0027 member to aid in more explicit ordering of traits if\nnecessary.\n\nChange-Id: I35a80f710d6f79ac90649fcc5120a25acd1ea989\nSigned-off-by: Clif Houck \u003cme@clifhouck.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/c42134c544ca209a82631f152bcf2e4f13a2fa87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/c42134c544ca209a82631f152bcf2e4f13a2fa87"}]},"branch":"refs/heads/master"},"90436bf53d31f24aa0d1b0c34f4692b8f363df36":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-01-28 17:29:54.000000000","uploader":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"ref":"refs/changes/76/974776/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/76/974776/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/76/974776/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/76/974776/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/76/974776/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/76/974776/3"}}},"commit":{"parents":[{"commit":"43c7b32de85f3469567cb1150029217132a6ade0","subject":"Filter out NoMatch actions in _vif_attach_tbn","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/43c7b32de85f3469567cb1150029217132a6ade0"}]}],"author":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2026-01-26 15:34:13.000000000","tz":-360},"committer":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2026-01-28 17:29:45.000000000","tz":-360},"subject":"Update TBN config file to improve trait structure","message":"Update TBN config file to improve trait structure\n\nAdds an explicit \u0027actions\u0027 member which is the list of actions belonging\nto a trait.\n\nAlso add an \u0027order\u0027 member to aid in more explicit ordering of traits if\nnecessary.\n\nChange-Id: I35a80f710d6f79ac90649fcc5120a25acd1ea989\nSigned-off-by: Clif Houck \u003cme@clifhouck.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/90436bf53d31f24aa0d1b0c34f4692b8f363df36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/90436bf53d31f24aa0d1b0c34f4692b8f363df36"}]},"branch":"refs/heads/master"},"7dd73ece62425b0d7f8c01d3437c3f9cfbf9024f":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-01-28 17:44:42.000000000","uploader":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"ref":"refs/changes/76/974776/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/76/974776/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/76/974776/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/76/974776/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/76/974776/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/76/974776/4"}}},"commit":{"parents":[{"commit":"8433d764f006d8d6fa5175bbc50dd3b60a308a2d","subject":"Filter out NoMatch actions in _vif_attach_tbn","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/8433d764f006d8d6fa5175bbc50dd3b60a308a2d"}]}],"author":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2026-01-26 15:34:13.000000000","tz":-360},"committer":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2026-01-28 17:44:35.000000000","tz":-360},"subject":"Update TBN config file to improve trait structure","message":"Update TBN config file to improve trait structure\n\nAdds an explicit \u0027actions\u0027 member which is the list of actions belonging\nto a trait.\n\nAlso add an \u0027order\u0027 member to aid in more explicit ordering of traits if\nnecessary.\n\nChange-Id: I35a80f710d6f79ac90649fcc5120a25acd1ea989\nSigned-off-by: Clif Houck \u003cme@clifhouck.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/7dd73ece62425b0d7f8c01d3437c3f9cfbf9024f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/7dd73ece62425b0d7f8c01d3437c3f9cfbf9024f"}]},"branch":"refs/heads/master"},"d1f2364091ff0f0a6a1ffac5cb51a618169c6e2a":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-01-29 17:58:21.000000000","uploader":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"ref":"refs/changes/76/974776/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/76/974776/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/76/974776/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/76/974776/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/76/974776/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/76/974776/5"}}},"commit":{"parents":[{"commit":"e9b8e8367215943823b1da90e7f20b03156fe62c","subject":"Filter out NoMatch actions in _vif_attach_tbn","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/e9b8e8367215943823b1da90e7f20b03156fe62c"}]}],"author":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2026-01-26 15:34:13.000000000","tz":-360},"committer":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2026-01-29 17:44:21.000000000","tz":-360},"subject":"Update TBN config file to improve trait structure","message":"Update TBN config file to improve trait structure\n\nAdds an explicit \u0027actions\u0027 member which is the list of actions belonging\nto a trait.\n\nAlso add an \u0027order\u0027 member to aid in more explicit ordering of traits if\nnecessary.\n\nChange-Id: I35a80f710d6f79ac90649fcc5120a25acd1ea989\nSigned-off-by: Clif Houck \u003cme@clifhouck.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/d1f2364091ff0f0a6a1ffac5cb51a618169c6e2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/d1f2364091ff0f0a6a1ffac5cb51a618169c6e2a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
