)]}'
{"id":"openstack%2Fironic~975365","triplet_id":"openstack%2Fironic~master~Idbed1770c79b2f84bb27c8545395a0391a7b4b9c","project":"openstack/ironic","branch":"master","topic":"artifact-detect","attention_set":{},"removed_from_attention_set":{"32177":{"account":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"last_update":"2026-02-17 06:10:24.000000000","reason":"Change was submitted"},"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2026-02-17 06:10:24.000000000","reason":"Change was submitted"},"4571":{"account":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"last_update":"2026-02-17 06:10:24.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"Idbed1770c79b2f84bb27c8545395a0391a7b4b9c","subject":"Add bootc deploy to autodetect default","status":"MERGED","created":"2026-02-01 21:36:35.000000000","updated":"2026-02-17 06:13:42.000000000","submitted":"2026-02-17 06:10:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"975365-artifact-detect","meta_rev_id":"b0ddc65744a47da42099834923dff2980505c206","_number":975365,"virtual_id_number":975365,"owner":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-17 06:10:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":0,"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},{"value":2,"date":"2026-02-17 03:56:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":2,"date":"2026-02-13 12:20:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":0,"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},{"value":1,"date":"2026-02-17 03:56:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-01 22:01:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-02 00:08:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-06 15:56:00.000000000","updated_by":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"reviewer":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"state":"REVIEWER"},{"updated":"2026-02-08 21:17:26.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2026-02-13 12:20:14.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2026-02-17 03:56:51.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"}],"messages":[{"id":"c3a74b07fc2e06922a9a27546b75a2e5437aeafe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2026-02-01 21:36:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"748ca1490c6f26b07c0adf97e962e914afe34c24","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-01 22:01:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93eb96afffad42f990901126c9e199c3\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4641a01f4b404094b7e2a3f63765cde3 : SUCCESS in 13m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c3a652c94dae4853b800d164b2722333 : SUCCESS in 15m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b4f9d0c91241483fa20288022f9e61e6 : RETRY_LIMIT in 4m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ae427830759f15b7578cc531a2bb804893ac5720","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 00:08:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d87e57cb0e1344aba82ec3ce7c38a8d6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad4744d980d14f11a9d3860ea6fe233d : SUCCESS in 15m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d36a67406c724daa92dae2b5d52ded38 : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e4c0adc5073e4c159b4dc20d423f0aad : SUCCESS in 9m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fcf536c007c345adac80fd7281ad1a4f : SUCCESS in 8m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdbec5815428469c8613b3a6450c2634 : SUCCESS in 18m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a9a8fbc3fd94aa38f6a851dd7dc8be4 : FAILURE in 20m 10s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/58ec72df6ec848929187178d41492445 : SUCCESS in 7m 07s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/1e9a6ea49409429c963b5a36fdc29910 : SUCCESS in 5m 15s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/f8ae50eb49104268a1df75620e6375b5 : SUCCESS in 7m 00s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/f86c78faa3a24d7bb391dd931a4fac12 : SUCCESS in 26m 26s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/6271aa9e155c4a1a9d76b4d76056b6d3 : SUCCESS in 59m 01s (non-voting)\n- ironic-tempest-standalone-advanced https://zuul.opendev.org/t/openstack/build/5b337a8e116b4180bc0365f1cff73344 : SUCCESS in 1h 08m 24s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/e4eeba99a6884afdae892bde352229f9 : SUCCESS in 43m 54s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/64a005abc5864766855bca3f405e26e5 : SUCCESS in 47m 34s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/8b32f65e76ac4b168d044d5676b9eea1 : SUCCESS in 47m 53s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/b8a81eb969e94ba586d5be40c889bc20 : SUCCESS in 42m 42s (non-voting)\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/befc36eff6b74785833fd44418367262 : SUCCESS in 42m 21s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-multinode https://zuul.opendev.org/t/openstack/build/4d03f84b445145edbc6348b3ce3ff11b : SUCCESS in 2h 18m 45s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-multinode-shard https://zuul.opendev.org/t/openstack/build/a5e7963575804424bc32f101f3b58f1f : SUCCESS in 1h 34m 30s (non-voting)\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/07128225d80d47599627af34c71ace9f : SUCCESS in 58m 46s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/de86e2a361724890baca93cea546cafc : SUCCESS in 41m 09s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/71259d28e77d4802bb84cf8741995ed0 : SUCCESS in 46m 57s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/632323a4e56b4328aefed4a498809aec : SUCCESS in 1h 11m 28s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/7a9a107dc08243f6a5bf6fd8aa630922 : SUCCESS in 1h 07m 25s (non-voting)\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/ea26c9c520754f55adbb7ffd4a296f93 : SUCCESS in 1h 17m 59s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/35b17e68d08843e5b7c21a8a7933d4ff : SUCCESS in 29m 27s\n- metal3-ironic-standalone-operator-functional https://zuul.opendev.org/t/openstack/build/090017c97f39467bbcb27fbd2050f1cd : FAILURE in 8m 05s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/b89e17b72ccc4f8e8c02bbf2451dec4c : SUCCESS in 41m 05s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/c7cda54f75d14969a5f61a6678506663 : SUCCESS in 28m 20s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/7f89320a977e4643acd205acd11ba60a : SUCCESS in 56m 05s (non-voting)\n- bifrost-integration-on-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3058b14419c04888b25b3ce85ef11e12 : SUCCESS in 47m 10s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7d8f657ad391490f991d5e62c27bc704 : SUCCESS in 46m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9997fefbb7abaa9e06861259511fd021f91f75d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2026-02-03 00:27:14.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c159219abbf380516523d28d849a9d08e778a1c6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 00:49:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71262766d47f4854b30d2b44ac8d72ff\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/578f5c77a0784a4a8e5722b2fbe78f7b : SUCCESS in 14m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/02dcfcf8b537487591d9ee28848e6f78 : SUCCESS in 14m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/25cc743dbf494eff8accb6f188395f27 : RETRY_LIMIT in 4m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1bd50e21b38391f2983601e7458b014c51dd51b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 02:08:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e5b2bf2aa1d4a8da4ef04722bb37901\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ca6df36c41340d7b5df120613f70b2f : SUCCESS in 9m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/207a3b994e0348c88c5ca246a3383408 : SUCCESS in 6m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1d296381c8444c4caeddec235ccc72de : SUCCESS in 8m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/71948a073648442abe73545d8a201bd1 : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9e7e36d50694c35ab67f90c3f7b5bdd : SUCCESS in 17m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/16b4f2b0a193476684da926b62ad68ad : SUCCESS in 22m 25s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/7dc2863c64c24d3c81c5108e8cacef3f : SUCCESS in 2m 47s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/fc9d581247354a67ad8f87c31d7bb7cc : SUCCESS in 10m 23s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/952d845ad1b2406d843a78685237e8b0 : SUCCESS in 7m 22s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/fa752e97d8b041928ca4058c7786700a : SUCCESS in 19m 28s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/516da2da27ae428391572684833bb1f0 : SUCCESS in 53m 34s (non-voting)\n- ironic-tempest-standalone-advanced https://zuul.opendev.org/t/openstack/build/816b318ad8d74385b9e3103e0fed963d : SUCCESS in 54m 35s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/94048f8df96a474cbddc0958c86f6897 : SUCCESS in 42m 46s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/a8155de8305c49dd89559f851141e317 : SUCCESS in 37m 07s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/a36e410702334675a9d042cee1faa4c8 : SUCCESS in 38m 15s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/c0acfc7dafb94d7da48e8d148d8914cb : SUCCESS in 49m 46s (non-voting)\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/4c4b66f7a80442fea37a9c5dcf089ac4 : SUCCESS in 28m 02s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-multinode https://zuul.opendev.org/t/openstack/build/bf51ed6ea1fb41e49662535447e9ab51 : SUCCESS in 1h 37m 15s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-multinode-shard https://zuul.opendev.org/t/openstack/build/cbec14880033497fa902206870e92b71 : SUCCESS in 1h 29m 17s (non-voting)\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/726e5ee4dc9c419ebed4c19eb470c26e : SUCCESS in 41m 50s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/19a217f05e1d4b019cf2e6215695557c : SUCCESS in 18m 26s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/cc9a3a091cd947c88dbb92fbec23510b : SUCCESS in 23m 13s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/aa6a9df344844fb997a3c6f130260087 : SUCCESS in 45m 45s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/6c779a4878c44a5c9bc5fc676bb5ce5e : SUCCESS in 1h 29m 11s (non-voting)\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/76bc12b3eecb472bbe7f523ce05db2e9 : SUCCESS in 41m 24s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/4ba0672ea45d49f2b3d6dc85b67793e6 : SUCCESS in 29m 26s\n- metal3-ironic-standalone-operator-functional https://zuul.opendev.org/t/openstack/build/4408dc380d7645d2825a868cd9c0693f : FAILURE in 7m 43s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/2af8ac87ac884a1fab40697f2a411b18 : SUCCESS in 40m 38s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/b8b18cb7234a43dab137657cf0c25972 : SUCCESS in 51m 47s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/25baec02fcb446aca53c54bc5dd6fe2e : SUCCESS in 59m 38s (non-voting)\n- bifrost-integration-on-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5af83baed0564cebbd25f0a6cda85aeb : SUCCESS in 48m 11s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4a36ae8c31184509aa9649135f6a90c0 : SUCCESS in 53m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2cf200c14d5d5a64338ea92b91d517082982b3c7","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2026-02-03 19:40:10.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":2},{"id":"11801ace2c6756679eb4c6733deda1aeaa9a027c","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2026-02-06 15:56:00.000000000","message":"Patch Set 2: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d70a0789de8610653e42ca4ca8ce097fe130f3ef","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2026-02-08 21:17:26.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0f78da58dd70d5f0a7b2fa46edc8725e41185d25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2026-02-11 12:46:48.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"b13a8b9e1b32bf2bf12aa55bf0a8f64225b511eb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 14:01:40.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4964ec6ea4094c2a8d10f5a6e0bdd6f9\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bca63b67888b425b973316a1316c6cdd : SUCCESS in 14m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d885da37463b444ba39215ed1b21849e : SUCCESS in 9m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a120975341304cd3acd3b202ede3a71d : RETRY_LIMIT in 4m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cdf3a1111b62c82c20379119dd3dfcb237ed6256","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 15:02:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95dad11ad4184e6d94d0e32bfe6b476e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/88013d1ce3d943a5a56979ed1132dacb : SUCCESS in 17m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bd8937f841b459cbb2cf8e3209b77d0 : SUCCESS in 7m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b562037eb1ec4e18988343a2a8c3f8ed : SUCCESS in 9m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fb83ab25cb6f4ca5a83943ba2d45ab26 : SUCCESS in 7m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b17f4a2076954f5f8d086a796072c931 : SUCCESS in 17m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/552d4589cafe4b72b56dfb2f9cd15dc4 : SUCCESS in 20m 07s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/f622fca3703b41d682a6bce113ca6603 : SUCCESS in 5m 19s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/30aef732cb414ce6ac8314c7ad886b2a : SUCCESS in 8m 42s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/9daeee2a4f99474b9e4a915c31d6313c : SUCCESS in 8m 42s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/943194911a94499784bad15283a5a504 : SUCCESS in 40m 24s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/b546f9dcc3904267ad183bebf3bb1957 : SUCCESS in 1h 00m 49s (non-voting)\n- ironic-tempest-standalone-advanced https://zuul.opendev.org/t/openstack/build/e3ea149f93a14a388d144dc80b0af702 : SUCCESS in 1h 20m 49s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/8fd8b580bcf4486dae0bafe0bd41b00b : SUCCESS in 45m 16s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/4671b10cac684a508497a751de6acb77 : SUCCESS in 52m 50s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/a53a69949a7b487490dc48b1e895fb0f : SUCCESS in 44m 44s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/83f9cae61fbc4699b0ddce0d41322243 : SUCCESS in 47m 02s (non-voting)\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/2fc7eddb7383455385421df4f4ca3f81 : SUCCESS in 1h 01m 39s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-multinode https://zuul.opendev.org/t/openstack/build/4955f82770754e309c00144b8cca3361 : SUCCESS in 2h 08m 30s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-multinode-shard https://zuul.opendev.org/t/openstack/build/1e1c9789cad54346b2e494163d7ca3e9 : SUCCESS in 1h 25m 17s (non-voting)\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/54e999d32e1d4d61b4918758c778f7cf : SUCCESS in 33m 44s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/b64f8b0149e445b591a86859344a8c7b : SUCCESS in 21m 34s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/e148bc889a354efa9df879090f34c226 : SUCCESS in 1h 01m 34s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/6ed6cb99f3344211ae2d04e632e80b20 : SUCCESS in 1h 17m 44s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/19304d8ddbea4408a2b563fd749cc6dc : FAILURE in 1h 02m 46s (non-voting)\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/b0e3bc47615141dbab210489b09b8233 : SUCCESS in 2h 06m 54s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/a085d6868e804e51b98b26b1269038aa : SUCCESS in 34m 38s\n- metal3-ironic-standalone-operator-functional https://zuul.opendev.org/t/openstack/build/92ee3c0a9a274a779315137ad18dfdb7 : FAILURE in 8m 11s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/ee9f643b62ac43ab956bc3c1c1caed6c : SUCCESS in 20m 45s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/3ec37835d783470988b941c078e83308 : SUCCESS in 34m 06s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/5e78293842f944afbaafa5bdc91533b0 : SUCCESS in 53m 36s (non-voting)\n- bifrost-integration-on-ubuntu-noble https://zuul.opendev.org/t/openstack/build/26a8eff3b8c64e8285f7797174eff441 : SUCCESS in 42m 56s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/152d4d624c1244f992001b830366fd7b : SUCCESS in 28m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"322484be8b4eddbf83e8bf1426c7950a3d7292a6","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2026-02-12 10:11:18.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"03ba87cf8d28c04da9678360915eb52e162db3ea","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2026-02-13 12:20:14.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d147dd6647470bbcdee09d98979ac78f164ac2ef","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-02-17 03:56:51.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ab68082e5d22adab87a6fb45285a5675dbef0012","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 03:58:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9972642b3c85c18582be88a6aaa6188fe3a5af37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 06:10:23.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9cb870c38f4459caacd8d6db35e6db1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83adc7d268ec489eb11ddd7aeb0d500d : SUCCESS in 8m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e581d7d521254df091ea80a83acebf72 : SUCCESS in 9m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e9799fb8cc124d7887389084babdc9ed : SUCCESS in 8m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88310b7c0872487391623f04fd7e3a5e : SUCCESS in 8m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a27470cd70a4dd08109676704fafd71 : SUCCESS in 16m 09s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/a0a829b82b7e479d97cd2b9bcadf4768 : SUCCESS in 6m 16s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/cc3a7d25e20148bf81241e01404d5ccc : SUCCESS in 5m 15s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/031772425bec410e83a9d7ec62d0ae90 : SUCCESS in 29m 53s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/6e5944a01a3d4776bf0ba3bfe0674a38 : SUCCESS in 42m 18s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/af423e46b5ef424290b9e4a9d3f17347 : SUCCESS in 48m 18s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/f7ee05c7c611434c82c874cfa942d1e1 : SUCCESS in 33m 45s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/e1c65b1733cc4fd2af5b37a4b6b77268 : SUCCESS in 53m 36s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/c701f8a2ed844477b1baf84d017ef253 : SUCCESS in 30m 13s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/9f81c7ee258b4a839cf8a660507aa2f0 : SUCCESS in 45m 35s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/e9a2527b5ea742048f4f337bb9659163 : SUCCESS in 1h 01m 12s","accounts_in_message":[],"_revision_number":3},{"id":"e27962cade25ac92274a59d8246d31dc1b79dd14","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 06:10:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"b0ddc65744a47da42099834923dff2980505c206","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 06:13:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e5cad8766a044218a37757d9735ca13\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be41c7bbc07f4ae8a9765901823378a7 : SUCCESS in 1m 10s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1e50d16f28347fb9b0e2ae49b06d659 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1a0a480f65d5e0ce64e73e44d840e22617042994","revisions":{"3d9be26c22900dbaeb0132dbf41b4bef42d2c461":{"kind":"REWORK","_number":1,"created":"2026-02-01 21:36:35.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/65/975365/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/65/975365/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/65/975365/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/65/975365/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/65/975365/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/65/975365/1"}}},"commit":{"parents":[{"commit":"66ba356ae901b4ada6170a1f78afa7c03101c7a2","subject":"Replace ironic-tempest-bios-ipmi-direct with -autodetect","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/66ba356ae901b4ada6170a1f78afa7c03101c7a2"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2026-02-01 21:30:20.000000000","tz":780},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2026-02-01 21:36:16.000000000","tz":780},"subject":"Add bootc deploy to autodetect default","message":"Add bootc deploy to autodetect default\n\nThe ``bootc`` deploy interfaces is now enabled by default in\n``enabled_deploy_interfaces`` and has also been added to the default\nlist of ``autodetect_deploy_interfaces`` for the ``autodetect`` deploy\ninterface. This allows nodes configured with the ``autodetect`` deploy\ninterface to automatically select the ``bootc`` deploy interface for\ncontainer images which are identified as bootc images through registry\nintrospection. The ``autodetect`` deploy interface will fall back to the\n``direct`` deploy interface for standard disk images and OCI artifacts.\n\nChange-Id: Idbed1770c79b2f84bb27c8545395a0391a7b4b9c\nSigned-off-by: Steve Baker \u003csbaker@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/3d9be26c22900dbaeb0132dbf41b4bef42d2c461"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/3d9be26c22900dbaeb0132dbf41b4bef42d2c461"}]},"branch":"refs/heads/master"},"c230d721ad91641218e0b8baeef6c58a6a9f960d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-02-03 00:27:14.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/65/975365/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/65/975365/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/65/975365/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/65/975365/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/65/975365/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/65/975365/2"}}},"commit":{"parents":[{"commit":"93a833fb768fe39f62a49c08086de0b8cd63bdae","subject":"Replace ironic-tempest-bios-ipmi-direct with -autodetect","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/93a833fb768fe39f62a49c08086de0b8cd63bdae"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2026-02-01 21:30:20.000000000","tz":780},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2026-02-03 00:27:07.000000000","tz":780},"subject":"Add bootc deploy to autodetect default","message":"Add bootc deploy to autodetect default\n\nThe ``bootc`` deploy interfaces is now enabled by default in\n``enabled_deploy_interfaces`` and has also been added to the default\nlist of ``autodetect_deploy_interfaces`` for the ``autodetect`` deploy\ninterface. This allows nodes configured with the ``autodetect`` deploy\ninterface to automatically select the ``bootc`` deploy interface for\ncontainer images which are identified as bootc images through registry\nintrospection. The ``autodetect`` deploy interface will fall back to the\n``direct`` deploy interface for standard disk images and OCI artifacts.\n\nChange-Id: Idbed1770c79b2f84bb27c8545395a0391a7b4b9c\nSigned-off-by: Steve Baker \u003csbaker@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/c230d721ad91641218e0b8baeef6c58a6a9f960d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/c230d721ad91641218e0b8baeef6c58a6a9f960d"}]},"branch":"refs/heads/master"},"1a0a480f65d5e0ce64e73e44d840e22617042994":{"kind":"REWORK","_number":3,"created":"2026-02-11 12:46:48.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/65/975365/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/65/975365/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/65/975365/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/65/975365/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/65/975365/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/65/975365/3"}}},"commit":{"parents":[{"commit":"44f5b0a630a11c1f6c9e5045b62f546449d005fd","subject":"Replace ironic-tempest-bios-ipmi-direct with -autodetect","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/44f5b0a630a11c1f6c9e5045b62f546449d005fd"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2026-02-01 21:30:20.000000000","tz":780},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2026-02-11 12:46:11.000000000","tz":780},"subject":"Add bootc deploy to autodetect default","message":"Add bootc deploy to autodetect default\n\nThe ``bootc`` deploy interfaces is now enabled by default in\n``enabled_deploy_interfaces`` and has also been added to the default\nlist of ``autodetect_deploy_interfaces`` for the ``autodetect`` deploy\ninterface. This allows nodes configured with the ``autodetect`` deploy\ninterface to automatically select the ``bootc`` deploy interface for\ncontainer images which are identified as bootc images through registry\nintrospection. The ``autodetect`` deploy interface will fall back to the\n``direct`` deploy interface for standard disk images and OCI artifacts.\n\nChange-Id: Idbed1770c79b2f84bb27c8545395a0391a7b4b9c\nSigned-off-by: Steve Baker \u003csbaker@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1a0a480f65d5e0ce64e73e44d840e22617042994"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1a0a480f65d5e0ce64e73e44d840e22617042994"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
