)]}'
{"id":"openstack%2Fironic~977553","triplet_id":"openstack%2Fironic~master~Ibfac9d3c0674f33217cdca2b6128cb2bc3fed5dc","project":"openstack/ironic","branch":"master","topic":"ironic-vxlan","attention_set":{},"removed_from_attention_set":{"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2026-02-23 16:27:29.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"Ibfac9d3c0674f33217cdca2b6128cb2bc3fed5dc","subject":"CI: Wait for neutron to be initalized before trying to change it.","status":"MERGED","created":"2026-02-21 19:36:49.000000000","updated":"2026-02-23 16:29:01.000000000","submitted":"2026-02-23 16:27:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"977553-ironic-vxlan","meta_rev_id":"748b65da6b5b22ebb054fc90bc43e2096785d843","_number":977553,"virtual_id_number":977553,"owner":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-23 16:27:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2026-02-23 15:12:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-02-23 14:23:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":1,"date":"2026-02-23 15:12:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-21 19:58:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-21 20:56:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-23 14:23:05.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2026-02-23 15:12:24.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"}],"messages":[{"id":"a3e31d5eedd9b404bb1b785ac2c1a46df6c7aa1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2026-02-21 19:36:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1c0b2dc89a3720c2030c653c76967c33d51112c7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-21 19:58:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/370963d934ce4ffab0e7c915b2e99a3d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/98547e90072a4718aec3db16a030d789 : SUCCESS in 13m 42s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5c651337dc7447a1b3120732222ed678 : SUCCESS in 13m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a01966d9c13c483a89ede793933d3a9b : RETRY_LIMIT in 4m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c14957bf0f94122f399c30b824ce8f4ecf1cf95d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-21 20:56:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/570120a1e05b48d599a6d3cae099c2bf\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d669517b4f74dba8540412aea5d45b7 : SUCCESS in 12m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9e3a64426ff4cf4a3d79592dd1818a6 : SUCCESS in 3m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/62ab8e0099454d1cb654d5b695afadfc : SUCCESS in 6m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/24570b9f5af14f4ab643b116594d790f : SUCCESS in 9m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/eb07447daa2d4db5b577415ac0e941bc : SUCCESS in 13m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/460a31ae7ec7423096e56acb09224751 : SUCCESS in 23m 30s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/2d004332b1bb46feb8130a07384e18b9 : SUCCESS in 5m 17s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/45c82ec4fcef4dfe995a388b93ea0f6e : SUCCESS in 9m 30s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/3ce228ea386c4d558cb0f9cacf45d04f : SUCCESS in 29m 09s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/f4d2e143e7844bb383808331af07f783 : SUCCESS in 1h 00m 58s (non-voting)\n- ironic-tempest-standalone-advanced https://zuul.opendev.org/t/openstack/build/8d85b3d9966a4334ad6265485f288b06 : SUCCESS in 48m 45s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/e8686c6b8b7c4331822f9462790e6dee : SUCCESS in 50m 09s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/ce3e2cb18423476e96e07074ae3a9f4d : SUCCESS in 23m 08s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/d3c47a855fdb476b9f2a4ea0723019c8 : SUCCESS in 33m 12s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/1e49436eefd34f7eb4811eccc97b56a0 : SUCCESS in 24m 31s (non-voting)\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/c79bcb5e391748018188d7a236d7992d : FAILURE in 1h 14m 00s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-multinode https://zuul.opendev.org/t/openstack/build/925aa9dfdf614896bfd2eeb63bd0439f : SUCCESS in 54m 58s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-multinode-shard https://zuul.opendev.org/t/openstack/build/b7243cb1a2104eb4837358414de38186 : SUCCESS in 38m 41s (non-voting)\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/2d406ddd7e6f44a98109e46e3b6457ed : SUCCESS in 30m 44s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/9699416ef3584133a4dbc038de972311 : SUCCESS in 33m 06s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/27c2e0f9eb7840a098de9b1e987ce997 : SUCCESS in 29m 30s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/4da1b67bc690420e95f76982b6839fb0 : SUCCESS in 52m 19s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/2e95f1bee52c459ba4525a28ecd1a530 : SUCCESS in 51m 44s (non-voting)\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/9efce51554ae45b89fabb6ad086478ee : SUCCESS in 40m 25s (non-voting)\n- metal3-ironic-standalone-operator-functional https://zuul.opendev.org/t/openstack/build/a7c0a3c2a4fd438f882470ea5d5dbb0a : FAILURE in 7m 44s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/2b2b5877f67c44709cb000190a250683 : SUCCESS in 32m 18s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/42eb18084b2d4c49bb39b58d42540b57 : SUCCESS in 49m 14s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/9276045c2021420c9b07b4274196ba54 : SUCCESS in 43m 48s (non-voting)\n- bifrost-integration-on-ubuntu-noble https://zuul.opendev.org/t/openstack/build/54e6aa5a0c55491ca62050b0bb254498 : SUCCESS in 43m 41s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/caab5a51d9a24d429c6a8706835afda8 : SUCCESS in 24m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0a36712ca04bc20e77bb497fe88483b0b63e5e73","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2026-02-22 00:23:01.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":1},{"id":"d9719b30a7ac5f9fb600e359f7dc31ba4d9d6334","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2026-02-22 00:23:31.000000000","message":"Topic set to ironic-vxlan","accounts_in_message":[],"_revision_number":1},{"id":"08173e78c29a96749c8affde400f50570b42ff81","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2026-02-23 14:23:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"961a2bd21f3fb1066d080ac19ed68cbba0e154d1","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2026-02-23 15:12:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a8ac06a15451cc432fb26ba73bc21cf8e21b23a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 15:12:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d2d0195d38bde175372b93b212164cf27d8280dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 16:27:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9855c81ca7bb4cc28351c86883637124\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc64267092ee47de891cd89e2caf3980 : SUCCESS in 7m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/974966e40a75404187fb8ac3d009fde5 : SUCCESS in 9m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/55054feb495847458de81b91cc2d623f : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06b988eb16b5476f8d2453500df4afa3 : SUCCESS in 15m 01s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/a8baddf979324a25b1e4c91e43415dac : SUCCESS in 3m 44s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/143b04375f7242d5b4f6e0aad77c0af7 : SUCCESS in 5m 44s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/2047cc6facd64fdc898ecd39a518012c : SUCCESS in 22m 32s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/7ed79a24c9a04bbb9f9237714a1ce899 : SUCCESS in 48m 36s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/814550d92ff94d919c07c586b86aabb2 : SUCCESS in 48m 25s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/b39b61bcef3e4833ac77004885c4a05e : SUCCESS in 42m 26s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/03d03f3777dc4e80a579fcc407d63ecd : SUCCESS in 30m 10s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/5f354f24296f4412a3a4b337e035472d : SUCCESS in 39m 04s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/1a6856240cd240b5bde06ecc0c3ae417 : SUCCESS in 53m 51s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/0c6393be997e495199eb7421b3f15e09 : SUCCESS in 51m 09s\n- metal3-integration https://zuul.opendev.org/t/openstack/build/fecc26e39e4245539b5f1109afe96fde : SUCCESS in 32m 52s","accounts_in_message":[],"_revision_number":1},{"id":"d93e15e4f0fceb029b72aa8f2c8192f875fe367d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 16:27:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"748b65da6b5b22ebb054fc90bc43e2096785d843","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 16:29:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/653cb12935244f2c9fbbe77b42f7e872\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d51d63b69fef4d9fb780756cadccdd41 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"32b2fbba1290c9c349047ae9eb4552d6a0186605","revisions":{"32b2fbba1290c9c349047ae9eb4552d6a0186605":{"kind":"REWORK","_number":1,"created":"2026-02-21 19:36:49.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/53/977553/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/53/977553/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/53/977553/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/53/977553/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/53/977553/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/53/977553/1"}}},"commit":{"parents":[{"commit":"ff1016f96fac225c19cc0981a4d36f6e080bbcf4","subject":"Merge \"Add bootc deploy to autodetect default\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/ff1016f96fac225c19cc0981a4d36f6e080bbcf4"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2026-02-21 19:35:41.000000000","tz":-480},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2026-02-21 19:36:22.000000000","tz":-480},"subject":"CI: Wait for neutron to be initalized before trying to change it.","message":"CI: Wait for neutron to be initalized before trying to change it.\n\nNoticed some networking-baremetal jobs sometimes fail becasue the job\nruns ahead too fast before neutron is up. This should remedy that.\n\nChange-Id: Ibfac9d3c0674f33217cdca2b6128cb2bc3fed5dc\nSigned-off-by: Julia Kreger \u003cjuliaashleykreger@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/32b2fbba1290c9c349047ae9eb4552d6a0186605"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/32b2fbba1290c9c349047ae9eb4552d6a0186605"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
