)]}'
{"id":"openstack%2Fironic~978112","triplet_id":"openstack%2Fironic~master~I4c6e8db7878e1f8ffe18fa75c10eb92fa5f0c62c","project":"openstack/ironic","branch":"master","attention_set":{},"removed_from_attention_set":{"5890":{"account":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"last_update":"2026-02-27 20:36:32.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"I4c6e8db7878e1f8ffe18fa75c10eb92fa5f0c62c","subject":"fix up the behavior of serial number reading on iDRAC","status":"MERGED","created":"2026-02-26 17:53:36.000000000","updated":"2026-02-27 20:38:03.000000000","submitted":"2026-02-27 20:36:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978112","meta_rev_id":"59671d841748451e02c74a01d4561e95a92c650f","_number":978112,"virtual_id_number":978112,"owner":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-27 20:36:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":2,"date":"2026-02-27 19:13:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":2,"date":"2026-02-27 09:04:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":1,"date":"2026-02-27 19:13:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-26 18:43:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-26 19:35:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-27 09:04:13.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2026-02-27 19:13:26.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"}],"messages":[{"id":"64134881e64c8869f3c1118e022a6c3eb41f26c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-02-26 17:53:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"358806e5bf02f18e44ce89064c615f8258f313f9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 18:43:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0154ded6545c4304b7178904273efb6f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b917fa5a5c3d4d9d841052d2f83ba9f1 : SUCCESS in 12m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6388ea033a6648d89706cf602814028e : RETRY_LIMIT in 4m 05s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/fd3543f285654311860b55dbf5eab2f2 : SUCCESS in 14m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"57e62ee54a5933a5ec031eada1484e16ada33381","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 19:35:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c592e0368ce247449fb15ed7679089c4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5287ede351754c82a3e8c5758de3c3f2 : SUCCESS in 12m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f2c42e146dc4de8905cc4294ee54234 : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/38587cfe132f433595f9a87bdb9d8ef7 : SUCCESS in 7m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/977482e8cd2d491c9a7b31d5629b6e79 : SUCCESS in 6m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f9c32985c4cd407fac9aefe5354baa20 : SUCCESS in 13m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc309d2c03214d8798e4893713d56d3b : SUCCESS in 15m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7bc5169af4e43d28823ab8889a7c6d2 : SUCCESS in 16m 27s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/a87a91d822a1472393312a4ff892ee4c : SUCCESS in 5m 36s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/d7556b7a0d554dc3ac9339edff4204db : SUCCESS in 5m 30s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/f30adef6d099448485995ecdffcf21e5 : SUCCESS in 7m 34s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/8028d202a2ec43109a71010e55041d44 : SUCCESS in 37m 46s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/894284f47b4d469ca53353583e2b1537 : SUCCESS in 53m 49s (non-voting)\n- ironic-tempest-standalone-advanced https://zuul.opendev.org/t/openstack/build/16bb2408692a4032a2201754d5eae8ff : SUCCESS in 1h 12m 47s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/2541c6a6321e4138a76b1bd75a0484bc : SUCCESS in 25m 34s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/b63b2fd7b6a9482cb9072211aaed5899 : SUCCESS in 53m 13s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/4b2bc22896c34b6b9d4d38765a0e2640 : SUCCESS in 48m 14s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/058b2849b4714c678255eefeb2b5cda2 : SUCCESS in 48m 52s (non-voting)\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/c4535848a82745539986d79c6ff8718e : FAILURE in 55m 48s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-multinode https://zuul.opendev.org/t/openstack/build/286a4e6f222f448f93c0ace61d5ac41a : SUCCESS in 1h 12m 50s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-multinode-shard https://zuul.opendev.org/t/openstack/build/521359ae481142f98947509c00765a1d : SUCCESS in 1h 33m 06s (non-voting)\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/3345aec6ebb54352a1e2586b93cff030 : SUCCESS in 51m 43s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/bd2c9671f9a745b9a3d3721180875b98 : SUCCESS in 25m 20s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/dadd6f4c10564938908321055fb081d7 : SUCCESS in 49m 35s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/9c8e2b1f9a824026ae1ec4c837d8706b : SUCCESS in 1h 08m 28s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/b87dc16a59bf43ce84ce52788125bf81 : SUCCESS in 1h 23m 25s (non-voting)\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/7261806b2fee46ac8fdf53f16061dbd8 : SUCCESS in 1h 10m 46s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/bb698faf8abf4788a44f4f122dc50820 : SUCCESS in 35m 55s\n- metal3-ironic-standalone-operator-functional https://zuul.opendev.org/t/openstack/build/b512fab16edd4e80bf73e74a8914af58 : FAILURE in 11m 07s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/b29b24bdfdc8459ea62c68408b29c378 : SUCCESS in 36m 28s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/b9c03b893438441aabf6e3045da5d374 : SUCCESS in 45m 25s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/288ca72f51624dbeb070081f7d57353a : SUCCESS in 53m 12s (non-voting)\n- bifrost-integration-on-ubuntu-noble https://zuul.opendev.org/t/openstack/build/996bef022bf7420cb840d2d58218f687 : FAILURE in 6m 57s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ee5fe727053f456bb2640c33b16471d4 : FAILURE in 6m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0e6370c79b82145f1d51e4d0335af80825ad96a2","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-02-26 20:26:57.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":1},{"id":"7227fa844363ed9a3c367efbf810083ced2d89d9","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2026-02-27 09:04:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"900a834358b10520d41f3aefdfcd3d790336eec7","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2026-02-27 19:13:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f009bcddfc173255766358dbd041f0ba3a714763","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 19:13:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7155ab51c3b8036fa37736a8d2696e00d0f6ada2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 20:36:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/707807679a8b4401bb5113022cfc2c99\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/410224bffec14203b3f1e5b3c1910eec : SUCCESS in 6m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6f7ca8a23b694f76894122521bc537de : SUCCESS in 8m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6488fc90d99c42d7ab1835fdd0862bbf : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e35bfd89fd0d48ffa76cf41e13ad56ad : SUCCESS in 16m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee0eac18363a4d90960afc40d6fb4e0d : SUCCESS in 19m 47s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/1552338f296e465381899d8acf7e3da4 : SUCCESS in 5m 11s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/5359a30be3914115af38892655423bc1 : SUCCESS in 9m 50s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/4133083602b0459d8e4e652e592ddcfe : SUCCESS in 42m 51s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/eaef08da74cf46d89abc23cafd8b8855 : SUCCESS in 37m 40s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/34e6e203f22948c486f8ff8238c7156f : SUCCESS in 52m 15s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/3cd2b2213d824908906f994fd02a87dd : SUCCESS in 25m 18s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/356f8cfcb7754b6c87e42c7e16a6fb2d : SUCCESS in 1h 05m 05s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/443d5eca0cca4f808387441d0ae86fb4 : SUCCESS in 42m 07s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/65a4cbaf94614db6a62d7c76a23217b2 : SUCCESS in 50m 23s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/ed9900ed2e3a48769f0a6190bcd78fdc : SUCCESS in 1h 16m 54s\n- metal3-integration https://zuul.opendev.org/t/openstack/build/460e6ba2e4c9419b87c8da6422ef1546 : SUCCESS in 30m 46s","accounts_in_message":[],"_revision_number":1},{"id":"425dcc26cadce9d3455154811df0af5975f16878","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 20:36:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"59671d841748451e02c74a01d4561e95a92c650f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 20:38:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ce718ea864a47c19802067ceb2f34cf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f11f1d5d65e2427da2a81f198143d778 : SUCCESS in 1m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/256a139fac6c4fcab4a80921be3cf229 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6205f00905a46954b8a23bf62de5cc52230cc3f2","revisions":{"6205f00905a46954b8a23bf62de5cc52230cc3f2":{"kind":"REWORK","_number":1,"created":"2026-02-26 17:53:36.000000000","uploader":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"ref":"refs/changes/12/978112/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/12/978112/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/12/978112/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/12/978112/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/12/978112/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/12/978112/1"}}},"commit":{"parents":[{"commit":"09c4bcd155a02463a36d7fa590f01979928a8938","subject":"Merge \"Enable parallel processes in sphinx-build to speed up tox \"docs\" target\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/09c4bcd155a02463a36d7fa590f01979928a8938"}]}],"author":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2026-01-22 15:26:53.000000000","tz":-360},"committer":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2026-02-26 17:51:32.000000000","tz":-360},"subject":"fix up the behavior of serial number reading on iDRAC","message":"fix up the behavior of serial number reading on iDRAC\n\nWe added the \u0027sku\u0027 field this cycle to handle the iDRAC reporting the\nserial number incorrectly. Instead of generically extending the redfish\ndriver to include the sku field which isn\u0027t useful on hardware outside\nof iDRAC just fix up the data in the iDRAC case.\n\nChange-Id: I4c6e8db7878e1f8ffe18fa75c10eb92fa5f0c62c\nSigned-off-by: Doug Goldstein \u003ccardoe@cardoe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/6205f00905a46954b8a23bf62de5cc52230cc3f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/6205f00905a46954b8a23bf62de5cc52230cc3f2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
