)]}'
{"id":"openstack%2Fironic~978125","triplet_id":"openstack%2Fironic~stable%2F2025.2~I786c90effda9892bfef3a90700d1fc2a3372cfde","project":"openstack/ironic","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"10273":{"account":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"last_update":"2026-02-27 00:18:44.000000000","reason":"Change was submitted"},"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2026-02-27 00:18:44.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"I786c90effda9892bfef3a90700d1fc2a3372cfde","subject":"Fix async periodic on redfish for servicewait","status":"MERGED","created":"2026-02-26 18:53:56.000000000","updated":"2026-02-27 00:21:41.000000000","submitted":"2026-02-27 00:18:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978125","cherry_pick_of_change":971581,"cherry_pick_of_patch_set":2,"meta_rev_id":"46d9105af168bef2c056b76e98e0cec14d69a07d","_number":978125,"virtual_id_number":978125,"owner":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-27 00:18:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":2,"date":"2026-02-26 22:58:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":1,"date":"2026-02-26 22:58:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-26 18:53:56.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"CC"},{"updated":"2026-02-26 18:55:19.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2026-02-26 20:21:29.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"CC"},{"updated":"2026-02-26 22:54:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-26 22:58:03.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"}],"messages":[{"id":"7f75171754569c0cd572a57b3b4461f066b3757a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2026-02-26 18:53:56.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"842d236ede3113a6478903efeed3115f4ef1d18c","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2026-02-26 18:55:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b50d8ba1df09cd7a8b928a20c1dc5df8d0a2e845","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2026-02-26 18:55:28.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":1},{"id":"775944998c6ed18b176fad3e866dfe91bd59d606","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-02-26 20:21:29.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"7838c2bc32c2b53f866bbae24ce35a60601b87ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 22:54:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21cbf878ff104093855645900eac1adb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf2d0fa90e85412e9eca97ebd356b9f8 : SUCCESS in 12m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eee1e440ccd34293932cfa6e785b6d7d : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6d4aae43ea8a43d1af941d427d3b9a40 : SUCCESS in 6m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d5f8c03bb26f46869fb64c69d731e2e2 : SUCCESS in 9m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f1a1dbd75c8844beb7d98de2ef8006fe : SUCCESS in 12m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f51f7fcd0264c43a8bed4dad3f1be89 : SUCCESS in 13m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e47df9402d634c1d89ea18c146832da6 : SUCCESS in 16m 11s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/94ee55aa92a047dc92e65373f88c39e3 : SUCCESS in 3m 16s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/781159f6c9b9450794606e3d1acae2f2 : SUCCESS in 11m 04s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/96c49593401545bf8128f49e34e1306e : SUCCESS in 7m 16s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/b2ab00b3a1ad4a388bc33826b2a495d8 : SUCCESS in 36m 09s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/1af5eac5d2a2468d9c944ee496acfd8a : SUCCESS in 1h 01m 37s (non-voting)\n- ironic-tempest-standalone-advanced https://zuul.opendev.org/t/openstack/build/21be777528774d7ca8149876bdd6ba6f : SUCCESS in 56m 55s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/affb4cac7fc148ffa2352ccfa46d0dba : SUCCESS in 45m 10s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/1dde192171cb447db8aed068ad77553f : SUCCESS in 29m 35s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/68f3aadb66a84228bf791b4aa92b36b3 : FAILURE in 15m 23s (non-voting)\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/5995e772263740259536206349cef6f6 : SUCCESS in 47m 48s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/a938c794064e496b87c188da602e8418 : SUCCESS in 52m 29s\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/9bd7df73d3454bbd94512fe840f2b759 : SUCCESS in 32m 47s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/dffdf7b80e314ee4960a41c6bf324dc4 : SUCCESS in 2h 26m 48s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-shard https://zuul.opendev.org/t/openstack/build/353f237782b54c7e837bf1bd67f5782a : SUCCESS in 1h 12m 39s (non-voting)\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/7c0da1eba98d4d73b5246e7354836751 : SUCCESS in 1h 12m 38s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/8e701c5caefd42b3b512eec19d672ba4 : SUCCESS in 31m 22s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/2f8a0e0fde0048fd8ae6cdb1d3fe4e79 : SUCCESS in 31m 49s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/75e6672fa08c4f6ea75c3bba555b0491 : SUCCESS in 30m 47s\n- metalsmith-integration-glance-centos9-uefi https://zuul.opendev.org/t/openstack/build/7b3798008366409a9980d0e271bfc1f3 : FAILURE in 28m 29s (non-voting)\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/4d5808d139bc4e37bb343a65d0a44056 : SUCCESS in 1h 08m 56s (non-voting)\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/a0106dc942e44617b518a114f5172f38 : SUCCESS in 37m 33s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/90276cdfe81d4e8cb5e97c3219bdea13 : SUCCESS in 47m 23s (non-voting)\n- ironic-inspector-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/1a22b91d695f40e8be3043fc4b067374 : SUCCESS in 19m 36s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/a4fe4f6cbeaf468b864b48520a92a5cc : FAILURE in 21m 29s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/08aae46828424adebe4b26f1c9879548 : SUCCESS in 48m 08s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/4568f14be38a4f4fa4ce68ec885a7679 : SUCCESS in 23m 06s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/44b0f67b0b154968a88d83b2c3c93896 : SUCCESS in 42m 33s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/70441215e8c14a10ae135b1f1949c6d5 : SUCCESS in 20m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c77187e1764917f846f3d3f46dbc88e67819ea39","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-02-26 22:58:03.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"41110beaaab284cf1d518486fbf5b1a0fd2c3dc6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 22:58:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"120635d19bf7dc3d98f319b0b4d703af1852bddf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 00:18:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54969a6780bf4caba134e0d6b0c9e184\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/349169e61bca433fb266438df42f27a2 : SUCCESS in 6m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3d3c556e1b38409bb8e4492bee704454 : SUCCESS in 7m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/01809d21cafa4ee090607c8918513704 : SUCCESS in 11m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/962fc635d72742e189c996530fe34425 : SUCCESS in 12m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/642db4beb1b74f008db4d0670e0eadfb : SUCCESS in 16m 02s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/4558195af3da41c1a06c3c039753e59a : SUCCESS in 5m 04s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/33fb1a70e91447768e3e826e9564615c : SUCCESS in 11m 33s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/b8c0ad40cffa4385a2f8d6991edc9558 : SUCCESS in 29m 40s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/a74cba46ecc74c49ad78c7a9519488c2 : SUCCESS in 55m 11s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/342abc1cbb724ab6aed22227c8a0467c : SUCCESS in 56m 38s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/e1e1ea510c164251a8d14bba87f81532 : SUCCESS in 51m 43s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/784b228f9cee4f728d8b76ceab5451f0 : SUCCESS in 46m 32s\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/dc521f5c9a334041a121789c188432e4 : SUCCESS in 1h 05m 12s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/2a63ca2069ab4081a29692d8be903355 : SUCCESS in 40m 17s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/88937c3d9526425fa797021ee893d83a : SUCCESS in 47m 55s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/ecbe892aaa4947559f686e34fb369e11 : SUCCESS in 1h 15m 19s","accounts_in_message":[],"_revision_number":2},{"id":"856f1915a24cdbc2e6846ba2bae84f54dcf81c33","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 00:18:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"46d9105af168bef2c056b76e98e0cec14d69a07d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 00:21:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cb8f5ae08604f17a603282778cc380c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8e8849cd8a5498eb2fc16350ffb196e : SUCCESS in 2m 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/02ef64ee04764041aecca01137b80a4a : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4f8b37f3ae474ebb7a018d2e3c7577773148baa7","revisions":{"97172802eaf4e726d93a8d478b89c2919e635bb5":{"kind":"REWORK","_number":1,"created":"2026-02-26 18:53:56.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/25/978125/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/25/978125/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/25/978125/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/25/978125/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/25/978125/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/25/978125/1"}}},"commit":{"parents":[{"commit":"d5477042f91b8ece9a3ab4dbb0dcfa4f6603928b","subject":"fix: missing staticmethod decorator on inspection_rules func","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/d5477042f91b8ece9a3ab4dbb0dcfa4f6603928b"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2025-12-19 23:09:17.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2026-02-26 18:53:56.000000000","tz":0},"subject":"Fix async periodic on redfish for servicewait","message":"Fix async periodic on redfish for servicewait\n\nsdmitriev1 reported issues using the service verb to perform firmware\nupdates utilizing the management interface. Turns out we had a bug\nwhich was not detected previously as the tested paths went through\nfirmware interface updates and the dell idrac variation interface.\n\nIn any event, this change should fix these issues. Also adds deploywait,\nbecause operators can choose different firmware for deploy templates as\nwell, and just to be on the safe side while also matching the other\nstep handling logic.\n\nFurther enough, we needed to properly call resumption of the next\nstep, which was furhter pointed out by the reporter.\n\nClaude was kind enough to write some unit tests, and noted as assisted\nby for it\u0027s assistance.\n\nCloses-Bug: 2136895\nAssisted-By: Claude Code/Claude Sonnet 4\nChange-Id: I786c90effda9892bfef3a90700d1fc2a3372cfde\nSigned-off-by: Julia Kreger \u003cjuliaashleykreger@gmail.com\u003e\nCo-Authored-By: Stanislav Dmitriev \u003csdmitriev1@gmail.com\u003e\n(cherry picked from commit 98cda8f2c58abde4f5e4ad42c26be8d1d3128d41)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/97172802eaf4e726d93a8d478b89c2919e635bb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/97172802eaf4e726d93a8d478b89c2919e635bb5"}]},"branch":"refs/heads/stable/2025.2"},"4f8b37f3ae474ebb7a018d2e3c7577773148baa7":{"kind":"REWORK","_number":2,"created":"2026-02-26 20:21:29.000000000","uploader":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"ref":"refs/changes/25/978125/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/25/978125/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/25/978125/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/25/978125/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/25/978125/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/25/978125/2"}}},"commit":{"parents":[{"commit":"d5477042f91b8ece9a3ab4dbb0dcfa4f6603928b","subject":"fix: missing staticmethod decorator on inspection_rules func","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/d5477042f91b8ece9a3ab4dbb0dcfa4f6603928b"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2025-12-19 23:09:17.000000000","tz":-480},"committer":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2026-02-26 20:21:12.000000000","tz":-360},"subject":"Fix async periodic on redfish for servicewait","message":"Fix async periodic on redfish for servicewait\n\nsdmitriev1 reported issues using the service verb to perform firmware\nupdates utilizing the management interface. Turns out we had a bug\nwhich was not detected previously as the tested paths went through\nfirmware interface updates and the dell idrac variation interface.\n\nIn any event, this change should fix these issues. Also adds deploywait,\nbecause operators can choose different firmware for deploy templates as\nwell, and just to be on the safe side while also matching the other\nstep handling logic.\n\nFurther enough, we needed to properly call resumption of the next\nstep, which was furhter pointed out by the reporter.\n\nClaude was kind enough to write some unit tests, and noted as assisted\nby for it\u0027s assistance.\n\nConflict:\n  ironic/tests/unit/drivers/modules/redfish/test_management.py pep8\n\nCloses-Bug: 2136895\nAssisted-By: Claude Code/Claude Sonnet 4\nChange-Id: I786c90effda9892bfef3a90700d1fc2a3372cfde\nSigned-off-by: Julia Kreger \u003cjuliaashleykreger@gmail.com\u003e\nCo-Authored-By: Stanislav Dmitriev \u003csdmitriev1@gmail.com\u003e\n(cherry picked from commit 98cda8f2c58abde4f5e4ad42c26be8d1d3128d41)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4f8b37f3ae474ebb7a018d2e3c7577773148baa7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4f8b37f3ae474ebb7a018d2e3c7577773148baa7"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
