)]}'
{"id":"openstack%2Fironic~978280","triplet_id":"openstack%2Fironic~stable%2F2025.2~I4adef236f8fb09509f05141d1dfb9f5de38759bb","project":"openstack/ironic","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"5890":{"account":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"last_update":"2026-03-02 14:36:43.000000000","reason":"\u003cGERRIT_ACCOUNT_5890\u003e replied on the change","reason_account":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}}},"hashtags":["ironic-week-prio"],"change_id":"I4adef236f8fb09509f05141d1dfb9f5de38759bb","subject":"fix inspection rules loaded from a file to not fail in logging","status":"MERGED","created":"2026-02-27 23:35:06.000000000","updated":"2026-03-02 16:34:24.000000000","submitted":"2026-03-02 16:15:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978280","cherry_pick_of_change":978116,"cherry_pick_of_patch_set":1,"meta_rev_id":"86bdf2455927604fd1479aeb59eddcdc383db38e","_number":978280,"virtual_id_number":978280,"owner":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-02 16:15:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-03-02 12:11:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":1,"date":"2026-03-02 14:36:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-28 01:38:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-02 12:11:32.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"d3e1e6654f3afa23cc1ee88e11cc113a9cf1a9d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-02-27 23:35:06.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"cfe3cc62bb8f48180eb21b4d48e02fa4ace858ab","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-02-27 23:59:12.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":1},{"id":"6ce1de1902e5955fe14694aed4d8ce9b41fc6b41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 01:38:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc5c6e886df74615bf31880931caee1b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6fce4f45268e4755a13b64d8d8b7f320 : SUCCESS in 13m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00cce4ee65a44b5486c93a9b55dbf2e6 : SUCCESS in 5m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/efa34d057949498e89266d990a3503d0 : SUCCESS in 9m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e6409158a65e473f8d33248577b8eb77 : SUCCESS in 8m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3998c92156e44f80b5e6521f8a3ce7a2 : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5df3f25ad4384913817fb54f52fa10fb : SUCCESS in 17m 57s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/bf944231be334c26a48fe83134a0c770 : SUCCESS in 4m 59s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/89896b96ec814afd9690d75f2eea22c0 : SUCCESS in 6m 48s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/d081bcd6fb474d40af9b96d8b53881db : SUCCESS in 6m 41s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/8fe3afcf1dea4d8eba1d8394c97a9796 : SUCCESS in 18m 59s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/15c10a3448894d4eb5eed2509d2cffcd : SUCCESS in 52m 30s (non-voting)\n- ironic-tempest-standalone-advanced https://zuul.opendev.org/t/openstack/build/285d43b92fb14940b429425e26a34799 : SUCCESS in 1h 23m 06s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/df484b9f77f64152b0efc197582763b1 : SUCCESS in 49m 14s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/7edb1ba07fc5462ca40edbb32f89a61e : SUCCESS in 49m 44s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/19d3d8e3486045ee9f0f1dfd7c80b419 : FAILURE in 14m 05s (non-voting)\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/e902840bf9754c92944afe77efd0cb16 : SUCCESS in 18m 34s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/51690ab270354a7bb6b47b4773a0bee8 : SUCCESS in 19m 44s\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/c2305335c331441688dc2f5b73464d93 : SUCCESS in 1h 06m 09s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/249d1c0b6bad40b4bdbe38ab08641566 : SUCCESS in 1h 57m 13s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-shard https://zuul.opendev.org/t/openstack/build/d634a879c28646f0891e0ff8cbe6aafc : SUCCESS in 1h 14m 33s (non-voting)\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/86db64134c994b409badd6a1b34cf6b4 : SUCCESS in 1h 13m 49s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/609b326673dd4a06a407ba86d1b1ef2d : SUCCESS in 21m 56s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/301c093b4ea34542ae179974061f06f1 : SUCCESS in 52m 09s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/a08ab910ba944d3faf39c6a2e2c69106 : SUCCESS in 1h 09m 27s\n- metalsmith-integration-glance-centos9-uefi https://zuul.opendev.org/t/openstack/build/88a0c46430c9481d85b9457f7061200f : FAILURE in 25m 41s (non-voting)\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/556822370811444784e9b8d3c6c70c35 : SUCCESS in 1h 09m 53s (non-voting)\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/b4517d6248714eae816e0dfb6d0833a3 : SUCCESS in 41m 55s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/dcf5e8cc7a5941669c72058f7a309146 : SUCCESS in 47m 00s (non-voting)\n- ironic-inspector-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/53fa2aa4163b431db8f8eddf438dd58b : SUCCESS in 43m 52s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/3d31c6a3292e4fe7b3d9e185f4d70a1e : FAILURE in 27m 20s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/7184ed655a3e4a74a5095a645f84935b : SUCCESS in 33m 48s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/7b9e0f0e3f9a4a499cc71de4e0945c49 : SUCCESS in 57m 49s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/349ad5b71c3f40eaa533eb98b086db8c : SUCCESS in 34m 25s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/35f905f5d6c3470c9dd4666ee43598c7 : SUCCESS in 32m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"96e129732920762a6a99e2836bc2b7e487a5fdb8","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2026-03-02 12:11:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"11a0c255cd45b1fe4b6665223bfc3875c1c70ba8","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-03-02 14:36:43.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"37b38c78e3fffabfe169b8708e5139496ee244e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 14:37:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"15c215e44e6eecca4982a12b1f7fce800b35850f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 16:15:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/367e4d4c73534ca2a3d365f16fd0f75d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e48dfa1ac1f437b96e0d601b7fa943c : SUCCESS in 4m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7be6d82a631248ddb020babee7173731 : SUCCESS in 8m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7df346b0d4d44bc4bc753563561c6799 : SUCCESS in 7m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51bb636637ec453a9717f5f2e8d75f8b : SUCCESS in 16m 34s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/c22a18415c88428ebbf7e972763bd32d : SUCCESS in 6m 12s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/b3aa11e269af468280b35294a5f1307a : SUCCESS in 11m 37s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/ca6c862b153442c6afd43fbf5437c6cc : SUCCESS in 41m 30s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/16c4f858f6724d9d9cb682fc6405edf7 : SUCCESS in 47m 00s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/70287b7f3c10457fbab1c5304a94b328 : SUCCESS in 51m 36s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/8b8772c45c3b491fa6fafde5a8e24531 : SUCCESS in 48m 35s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/2e97b09861344afbbe4afbbc38ebce01 : SUCCESS in 27m 22s\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/d63d04d8e11f44998b9097731728ed24 : SUCCESS in 49m 08s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/b44841fec30a4653b53e151b08b40ef4 : SUCCESS in 42m 03s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/46fde1013da9432991b40dc8a98d3aaa : SUCCESS in 35m 58s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/88654c9c22e94e019c0d069554dc78fe : SUCCESS in 1h 24m 20s","accounts_in_message":[],"_revision_number":1},{"id":"71670a261456af6856d599840df508bbcc0e47cd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 16:15:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"86bdf2455927604fd1479aeb59eddcdc383db38e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 16:34:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ec3c78b6edb40a1bdfc3e48de90b6eb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddac6f0729d64534b81bb8c4213a8165 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1bed3629ff56d4e957becf9db15421dcaefc0954","revisions":{"1bed3629ff56d4e957becf9db15421dcaefc0954":{"kind":"REWORK","_number":1,"created":"2026-02-27 23:35:06.000000000","uploader":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"ref":"refs/changes/80/978280/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/80/978280/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/80/978280/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/80/978280/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/80/978280/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/80/978280/1"}}},"commit":{"parents":[{"commit":"4f8b37f3ae474ebb7a018d2e3c7577773148baa7","subject":"Fix async periodic on redfish for servicewait","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4f8b37f3ae474ebb7a018d2e3c7577773148baa7"}]}],"author":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2026-02-26 17:58:17.000000000","tz":-360},"committer":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2026-02-27 23:35:06.000000000","tz":0},"subject":"fix inspection rules loaded from a file to not fail in logging","message":"fix inspection rules loaded from a file to not fail in logging\n\nInspection rules must have UUIDs because all the logging lines use the\nUUID. However we do not require a UUID for the rules loaded from a file\nso the code will fail when run with debug. This adds a synthetic UUID\nthat\u0027s specifically not random to make it more clear to users where\nthese rules came from.\n\nChange-Id: I4adef236f8fb09509f05141d1dfb9f5de38759bb\nSigned-off-by: Doug Goldstein \u003ccardoe@cardoe.com\u003e\n(cherry picked from commit e070ae5f98bc335abd676c0b72fdc8bac37f2244)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1bed3629ff56d4e957becf9db15421dcaefc0954"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1bed3629ff56d4e957becf9db15421dcaefc0954"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
