)]}'
{"id":"openstack%2Fironic~987439","triplet_id":"openstack%2Fironic~master~I29f24490d97026bd039c667b9de0610131ea48ee","project":"openstack/ironic","branch":"master","attention_set":{},"removed_from_attention_set":{"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2026-05-07 01:44:24.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"I29f24490d97026bd039c667b9de0610131ea48ee","subject":"Cleanup [conductor]file_url_allowed_path","status":"MERGED","created":"2026-05-05 22:30:54.000000000","updated":"2026-05-07 01:45:41.000000000","submitted":"2026-05-07 01:44:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"987439","meta_rev_id":"7f5e976d7f51d3a9f0f87ef155642676d4f30a4d","_number":987439,"virtual_id_number":987439,"owner":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-07 01:44:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":2,"date":"2026-05-06 15:44:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":2,"date":"2026-05-07 00:50:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":1,"date":"2026-05-07 00:50:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-05 23:50:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-06 00:14:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-06 15:44:50.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"},{"updated":"2026-05-07 00:50:15.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"REVIEWER"}],"messages":[{"id":"f5f60e21779551a9ada67328ab791f5ce9d5b0f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2026-05-05 22:30:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a9ab409760da1950e169c78ad01bc5d0632236dc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 23:50:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d71c1a753cdd4d4f87bc843e60726b36\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/390c22602aed44e1aa62c9e0561fe531 : FAILURE in 10m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/532abfb6bf514564b0ae1375a48cfcb2 : SUCCESS in 9m 23s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/419388b16d9f4957b19773493dd390bf : SUCCESS in 16m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5a84dbf1a8664697855dfc96025f545550d36e1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 00:14:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce29363dd8054f2c9649b951c90f0802\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/374a80d069984b218729d282916b0d16 : SUCCESS in 12m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cc4f4e4b6b24014832f2fbd446f7737 : SUCCESS in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/99882ed70a4744f7bd9f670ec4b52391 : SUCCESS in 10m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f0162d005c8a48b98c015fd8f342f15c : SUCCESS in 8m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9f599d7720ef48cb9f7642fe8f5220b5 : SUCCESS in 13m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d472a1d0ba214ae39f13df519155ed5d : SUCCESS in 9m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2af828e3bc294501a73e441f59e665a0 : SUCCESS in 14m 59s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/2fb2e69358d64ea59ca058f084d758d8 : SUCCESS in 3m 44s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/68b974016dc84bcc9a8a09936fde6d81 : SUCCESS in 11m 48s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/8d921c665a9f457da3f0434bb9c16a2f : SUCCESS in 6m 08s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/dc56dc2372df4c5e8fe94eadb9811cb9 : SUCCESS in 20m 55s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/09a21f124e7d4aa7af9a5f9a4588e5ed : SUCCESS in 58m 23s (non-voting)\n- ironic-tempest-standalone-advanced https://zuul.opendev.org/t/openstack/build/fa48f3d30a924e4083c670148ddf4046 : SUCCESS in 27m 44s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/fc5580ff065f4466ab5d87da345b7b3d : SUCCESS in 44m 43s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/e62d8e9507ee43179d1c74417a14c22a : SUCCESS in 43m 53s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/25c0f20664c54e00bf7be7fcffacb790 : SUCCESS in 20m 16s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/1f34915271e74e6da10e100547917f1f : SUCCESS in 46m 50s (non-voting)\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/d504b1f75cac421aa596181e45b2ff35 : FAILURE in 1h 08m 47s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-multinode https://zuul.opendev.org/t/openstack/build/fdb52939c69c49d2be444d42f542ef5e : SUCCESS in 1h 37m 38s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-multinode-shard https://zuul.opendev.org/t/openstack/build/0c3f9083f87646bbb2b737cf1c3cbb18 : SUCCESS in 1h 14m 30s (non-voting)\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/b3e8720d88144ebcbf07f3e4c42e8f16 : SUCCESS in 57m 23s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/d14f6711782f49c48e4b5cf1910c57bf : SUCCESS in 44m 24s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/5405fdd15688400d97003e229862f0fd : SUCCESS in 44m 10s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/a2e9eb27c3c8402e91f2ffa6b96f5239 : SUCCESS in 42m 28s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/82f191fdac644e8cb86fc73be34af8ed : SUCCESS in 1h 30m 05s (non-voting)\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/1c11ba157cfc4527b92258f3dfe51ad3 : SUCCESS in 1h 22m 57s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/7fbe0c52504b432cb7261c85eb034f25 : SUCCESS in 34m 25s\n- metal3-ironic-standalone-operator-functional https://zuul.opendev.org/t/openstack/build/fa50dd8604ab4bb6864b98abcdc7141c : FAILURE in 9m 24s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/6008eb6798ef45d9ae15a8811bf3307e : SUCCESS in 40m 10s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/937779a806ea44e2b4ae3bf9c0ab1670 : SUCCESS in 28m 18s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/f5ae15ff0d3541538211cb118a3483d3 : SUCCESS in 1h 12m 27s (non-voting)\n- bifrost-integration-on-ubuntu-noble https://zuul.opendev.org/t/openstack/build/58b8c441cbfd4708af63e7817cb2cbd4 : SUCCESS in 49m 35s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a8fb05cb7156497ebde13b833a4b7182 : SUCCESS in 1h 00m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"235979d71074a8497afb3533349f247cac4ccd47","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2026-05-06 14:46:24.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":1},{"id":"10900d9ec17ed7a054e10c137d7165c59ef477dc","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-05-06 15:44:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3485d2c279ca92ca1e6f52ad9e9919543a8627e8","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2026-05-07 00:50:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"144a6a5d53838ce4e94be0c6dae8c76e77643527","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 00:50:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"da06c7059ee1c509c7720dccd01bda1cc36264c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 01:44:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a241e8c8b0cb462d87de540a53cfb5ca\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8076fc8ecee54066be913201f4e8242b : SUCCESS in 3m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/17f9d707532642b780aca3225fd4f0a1 : SUCCESS in 5m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4834edc3e6944157a9a9263574bdc739 : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/832596208e934b37b53c70ef55da9467 : SUCCESS in 8m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1dda9af117c94cc1b8aa209965c82274 : SUCCESS in 16m 09s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/759dce391b3e4c08bf7c72f3a26eb776 : SUCCESS in 4m 49s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/06a54e4f48264370b15c39d19357fabd : SUCCESS in 8m 01s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/ae21a7eefdb046d3853a2f43e38fdc88 : SUCCESS in 22m 37s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/c2ef0f62e73e4ffd8734b89116294237 : SUCCESS in 22m 13s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/614820e992c548dbb7b56129f78b97dc : SUCCESS in 19m 19s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/651fb66c943044c2ba06299730135bb1 : SUCCESS in 46m 17s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/4774148e7fc34137afad85267e00ace8 : SUCCESS in 26m 33s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/8483bbb5ee1e4a60947e6b244878dd2a : SUCCESS in 38m 05s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/1c7e22a4e3bd4372a2a4e58d2994732f : SUCCESS in 32m 55s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/48827940eafe498d92f0635a77249def : SUCCESS in 48m 09s\n- metal3-integration https://zuul.opendev.org/t/openstack/build/c113fc8798a94866981c0bf6ada6b298 : SUCCESS in 33m 40s","accounts_in_message":[],"_revision_number":1},{"id":"24df21d6955b2a310146a934a9377ac0b20fe624","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 01:44:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7f5e976d7f51d3a9f0f87ef155642676d4f30a4d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 01:45:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5793556290140be873074311bc25d8b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9eac7be70da94a9987c739cde567c7fa : SUCCESS in 1m 09s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa07c4a65d194569a839d7ef068d22dc : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"08067f0fba10670ae8b258fa5575aa4f5812df55","revisions":{"08067f0fba10670ae8b258fa5575aa4f5812df55":{"kind":"REWORK","_number":1,"created":"2026-05-05 22:30:54.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/39/987439/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/39/987439/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/987439/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/987439/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/987439/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/39/987439/1"}}},"commit":{"parents":[{"commit":"1114d43ca8cde969e87dd99f5c70cc9c5a57fe93","subject":"Merge \"Remove deprecated configuration molds feature\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1114d43ca8cde969e87dd99f5c70cc9c5a57fe93"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2026-05-05 22:26:14.000000000","tz":-420},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2026-05-05 22:30:43.000000000","tz":-420},"subject":"Cleanup [conductor]file_url_allowed_path","message":"Cleanup [conductor]file_url_allowed_path\n\nWhen this config option was created, it was as part of an OSSA/CVE and\nnon-disruptiveness was the #1 priority. Now that we\u0027re well past that,\nwe need to make this a good default -- rather than one that just\ndescribes existing downstream use cases.\n\nOperators who need the existing paths should update their config.\n\nChange-Id: I29f24490d97026bd039c667b9de0610131ea48ee\nSigned-off-by: Jay Faulkner \u003cjay@jvf.cc\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/08067f0fba10670ae8b258fa5575aa4f5812df55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/08067f0fba10670ae8b258fa5575aa4f5812df55"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
