)]}'
{"id":"openstack%2Fironic~990085","triplet_id":"openstack%2Fironic~master~I35f85b143a77e7803db87c565d8e1a6d56815805","project":"openstack/ironic","branch":"master","topic":"correct_change_number_in_docs_for_is_root_volume","attention_set":{},"removed_from_attention_set":{"35099":{"account":{"_account_id":35099,"name":"Jakub Jelinek","email":"kuba.jelinek@nscale.com","username":"kubajj"},"last_update":"2026-05-27 07:43:14.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"I35f85b143a77e7803db87c565d8e1a6d56815805","subject":"Correct the change number linked in docs for is_root_volume","status":"MERGED","created":"2026-05-26 11:38:10.000000000","updated":"2026-05-27 07:44:28.000000000","submitted":"2026-05-27 07:43:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"990085-correct_change_number_in_docs_for_is_root_volume","meta_rev_id":"06d77004863d7e4a63f7d186686901bff20ca072","_number":990085,"virtual_id_number":990085,"owner":{"_account_id":35099,"name":"Jakub Jelinek","email":"kuba.jelinek@nscale.com","username":"kubajj"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-27 07:43:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":2,"date":"2026-05-26 21:07:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":2,"date":"2026-05-27 07:09:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":1,"date":"2026-05-27 07:09:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-26 12:04:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-26 21:07:51.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2026-05-27 07:09:09.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"}],"messages":[{"id":"b606af33d4e24f04864688c8c394f450d5d1d043","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35099,"name":"Jakub Jelinek","email":"kuba.jelinek@nscale.com","username":"kubajj"},"date":"2026-05-26 11:38:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9af202a7ac5c916fef2f047502018a58c540f473","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 12:04:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0073ddfdc9fe4136865c4e5069b83d6a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16fb833901534565a2c515e386cfb413 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42ca8b71827c4b93a73182977d737593 : SUCCESS in 14m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c3bd9a36ecc4f3db14972739f248059 : SUCCESS in 25m 18s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/2aa3f96570f54d118591de7788f874d5 : SUCCESS in 6m 47s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/25eb4a4e7e55406cbe9b3292870ea4b1 : SUCCESS in 11m 32s","accounts_in_message":[],"_revision_number":1},{"id":"89676c048dbff5a7a721137e6af07d4168d4573f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35099,"name":"Jakub Jelinek","email":"kuba.jelinek@nscale.com","username":"kubajj"},"date":"2026-05-26 12:24:26.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"0c05de9ba8b1adfa6b3fa5305adeab0476b3f964","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 12:54:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1c0271bc19a4ba49eb7f0b7543c5af8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2eee6f51e324c5b8ae6b9b88032ddb2 : SUCCESS in 7m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/376ddf9f43324c44ab46b3ebecec3c2b : SUCCESS in 7m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dcc6f9d496d64af299392ddf9d8dbcf0 : SUCCESS in 24m 58s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/6b233ce859ce4b2ca917fe3b07ea7f90 : SUCCESS in 2m 58s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/1030e2a8209942f0a5f567c837697c90 : SUCCESS in 9m 17s","accounts_in_message":[],"_revision_number":2},{"id":"04dc70405f6db266389fad0f61a8f219cdd1288e","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":35099,"name":"Jakub Jelinek","email":"kuba.jelinek@nscale.com","username":"kubajj"},"date":"2026-05-26 13:47:15.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":2},{"id":"8d1e1a67fd423caefa987daa44660ec93931226b","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2026-05-26 21:07:51.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7abb3b8034032db542eeb3ea9eac6884bb517ba6","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2026-05-27 07:09:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"61253eafd8ef881042412ca5cd4513831f154d5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 07:09:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9b53e779fbd96c87463c59df8ff005a11c810761","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 07:43:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53373ce5fb144dbf91afbd0bd2d9f50a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d507a293b46d4cc287f248ab151c2abb : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/589cd78ad9c54fcdafeb553997504482 : SUCCESS in 15m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d7adee0b2bf4d0eaa2e12652b811e0e : SUCCESS in 28m 30s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/9166a14a0ea9432f8e61e6e31bfa6756 : SUCCESS in 6m 53s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/b42603cc50c3472d86ba02f78e12a991 : SUCCESS in 10m 06s","accounts_in_message":[],"_revision_number":2},{"id":"23676a327299c0913b345bae74c12dc3f57fb72a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 07:43:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"06d77004863d7e4a63f7d186686901bff20ca072","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 07:44:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a43d5418436c49878d2a92d569613c34\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61c41dacc2b34a3aa277423df12e08eb : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"beae8cadf125f1eeb2d2e108b1c4f26c586460e8","revisions":{"0947b153f0515cc8c17348ccb7be3fd3237ad73c":{"kind":"REWORK","_number":1,"created":"2026-05-26 11:38:10.000000000","uploader":{"_account_id":35099,"name":"Jakub Jelinek","email":"kuba.jelinek@nscale.com","username":"kubajj"},"ref":"refs/changes/85/990085/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/85/990085/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/85/990085/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/85/990085/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/85/990085/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/85/990085/1"}}},"commit":{"parents":[{"commit":"c28e0d1f338d2e786a9f648155f182eaaa81828f","subject":"Merge \"novncproxy - Standardize ssl-\u003etls for parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/c28e0d1f338d2e786a9f648155f182eaaa81828f"}]}],"author":{"name":"Jakub Jelinek","email":"jakub.jelinek@cern.ch","date":"2026-05-26 11:35:19.000000000","tz":0},"committer":{"name":"Jakub Jelinek","email":"jakub.jelinek@cern.ch","date":"2026-05-26 11:35:19.000000000","tz":0},"subject":"Correct the change number linked in docs for is_root_volume","message":"Correct the change number linked in docs for is_root_volume\n\nThe change number linked in the is_root_volume docs points to change\nwhich down-played the role of is_root_volume and not implemented it.\nThis change fixes the error.\n\nChange-Id: I35f85b143a77e7803db87c565d8e1a6d56815805\nSigned-off-by: Jakub Jelinek \u003cjakub.jelinek@cern.ch\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/0947b153f0515cc8c17348ccb7be3fd3237ad73c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/0947b153f0515cc8c17348ccb7be3fd3237ad73c"}]},"branch":"refs/heads/master"},"beae8cadf125f1eeb2d2e108b1c4f26c586460e8":{"kind":"REWORK","_number":2,"created":"2026-05-26 12:24:26.000000000","uploader":{"_account_id":35099,"name":"Jakub Jelinek","email":"kuba.jelinek@nscale.com","username":"kubajj"},"ref":"refs/changes/85/990085/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/85/990085/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/85/990085/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/85/990085/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/85/990085/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/85/990085/2"}}},"commit":{"parents":[{"commit":"c28e0d1f338d2e786a9f648155f182eaaa81828f","subject":"Merge \"novncproxy - Standardize ssl-\u003etls for parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/c28e0d1f338d2e786a9f648155f182eaaa81828f"}]}],"author":{"name":"Jakub Jelinek","email":"jakub.jelinek@cern.ch","date":"2026-05-26 11:35:19.000000000","tz":0},"committer":{"name":"Jakub Jelinek","email":"jakub.jelinek@cern.ch","date":"2026-05-26 12:22:20.000000000","tz":0},"subject":"Correct the change number linked in docs for is_root_volume","message":"Correct the change number linked in docs for is_root_volume\n\nThe change number linked in the is_root_volume docs points to change\nwhich down-played the role of is_root_volume and not implemented it.\nThis change fixes the error.\n\nChange-Id: I35f85b143a77e7803db87c565d8e1a6d56815805\nSigned-off-by: Jakub Jelinek \u003cjakub.jelinek@cern.ch\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/beae8cadf125f1eeb2d2e108b1c4f26c586460e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/beae8cadf125f1eeb2d2e108b1c4f26c586460e8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
