)]}'
{"id":"openstack%2Fironic~991371","triplet_id":"openstack%2Fironic~bugfix%2F33.0~I31ee960f6f055e39dd248f54ac853e21838632df","project":"openstack/ironic","branch":"bugfix/33.0","attention_set":{},"removed_from_attention_set":{"14228":{"account":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"last_update":"2026-06-03 15:14:11.000000000","reason":"\u003cGERRIT_ACCOUNT_14228\u003e replied on the change","reason_account":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"}},"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2026-06-03 22:13:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I31ee960f6f055e39dd248f54ac853e21838632df","subject":"Ensure kernel_append_params are valid kernel parameters","status":"MERGED","created":"2026-06-03 14:26:50.000000000","updated":"2026-06-03 22:14:55.000000000","submitted":"2026-06-03 22:13:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"991371","meta_rev_id":"d2921fbfb9c9acb188a1d79d2430175a9ea1cf2c","_number":991371,"virtual_id_number":991371,"owner":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-03 22:13:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"all":[{"value":2,"date":"2026-06-03 15:14:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},{"value":2,"date":"2026-06-03 17:33:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":0,"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},{"value":1,"date":"2026-06-03 20:46:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-03 14:26:50.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"state":"REVIEWER"},{"updated":"2026-06-03 16:06:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-03 17:33:40.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"}],"messages":[{"id":"85b8422518887cbe76fc7b84ff88e79dc59b4be8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2026-06-03 14:26:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9c8dea6ec732917ecc0d94b4677cca5b5e314062","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2026-06-03 15:14:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"350739f6295b88deb299a636827a60ff81fef201","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 16:06:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af98c049f7a04e57bf4f96672f29d31b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8da6be7982334b30a0ff90797bef6bcc : SUCCESS in 9m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bad58c0b9fef4b61900d799db6e08932 : SUCCESS in 26m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f95df79d9b3f454c98942e5d3a0d9a06 : SUCCESS in 10m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2247536ec1f84a999204ea3debf1af5f : SUCCESS in 8m 51s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/24fb1a42203d41d984cadeec6a1e4076 : SUCCESS in 6m 13s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/8d3bc4647b4b4a58b22885c82e689bb9 : SUCCESS in 10m 43s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/a287c097b96e48b8aca28b13ae57bc0c : SUCCESS in 23m 10s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/52cd8b0abed345c687d66117658ba445 : SUCCESS in 39m 51s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/c48bfedb89fc4d11a3603f5cea066498 : SUCCESS in 51m 33s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/e2232511fc934bc59b002167fad60970 : SUCCESS in 25m 35s\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/9322b0c5ed9645af95689dc7674946fc : SUCCESS in 1h 22m 49s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/e0a69460e01d4ec09cd0f5f36197e6b1 : SUCCESS in 46m 36s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/a51b1a5622a24409b1772b58b1d31d0a : SUCCESS in 52m 49s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/c69cedfdf7d44bcdab0417770258c49d : SUCCESS in 1h 14m 05s","accounts_in_message":[],"_revision_number":1},{"id":"f79db538c2f07141cafcf92dc1de9b45275bba1f","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2026-06-03 17:33:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"acbc5d1d9b463f3c68d503d2962fe9bb60b8d079","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2026-06-03 20:46:50.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e2655856119e4c0f748b6a8808e013ca6c110a60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 20:47:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0308748bf169b7bc0d6fb512d5a53cd96738610c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 22:13:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e780929815164dc78d769c099a26f8aa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df4b44baadf7468b81696535cf5bcf2d : SUCCESS in 9m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e0774c18e62486ca5919c10f7fb9edb : SUCCESS in 30m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2e0d81ac42014a1ba5329b294e64bc94 : SUCCESS in 10m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e2f74fb4e0504233aeca8a7d9c6f978e : SUCCESS in 6m 52s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/b3fa6b7d56af42efaf98821ba48eb3ad : SUCCESS in 5m 29s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/6b8616aa034d4192865af305558d228b : SUCCESS in 12m 56s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/120d1d86996a4489a03026924702d52f : SUCCESS in 41m 05s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/93ed418f005647fd841b017679e1d402 : SUCCESS in 22m 53s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/7b45907713ee40a4b0ee92260ce2803c : SUCCESS in 38m 15s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/2e1fd01297404af0bed970fb3826e312 : SUCCESS in 38m 25s\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/ef3bd3cb874b401eb64fb0562fc4d667 : SUCCESS in 33m 37s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/8042bdbd441342999036ccdc2e13ff52 : SUCCESS in 34m 46s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/8ede6b4cc89c4fb4ac9a766d24885a6b : SUCCESS in 56m 45s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/0179252a92924ce38a06c3b236291376 : SUCCESS in 1h 20m 14s","accounts_in_message":[],"_revision_number":1},{"id":"f09724c65bc2e456ad6a79ee1887bef663cf48cd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 22:13:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d2921fbfb9c9acb188a1d79d2430175a9ea1cf2c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 22:14:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c17afeb356d441cc92607ffd412b8a96\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f90b74cb47444d4b41075bd778edf48 : SUCCESS in 1m 09s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc7e6b877d3e4bb69d369c140afcd847 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"00f15e5863c85ec0dbdcd4e296063d6045ef8a3d","revisions":{"00f15e5863c85ec0dbdcd4e296063d6045ef8a3d":{"kind":"REWORK","_number":1,"created":"2026-06-03 14:26:50.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/71/991371/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/71/991371/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/71/991371/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/71/991371/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/71/991371/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/71/991371/1"}}},"commit":{"parents":[{"commit":"db97efabc3ce29ecd9dd910e816cad0296bbbe80","subject":"Configure CI for bugfix/33.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/db97efabc3ce29ecd9dd910e816cad0296bbbe80"}]}],"author":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2026-05-05 22:21:57.000000000","tz":-300},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2026-06-02 19:32:48.000000000","tz":-420},"subject":"Ensure kernel_append_params are valid kernel parameters","message":"Ensure kernel_append_params are valid kernel parameters\n\nBy defining a kernel command line grammar and attemping to parse\nkernel_append_params. A successful parse indicates the input contained\nin kernel_append_params are valid kernel parameters. Unsuccessful\nparsing will raise and be rejected.\n\nThis parsing can be disabled through a new conductor configuration\noption: disable_kernel_parameter_parsing which is False by default.\n\nBasic kernel parameter sanitization (ie filtering newlines) is always\napplied in kernel_append_params since they are never valid for\ninclusion.\n\nFuture patches should extend kernel parameter parsing to all areas of\nIronic\u0027s code base in order to guarantee valid kernel parameters being\npassed along.\n\nThis commit addresses CVE-2026-46447.\n\nCloses-Bug: 2150624\nChange-Id: I31ee960f6f055e39dd248f54ac853e21838632df\nSigned-off-by: Clif Houck \u003cme@clifhouck.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/00f15e5863c85ec0dbdcd4e296063d6045ef8a3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/00f15e5863c85ec0dbdcd4e296063d6045ef8a3d"}]},"branch":"refs/heads/bugfix/33.0"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
