)]}'
{"id":"openstack%2Fironic~996469","triplet_id":"openstack%2Fironic~stable%2F2026.1~Iaa512f380926a9fc032eefea5bcdcbc5040a4739","project":"openstack/ironic","branch":"stable/2026.1","attention_set":{},"removed_from_attention_set":{"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2026-07-09 01:15:42.000000000","reason":"Change was submitted"},"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2026-07-09 00:19:57.000000000","reason":"\u003cGERRIT_ACCOUNT_11655\u003e replied on the change","reason_account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2026-07-09 01:15:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iaa512f380926a9fc032eefea5bcdcbc5040a4739","subject":"CVE-2026-44918: Prevent rehoming resources to nodes with different owner","status":"MERGED","created":"2026-07-08 13:56:23.000000000","updated":"2026-07-09 01:17:03.000000000","submitted":"2026-07-09 01:15:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"996469","meta_rev_id":"5b2165e7a3d3578ed3a0080b767dd9fc00e824f4","_number":996469,"virtual_id_number":996469,"owner":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-09 01:15:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":2,"date":"2026-07-08 19:25:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":2,"date":"2026-07-09 00:19:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":1,"date":"2026-07-09 00:19:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}],"CC":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-08 13:56:23.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"CC"},{"updated":"2026-07-08 14:55:06.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2026-07-08 15:57:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-09 00:19:57.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"}],"messages":[{"id":"39e53f1c11c44debffd7dfe790d12b05163e6303","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2026-07-08 13:56:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2092b2ef34214c620d468e3ee2ecfb6d8ab21e2","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2026-07-08 14:55:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"78d1d89f873d3dea6eaf5f4309797e0a7e2e3f7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 15:57:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a6c823a70b1c487cb3d53afd413c17cd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7bacacf032234a5cb04b3f2af3ccbc92 : FAILURE in 14m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4f79f8c4715424d8e7d7a3e76adf0fc : SUCCESS in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/984efd6da11c412faf3e0d04f57b958b : SUCCESS in 6m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1eaa48e1ccf94cd087759fa503005c84 : SUCCESS in 6m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/abf0486d8c5349bdb6c1597812009c63 : SUCCESS in 14m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24b5047f7c3749f0a81c98396603ab29 : SUCCESS in 8m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2899ab39786e48b8a11f88fb9070e574 : SUCCESS in 21m 05s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/5ac0ba58854b4f3d846b5749afbacc83 : SUCCESS in 5m 32s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/299fcc92e06a4df9b37a894ead0a46dd : SUCCESS in 6m 08s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/2d97191863e34e32bbcd1aee4371b7c8 : SUCCESS in 9m 15s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/b081ffdee2b7431abd0d35a06319a749 : FAILURE in 20m 41s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/ab03e3fd2b0a4f7e95c1f53038689927 : FAILURE in 40m 40s (non-voting)\n- ironic-tempest-standalone-advanced https://zuul.opendev.org/t/openstack/build/e4af4a791ad64b35b67f254df4748775 : POST_FAILURE in 23m 16s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/0244f64ef24541daa7f06a359e83b924 : FAILURE in 54m 34s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/fe5b25cbcca649b89ecafc1650b45b96 : FAILURE in 1h 12m 01s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/7e8648eb2a1544338f106c6e06796e8a : FAILURE in 1h 15m 28s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/cad6063a6e574688a5492b5c8eafd7c1 : FAILURE in 52m 34s (non-voting)\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/2b07b5ac29ed45259e4fe945b6567445 : FAILURE in 1h 05m 38s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-multinode https://zuul.opendev.org/t/openstack/build/37d3dc1e7b5b4e598f77cf5677728577 : FAILURE in 1h 50m 11s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-multinode-shard https://zuul.opendev.org/t/openstack/build/6722c77142814192890445c9c1d803cd : FAILURE in 1h 49m 18s (non-voting)\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/0adb2aa1eb6b4a808b622c859c8ec6fc : FAILURE in 1h 09m 42s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/5df4d471e46a4dbc908cddc06e12ee5d : FAILURE in 1h 05m 43s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/85190952eb5844e4b5dba8115f84eac4 : FAILURE in 1h 06m 33s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/1e252ed9502f445e8a74542a29fba7be : FAILURE in 1h 30m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bcbde3bf96ba491aafeea722fd54508e : SUCCESS in 57m 46s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/6f5152e492c64da6ba7fc7453d2c48f5 : FAILURE in 40m 28s (non-voting)\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/dbe55f35e3ad452bbecbf25df0d86132 : FAILURE in 36m 57s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/e3abdbef02b6440e88a1ed3f827ddbd2 : FAILURE in 28m 31s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/2d9d64f50fde4e0dbe0232513903289e : FAILURE in 57m 16s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/d671384b1dee48a49ad659bd1b420a99 : FAILURE in 1h 11m 49s (non-voting)\n- bifrost-integration-on-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af4a57016d664f87a8f4f88e290cb9f9 : FAILURE in 18m 04s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/87876d3451f2467db07bedddedaa4464 : FAILURE in 36m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4cb59879b9eac75709c367915a5e3db35b3fef4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2026-07-08 19:23:08.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"edc7bed0e3e75cdebc8179ce683adcbed3aeef9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2026-07-08 19:25:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"10dedfc05a1e8f6b7aeee591c6284965a06ee6a0","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2026-07-08 19:25:43.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"e3b73c10427b03ce58f41d2fd7f19ddb9ccd0756","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 21:35:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a14e3f80dfb4897a6feda801b7c481d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf71ea15cfc64d5197005b3756535406 : SUCCESS in 12m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72d5260c51fa48a6b3fa93e0f4fe9b23 : SUCCESS in 7m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7ac34c8654214dbeb23295d9ab17feea : SUCCESS in 7m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/84cfe21810c44d40a000e1ec61804926 : SUCCESS in 7m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5fa58cde470c4f718895aac0670f1288 : SUCCESS in 13m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9259a87fbfa4b3b9db1c668c0ea8d5d : SUCCESS in 18m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d84a1539359f4525bc73f1ec3e3c46b8 : SUCCESS in 24m 09s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/ae2fbed36af844fba9d0d18f396c5f1c : SUCCESS in 5m 50s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/7aa55168801b45808708a9592bb0ec74 : SUCCESS in 9m 42s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/9a644a7cd5ae4267a8280912c63d39b4 : SUCCESS in 8m 41s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/aac325b7619b49dc9646ece6fba6c5d7 : SUCCESS in 43m 32s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/61b13fda06914650969281fdefbfaf12 : SUCCESS in 54m 06s (non-voting)\n- ironic-tempest-standalone-advanced https://zuul.opendev.org/t/openstack/build/4bf81880df3247f696e455b8ad68d674 : SUCCESS in 1h 24m 28s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/32128845d71b4366bd49aeff6f2d8138 : SUCCESS in 46m 49s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/ad8eeac2c4c44cf9b76050c4455399d4 : SUCCESS in 49m 04s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/6ad28efdde854685bf87588ff0ab8663 : SUCCESS in 26m 32s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/0006a91a1fa74523b7e4d2e648995f93 : SUCCESS in 57m 11s (non-voting)\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/d72566bd2d934db198640f59c93e9dbc : FAILURE in 1h 11m 32s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-multinode https://zuul.opendev.org/t/openstack/build/1d2e5c03634f435291053dceb1e6abf0 : SUCCESS in 2h 05m 31s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-multinode-shard https://zuul.opendev.org/t/openstack/build/bfec2deb6e734549bb129c2aa5a7fe15 : FAILURE in 1h 39m 53s (non-voting)\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/3288005271ae4df8a17f921c531b4fc5 : SUCCESS in 34m 55s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/a21b5e5b44f94dc2a54449d59ab646f9 : SUCCESS in 34m 46s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/1ac9a43847f441d587c4e2474083e7bd : SUCCESS in 26m 08s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/4b343a5e381e4af09d41f950bc77c7dc : SUCCESS in 1h 13m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/86cce8bcf1cf45c1bac549e724a1f67f : SUCCESS in 1h 05m 30s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/f09fa3c8afd74357b846e40f28349908 : SUCCESS in 53m 40s (non-voting)\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/fa028dba040c4e6f8941d88628f926c1 : SUCCESS in 1h 13m 31s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/302669bc94534db7ae54157fdc160b56 : FAILURE in 28m 56s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/b5a2bfa425ec4cf88e33749c46247ec7 : SUCCESS in 37m 17s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/f4826171250746deac34256a364d3517 : SUCCESS in 22m 40s (non-voting)\n- bifrost-integration-on-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d5d2ea757fd4dea8a30d406486c45fe : SUCCESS in 46m 48s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f40a847a8fd342bda6649ad57f45971d : SUCCESS in 1h 01m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7e83cdbc0784b4f4356f9f10cfe1a604ceade7b5","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2026-07-09 00:19:57.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"05d8939e866e14913874bb7fdc34da41db921671","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 00:20:28.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f4fce37dc10526ef53e205d55dbbd5d892971430","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 01:15:42.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afc78a96239041c6bb49229a7400e6b9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10dcd13e5eee4b67ac8fd9d253f294bf : SUCCESS in 6m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b8cd39617cf54599a08d67e29b18ecfd : SUCCESS in 8m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1ec1f89de3b1429497da034809628dc0 : SUCCESS in 7m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7eb050119f8f41d0b23076993c3ae802 : SUCCESS in 17m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/337dc272d244416da433cbe61989c767 : SUCCESS in 20m 19s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/d5b61327ecd6489689dd29e90bffb593 : SUCCESS in 5m 07s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/7745092538a248f79b34a11b6785f59a : SUCCESS in 15m 26s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/5cba67d4baa74eb3b328acc0ea62a6d8 : SUCCESS in 37m 20s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/0be666ade94447eba42c2d250e676cfe : SUCCESS in 44m 23s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/b283563ce47c410996fad6297135ec42 : SUCCESS in 46m 52s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/238b23c4025a481386e3e62269dee28c : SUCCESS in 36m 41s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/01847ec8d7834624bd123da33b776f81 : SUCCESS in 28m 13s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/17704d97cc984ee7adf699a0773433a6 : SUCCESS in 40m 00s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/0897a3fab30242e49d90dd4da136e94c : SUCCESS in 50m 32s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/2a9f7195a25a43a9956cdb1ecf280a6d : SUCCESS in 50m 38s","accounts_in_message":[],"_revision_number":3},{"id":"99d75fa0a1d6c66fc8017cd56175f8ea63ac25aa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 01:15:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"5b2165e7a3d3578ed3a0080b767dd9fc00e824f4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 01:17:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0ef3d25c2be46de9e3f7fa76c70ac8c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47a72a0809f74708a4c1cc953eb3860f : SUCCESS in 1m 11s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c1944874501468787849b42f9069120 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1fe9279f4eb3fba3d49c572d42c45cfec2a37283","revisions":{"fe21cc27902ec4a049162b1a3df4a6ec67c4608b":{"kind":"REWORK","_number":1,"created":"2026-07-08 13:56:23.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/69/996469/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/69/996469/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/69/996469/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/69/996469/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/69/996469/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/69/996469/1"}}},"commit":{"parents":[{"commit":"730e2c402b944c326a33c4694b8417288d7442af","subject":"Fix bootfile_name_by_arch to consider boot mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/730e2c402b944c326a33c4694b8417288d7442af"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2026-05-04 15:14:07.000000000","tz":120},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2026-07-08 13:56:16.000000000","tz":-420},"subject":"CVE-2026-44918: Prevent rehoming resources to nodes with different owner","message":"CVE-2026-44918: Prevent rehoming resources to nodes with different owner\n\nIronic has a few similar patterns where only the source node is verified\nduring a create or update operation:\n1. A node can be created with a parent from a different project\n2. Node can be updated to have a parent from a different project\n3. Ports, port groups, volume targets and volume connectors can be\n   patched to be owned by a node from a different project.\n\nThis patch addresses all these cases and introduces a single helper to\ncover all of them.\n\nAlso makes sure that baremetal:node:get is checked even if the parent\nnode ID is a UUID (it was missing before).\n\nRelated-Bug: 2150450\nChange-Id: Iaa512f380926a9fc032eefea5bcdcbc5040a4739\nSigned-off-by: Julia Kreger \u003cjuliaashleykreger@gmail.com\u003e\nSigned-off-by: Jay Faulkner \u003cjay@jvf.cc\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/fe21cc27902ec4a049162b1a3df4a6ec67c4608b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/fe21cc27902ec4a049162b1a3df4a6ec67c4608b"}]},"branch":"refs/heads/stable/2026.1"},"87b570201adc1b138aa786c73f555bff1df65da0":{"kind":"REWORK","_number":2,"created":"2026-07-08 19:23:08.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/69/996469/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/69/996469/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/69/996469/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/69/996469/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/69/996469/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/69/996469/2"}}},"commit":{"parents":[{"commit":"c572103132ec50fffe98f179beb1a591e534e88f","subject":"security: block vendor.send_raw (CVE-2026-54423)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/c572103132ec50fffe98f179beb1a591e534e88f"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2026-05-04 15:14:07.000000000","tz":120},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2026-07-08 19:22:57.000000000","tz":-420},"subject":"CVE-2026-44918: Prevent rehoming resources to nodes with different owner","message":"CVE-2026-44918: Prevent rehoming resources to nodes with different owner\n\nIronic has a few similar patterns where only the source node is verified\nduring a create or update operation:\n1. A node can be created with a parent from a different project\n2. Node can be updated to have a parent from a different project\n3. Ports, port groups, volume targets and volume connectors can be\n   patched to be owned by a node from a different project.\n\nThis patch addresses all these cases and introduces a single helper to\ncover all of them.\n\nAlso makes sure that baremetal:node:get is checked even if the parent\nnode ID is a UUID (it was missing before).\n\nRelated-Bug: 2150450\nChange-Id: Iaa512f380926a9fc032eefea5bcdcbc5040a4739\nSigned-off-by: Julia Kreger \u003cjuliaashleykreger@gmail.com\u003e\nSigned-off-by: Jay Faulkner \u003cjay@jvf.cc\u003e\n(cherry picked from commit b58d4adc43551323b7ae8d5252636d796e32b949)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/87b570201adc1b138aa786c73f555bff1df65da0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/87b570201adc1b138aa786c73f555bff1df65da0"}]},"branch":"refs/heads/stable/2026.1"},"1fe9279f4eb3fba3d49c572d42c45cfec2a37283":{"kind":"REWORK","_number":3,"created":"2026-07-08 19:25:16.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/69/996469/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/69/996469/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/69/996469/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/69/996469/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/69/996469/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/69/996469/3"}}},"commit":{"parents":[{"commit":"c572103132ec50fffe98f179beb1a591e534e88f","subject":"security: block vendor.send_raw (CVE-2026-54423)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/c572103132ec50fffe98f179beb1a591e534e88f"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2026-05-04 15:14:07.000000000","tz":120},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2026-07-08 19:25:10.000000000","tz":-420},"subject":"CVE-2026-44918: Prevent rehoming resources to nodes with different owner","message":"CVE-2026-44918: Prevent rehoming resources to nodes with different owner\n\nIronic has a few similar patterns where only the source node is verified\nduring a create or update operation:\n1. A node can be created with a parent from a different project\n2. Node can be updated to have a parent from a different project\n3. Ports, port groups, volume targets and volume connectors can be\n   patched to be owned by a node from a different project.\n\nThis patch addresses all these cases and introduces a single helper to\ncover all of them.\n\nAlso makes sure that baremetal:node:get is checked even if the parent\nnode ID is a UUID (it was missing before).\n\nRelated-Bug: 2150450\nChange-Id: Iaa512f380926a9fc032eefea5bcdcbc5040a4739\nSigned-off-by: Julia Kreger \u003cjuliaashleykreger@gmail.com\u003e\nSigned-off-by: Jay Faulkner \u003cjay@jvf.cc\u003e\n(cherry picked from commit b58d4adc43551323b7ae8d5252636d796e32b949)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1fe9279f4eb3fba3d49c572d42c45cfec2a37283"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1fe9279f4eb3fba3d49c572d42c45cfec2a37283"}]},"branch":"refs/heads/stable/2026.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
