)]}'
{"id":"openstack%2Fironic~996471","triplet_id":"openstack%2Fironic~bugfix%2F34.0~Iaa512f380926a9fc032eefea5bcdcbc5040a4739","project":"openstack/ironic","branch":"bugfix/34.0","attention_set":{},"removed_from_attention_set":{"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2026-07-09 01:36:59.000000000","reason":"Change was submitted"},"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2026-07-09 00:22:57.000000000","reason":"\u003cGERRIT_ACCOUNT_11655\u003e replied on the change","reason_account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2026-07-09 01:36:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iaa512f380926a9fc032eefea5bcdcbc5040a4739","subject":"CVE-2026-44918: Prevent rehoming resources to nodes with different owner","status":"MERGED","created":"2026-07-08 13:57:14.000000000","updated":"2026-07-09 01:38:05.000000000","submitted":"2026-07-09 01:36:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"996471","cherry_pick_of_change":996469,"cherry_pick_of_patch_set":3,"meta_rev_id":"9371516dcdeb69d83732f51b34bbc607cfdde1a3","_number":996471,"virtual_id_number":996471,"owner":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-09 01:36:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":2,"date":"2026-07-08 19:27:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":2,"date":"2026-07-09 00:22:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":1,"date":"2026-07-09 00:22:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}],"CC":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-08 13:57:14.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"CC"},{"updated":"2026-07-08 14:55:02.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2026-07-08 15:29:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-09 00:22:57.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"}],"messages":[{"id":"961e5254423f2ee213247ca3177a72fe3da27f26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2026-07-08 13:57:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c9562b7cdbb6fbbc5daf85666e14f5f5b6043662","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2026-07-08 14:55:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8bb84663dfb1063e82c72400e0e454b85fd4a094","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 15:29:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ea1dc80d6f1a48cc83a7932579ad6666\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee84f417756a464f8776e861a7510ccb : SUCCESS in 17m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a87bdd4bd45b4930aaf2f6bd8f9f9a29 : SUCCESS in 21m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e1758394904047a4847d521dfadf2fab : SUCCESS in 6m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d2e1235bb97e4ab7a7dac7f69d69dff9 : SUCCESS in 8m 14s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/1b00e18ec7a04ead8b7acf00c7cbb1b5 : SUCCESS in 6m 10s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/531259f34d9044d0846d55f80e2902e4 : FAILURE in 17m 41s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/9746d59e72764cf9bdbb76ec7a226111 : FAILURE in 24m 41s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/3c1829c9265c44a1ab8fc5fd63697aa4 : FAILURE in 1h 06m 56s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/17e8b17e4c1e4ea7893b8df3aaa56bea : FAILURE in 1h 01m 35s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/bf2124f2f00e4ef7bdbadf0a59bd8d7c : FAILURE in 1h 11m 13s\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/785878238bce47bbb669c6f6f48c8e1e : FAILURE in 1h 21m 54s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/5bc817326a714e50b3783386caafe652 : FAILURE in 1h 10m 02s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/7f812a203fd5436c9c8e33dca4fac379 : FAILURE in 1h 08m 09s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/267133c347704089836125e9a494ae30 : FAILURE in 1h 21m 20s","accounts_in_message":[],"_revision_number":1},{"id":"bd2668e0edc1edb79714da59b96fcd7d4f75e414","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2026-07-08 19:27:14.000000000","message":"Patch Set 2: Cherry Picked from branch stable/2026.1.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"50100b2f53a284e1f07033e6fe5e53fa28165946","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2026-07-08 19:27:35.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4fc6e3bf4ccd6daca095fca74d8a37a97bc531aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 21:03:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7bf3d06481e4473ab07363fd1e67b93e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dc32ad2da10425e8852459dfb626e2d : SUCCESS in 16m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ca6c7cdd7f14a07a73d1d6631adda01 : SUCCESS in 8m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5f3a47edd77745418b420a7fa87ba28d : SUCCESS in 7m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d7d40f80bfe94ba28b246f884a194145 : SUCCESS in 4m 55s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/4dd1e8b32b284f05a38a749f9125ed50 : SUCCESS in 6m 18s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/913f01c9b6844849b4d64019363d2b4d : SUCCESS in 9m 59s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/2374ddbe8fa34e0da2004c12ca28e813 : SUCCESS in 39m 47s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/80af164c48e349b390bec4ebec10c117 : SUCCESS in 51m 00s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/96e19d95bd9a4283bf9a729ba89389a7 : SUCCESS in 49m 04s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/6668e509b73b4af7950a8e08b196b51d : SUCCESS in 51m 52s\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/2d7219adf9564b17a36be007d945cf79 : SUCCESS in 1h 29m 28s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/9e0d386af358414dad678587c51a66d0 : SUCCESS in 20m 35s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/3df3beae72654bdebce3b7c58b8e2b02 : SUCCESS in 38m 11s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/411129b67121470b9d3bceb0ace7677e : SUCCESS in 34m 14s","accounts_in_message":[],"_revision_number":2},{"id":"e35abb9885664918973ce20d1f99871b3e902e22","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2026-07-09 00:22:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b1f1de7092019a39228ac28282630580362864a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 00:23:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fc5c7c66d3b610ab0824e03cf8d22a5f743dca91","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 01:36:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64313122e5644e5bba22146de108caa2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f278aa49e4434a3aae3cfed37c82a2e7 : SUCCESS in 9m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ecbc76687461408ebc773aa090712539 : SUCCESS in 16m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fe0a65260126440f8bf85292e188e67a : SUCCESS in 6m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3f1176ab188945e792342885886b9436 : SUCCESS in 8m 36s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/8eb8cdd00d7249af8606815099420dff : SUCCESS in 6m 38s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/4899ba879dbf49d69913b254015459ce : SUCCESS in 6m 20s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/ca0d8320303849e982d73f6069f40649 : SUCCESS in 38m 59s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/1b60f20ab2ba4fc289adda59eb390311 : SUCCESS in 46m 48s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/7dbcea81b13b479181b895729d6038aa : SUCCESS in 54m 23s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/f401f27430b640098a391b8447ce76c1 : SUCCESS in 52m 24s\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/f731d5b1eb8b4d178ed71452c57e10c2 : SUCCESS in 35m 16s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/e74b0929d1d9433da992cc489a898ab8 : SUCCESS in 31m 04s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/99f6f210bfa14de7a6aa9325bf70e1fd : SUCCESS in 52m 04s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/66b252b652fc4618908484cc973600da : SUCCESS in 32m 07s","accounts_in_message":[],"_revision_number":2},{"id":"3f7ee7d2423ce939693e9128c8aed78c641d5c2d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 01:36:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9371516dcdeb69d83732f51b34bbc607cfdde1a3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 01:38:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89650e6eee6a4fe0ae9a9c896ab58939\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5bdb9f1cd32471fbeca9a95feb2bae0 : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a946f7363c84983af2a0a7ab34e8c35 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1288afe8c46a3c09c5daef6deef2299e7e7c1b84","revisions":{"cfa2aeaa22596a4cad546c821a366c19008307ee":{"kind":"REWORK","_number":1,"created":"2026-07-08 13:57:14.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/71/996471/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/71/996471/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/71/996471/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/71/996471/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/71/996471/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/71/996471/1"}}},"commit":{"parents":[{"commit":"b23998f830832953e53a8cb81636d8599760607e","subject":"Merge \"Add operator-configurable step disallow lists\" into bugfix/34.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/b23998f830832953e53a8cb81636d8599760607e"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2026-05-04 15:14:07.000000000","tz":120},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2026-07-08 13:57:07.000000000","tz":-420},"subject":"CVE-2026-44918: Prevent rehoming resources to nodes with different owner","message":"CVE-2026-44918: Prevent rehoming resources to nodes with different owner\n\nIronic has a few similar patterns where only the source node is verified\nduring a create or update operation:\n1. A node can be created with a parent from a different project\n2. Node can be updated to have a parent from a different project\n3. Ports, port groups, volume targets and volume connectors can be\n   patched to be owned by a node from a different project.\n\nThis patch addresses all these cases and introduces a single helper to\ncover all of them.\n\nAlso makes sure that baremetal:node:get is checked even if the parent\nnode ID is a UUID (it was missing before).\n\nRelated-Bug: 2150450\nChange-Id: Iaa512f380926a9fc032eefea5bcdcbc5040a4739\nSigned-off-by: Julia Kreger \u003cjuliaashleykreger@gmail.com\u003e\nSigned-off-by: Jay Faulkner \u003cjay@jvf.cc\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/cfa2aeaa22596a4cad546c821a366c19008307ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/cfa2aeaa22596a4cad546c821a366c19008307ee"}]},"branch":"refs/heads/bugfix/34.0"},"1288afe8c46a3c09c5daef6deef2299e7e7c1b84":{"kind":"REWORK","_number":2,"created":"2026-07-08 19:27:14.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/71/996471/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/71/996471/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/71/996471/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/71/996471/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/71/996471/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/71/996471/2"}}},"commit":{"parents":[{"commit":"b23998f830832953e53a8cb81636d8599760607e","subject":"Merge \"Add operator-configurable step disallow lists\" into bugfix/34.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/b23998f830832953e53a8cb81636d8599760607e"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2026-05-04 15:14:07.000000000","tz":120},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2026-07-08 19:27:14.000000000","tz":0},"subject":"CVE-2026-44918: Prevent rehoming resources to nodes with different owner","message":"CVE-2026-44918: Prevent rehoming resources to nodes with different owner\n\nIronic has a few similar patterns where only the source node is verified\nduring a create or update operation:\n1. A node can be created with a parent from a different project\n2. Node can be updated to have a parent from a different project\n3. Ports, port groups, volume targets and volume connectors can be\n   patched to be owned by a node from a different project.\n\nThis patch addresses all these cases and introduces a single helper to\ncover all of them.\n\nAlso makes sure that baremetal:node:get is checked even if the parent\nnode ID is a UUID (it was missing before).\n\nRelated-Bug: 2150450\nChange-Id: Iaa512f380926a9fc032eefea5bcdcbc5040a4739\nSigned-off-by: Julia Kreger \u003cjuliaashleykreger@gmail.com\u003e\nSigned-off-by: Jay Faulkner \u003cjay@jvf.cc\u003e\n(cherry picked from commit b58d4adc43551323b7ae8d5252636d796e32b949)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1288afe8c46a3c09c5daef6deef2299e7e7c1b84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1288afe8c46a3c09c5daef6deef2299e7e7c1b84"}]},"branch":"refs/heads/bugfix/34.0","conflicts":{"base":"c572103132ec50fffe98f179beb1a591e534e88f","ours":"b23998f830832953e53a8cb81636d8599760607e","theirs":"1fe9279f4eb3fba3d49c572d42c45cfec2a37283","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
