)]}'
{"id":"openstack%2Fironic~997136","triplet_id":"openstack%2Fironic~stable%2F2026.1~I0669782bacfb95b42dfa300e4f91ece62893847c","project":"openstack/ironic","branch":"stable/2026.1","topic":"use-service-auth-for-port-operations-stable/2026.1","attention_set":{},"removed_from_attention_set":{"5890":{"account":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"last_update":"2026-07-14 02:36:46.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"I0669782bacfb95b42dfa300e4f91ece62893847c","subject":"Use service auth for Neutron port operations","status":"MERGED","created":"2026-07-13 21:44:41.000000000","updated":"2026-07-14 02:39:18.000000000","submitted":"2026-07-14 02:36:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"997136-use-service-auth-for-port-operations-stable/2026.1","cherry_pick_of_change":989431,"cherry_pick_of_patch_set":2,"meta_rev_id":"7421a98149da3a1854f9e940ac58b0391920d23d","_number":997136,"virtual_id_number":997136,"owner":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-14 02:36:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"all":[{"value":2,"date":"2026-07-13 22:08:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-07-14 01:24:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-07-14 01:24:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":30314,"name":"Sebastian Lohff","email":"sebastian.lohff@sap.com","username":"seba"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-13 21:44:41.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":30314,"name":"Sebastian Lohff","email":"sebastian.lohff@sap.com","username":"seba"},"state":"CC"},{"updated":"2026-07-13 22:08:00.000000000","updated_by":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"reviewer":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"state":"REVIEWER"},{"updated":"2026-07-13 23:46:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-14 01:24:45.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"}],"messages":[{"id":"680d9f2d6ef0932e87cbe59f116b4c1b868cb811","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-07-13 21:44:41.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a86631600e808927d86e18c9596db6c256351ec9","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-07-13 21:45:13.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":1},{"id":"81c26cc1aadcceceab28fda9b9d171dfeae0fa43","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2026-07-13 22:08:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b044af0048027b66b32c42d1533f75a3d703c138","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-13 23:46:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efbec384b45140b7826af52750f7784c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a81b37745bab416382427fa9f101f9ce : SUCCESS in 9m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b9d0bda3c5840bbb8abb8c6cbc01e99 : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2134034afd5e44458533703e4f9dd2fd : SUCCESS in 10m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1f752146d664425ab1bdfb113a55479b : SUCCESS in 9m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/62194e29fdea4046ba311437e1972aa0 : SUCCESS in 15m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a8ade7915724b44ac0636a22552e7af : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aea4fcd6efdd48328674596bd4cfc2c9 : SUCCESS in 14m 33s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/252e5d624ffd438ab222aaec72035561 : SUCCESS in 3m 01s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/a78526da6ab1424d9a37dd43fe2853a4 : SUCCESS in 5m 46s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/54d89c00edbc4c1b9019e1021ab61339 : SUCCESS in 5m 54s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/a86392c18b1948b29464caca6793b6dd : SUCCESS in 39m 04s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/c84fd2dc8efa4049862f29218068d5bc : SUCCESS in 59m 01s (non-voting)\n- ironic-tempest-standalone-advanced https://zuul.opendev.org/t/openstack/build/ebcec22d907c4d7195659c6bb0647509 : SUCCESS in 1h 20m 17s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/db5cea9055304fcfa4c9f417271d6f2c : SUCCESS in 23m 22s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/96706742c72e4cbd8722e252408cc2b6 : SUCCESS in 52m 35s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/6930ac1402794a579f1075d2c2bb829c : SUCCESS in 25m 41s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/a164beac24e74773b12f945239ab3b9e : SUCCESS in 29m 42s (non-voting)\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/e22213cf46354e08ad27c3652e230364 : FAILURE in 1h 01m 57s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-multinode https://zuul.opendev.org/t/openstack/build/fdcec1b26dac4749901588113fc73e51 : SUCCESS in 1h 55m 12s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-multinode-shard https://zuul.opendev.org/t/openstack/build/1dbbe170c5084baa989d439355990c69 : SUCCESS in 1h 30m 43s (non-voting)\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/edfad9d9b5aa4cb29e5a2e577b8b0854 : SUCCESS in 1h 00m 01s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/c327a65b52a24e1cacf173c4d9e73fe4 : SUCCESS in 39m 27s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/60e20ab9a5c84a99886d4fa847e188a9 : SUCCESS in 54m 35s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/47085bf800f74084b2a0df3435ff9dfe : SUCCESS in 49m 38s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/b883b1e49a3d4974be9ca1db7bad634c : SUCCESS in 48m 00s (non-voting)\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/d28f3fe8b5194c8e84b87829aefe7583 : SUCCESS in 1h 19m 54s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/41abba2554304ab5b911bb35a003da9b : FAILURE in 26m 50s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/97a6e73d73304047b2686301ef00d334 : SUCCESS in 20m 47s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/b866f5c6c73c4d01bc6dc2acfd1a8709 : SUCCESS in 59m 27s (non-voting)\n- bifrost-integration-on-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1918b3088eb24f63835a312e7331da73 : SUCCESS in 37m 31s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/62abaf24f10c447aa9fd9a9cd272f7ee : SUCCESS in 25m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"106d2a953516827a090a412755384ac1704e759f","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2026-07-14 01:24:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"57bb88ac81eb08b71102f837044e851665dfbfa8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 01:25:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9db8235c9732fc9165a21581f6c0b9567680d259","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 02:36:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8c3bba631254beb8683aa26ff1b4ddf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e874d86624e544148f726de2556fe70a : SUCCESS in 6m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/715889b4d2cb49448e4e69bcf937d902 : SUCCESS in 9m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6af63556c1744dada9ffe28b5b214fc9 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cce5463a2a004fec8ae772958446fd1b : SUCCESS in 8m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/307c67a88174472ba94bade844efd6e5 : SUCCESS in 10m 00s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/0cc78c5f6daa4cd881f896c2b1db1f88 : SUCCESS in 3m 44s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/43bd5c135d78495cb4eebc4a29571a5a : SUCCESS in 5m 46s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/b662fb0fbaa84bef8d543ae8810b478e : SUCCESS in 20m 56s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/eab032e6b339412d843afda215251f47 : SUCCESS in 33m 52s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/2aeaf46a68c04909a875eb60522e5d2d : SUCCESS in 36m 09s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/b073371dca124f0a9c852f1500507df9 : SUCCESS in 36m 47s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/529fa1121a1049858e48dab463af5354 : SUCCESS in 32m 57s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/125554a8d8834237a53fcbac2155ca2c : SUCCESS in 30m 15s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/f7e89268d5f241179cd53cc815309d77 : SUCCESS in 56m 20s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/eb5ca8d64b264d68b272e8b0ac5cdbf6 : SUCCESS in 1h 09m 28s","accounts_in_message":[],"_revision_number":1},{"id":"9376d1c37615caf586f2c4e0efffcfe8c1a147d8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 02:36:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7421a98149da3a1854f9e940ac58b0391920d23d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 02:39:18.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5f8a3cf0467e484297317fe394cfc186\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5198c288be1c4b4c933f712820cfe423 : SUCCESS in 1m 32s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c9170297e81b463fbcbc91b2a78eda74 : FAILURE in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"61caa57beff1417c84ee6b8e9fd289166b192ce0","revisions":{"61caa57beff1417c84ee6b8e9fd289166b192ce0":{"kind":"REWORK","_number":1,"created":"2026-07-13 21:44:41.000000000","uploader":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"ref":"refs/changes/36/997136/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/36/997136/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/36/997136/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/36/997136/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/36/997136/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/36/997136/1"}}},"commit":{"parents":[{"commit":"4aeead39cf7d83abde1c740a231cddd7b149a418","subject":"Merge \"Fix BIOS firmware update not applied when TaskMonitor disappears\" into stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4aeead39cf7d83abde1c740a231cddd7b149a418"}]}],"author":{"name":"Sebastian Lohff","email":"sebastian.lohff@sap.com","date":"2026-05-18 12:32:57.000000000","tz":120},"committer":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2026-07-13 21:44:41.000000000","tz":0},"subject":"Use service auth for Neutron port operations","message":"Use service auth for Neutron port operations\n\nWhen a user triggers a manual operation like cleaning, most requests\ntowards the Neutron API are made with the authentication parameters\nprovided by the user\u0027s context. In case of port creation this can lead\nto ports being created with an unexpected project id (taken from the\nproject scope of the current user). To prevent this, we now use the\nservice credentials from config for add_ports_to_network() and\nremove_neutron_ports().\n\nCloses-Bug: #2121702\nChange-Id: I0669782bacfb95b42dfa300e4f91ece62893847c\nSigned-off-by: Sebastian Lohff \u003csebastian.lohff@sap.com\u003e\n(cherry picked from commit 27083f06332ecc48d17911a3c75f2334b8723444)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/61caa57beff1417c84ee6b8e9fd289166b192ce0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/61caa57beff1417c84ee6b8e9fd289166b192ce0"}]},"branch":"refs/heads/stable/2026.1","conflicts":{"base":"072619978e976d3bb9a0d883df648dc78c719e0e","ours":"4aeead39cf7d83abde1c740a231cddd7b149a418","theirs":"27083f06332ecc48d17911a3c75f2334b8723444","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
