)]}'
{"id":"openstack%2Fironic~998019","triplet_id":"openstack%2Fironic~stable%2F2025.2~I9d48a6827455d115e9e1176f3dee41c9e07251b8","project":"openstack/ironic","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2026-07-29 15:20:20.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"I9d48a6827455d115e9e1176f3dee41c9e07251b8","subject":"security: Fix poisoning of the allowed origin list","status":"MERGED","created":"2026-07-20 16:09:56.000000000","updated":"2026-07-29 15:24:46.000000000","submitted":"2026-07-29 15:20:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"998019","cherry_pick_of_change":998018,"cherry_pick_of_patch_set":1,"meta_rev_id":"2535b37f08d83c044842fe0efff1c88b9d1b172b","_number":998019,"virtual_id_number":998019,"owner":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-29 15:20:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":2,"date":"2026-07-28 22:03:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":2,"date":"2026-07-29 13:50:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":1,"date":"2026-07-29 13:50:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-20 18:06:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-29 13:50:56.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"}],"messages":[{"id":"9357a7b4a536f11c815b72ec50a1727a81d5c443","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2026-07-20 16:09:56.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2026.1.","accounts_in_message":[],"_revision_number":1},{"id":"d059c944cef155643f8cbca1e54d9067136b4607","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 18:06:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/66b70ef15b404733a87966aed65d42e7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be72401c76c04e958775c851d30852c0 : SUCCESS in 14m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecfca8ce9116490bb2bbf26ace8dd749 : SUCCESS in 7m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f475ba91880e4ab8bb6b6a34d3eadbf5 : SUCCESS in 9m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/144113a89ad84de2a4cd57e27349d96b : SUCCESS in 9m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/db537434f6ef4178ba7efc9d14b50184 : SUCCESS in 12m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d45823afe5c84a9c933c8d9df120644e : SUCCESS in 17m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a8f14040e164688ad2a811f9d115416 : SUCCESS in 22m 56s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/afaf5584ad0342098b3be740f4f2f795 : SUCCESS in 10m 22s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/6dc0d550fb7d41fa8002beed5b118e86 : SUCCESS in 12m 15s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/b2af31ea4abc4760b4a6ac9d80a57c82 : SUCCESS in 9m 04s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/b2c7b34290b54952a51445362ec780b3 : POST_FAILURE in 17m 26s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/087ed8b7cd6f49b69fa526f4ccf23511 : SUCCESS in 55m 28s (non-voting)\n- ironic-tempest-standalone-advanced https://zuul.opendev.org/t/openstack/build/902ada2214d54efdb63b7b0a069c835f : POST_FAILURE in 19m 27s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/c645d6540c6f4c2ea300bd0a52a280d6 : SUCCESS in 46m 39s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/ee1f43574ef44b17be72e6947de665ab : POST_FAILURE in 18m 03s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/ebf76afc6acf409a84e2087c2ff3085c : FAILURE in 21m 22s (non-voting)\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/7b71a10a79174d908661fa9bca8614ad : SUCCESS in 36m 00s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/cf78e64c5fbd4b319cd48f68dd29ba6d : SUCCESS in 25m 15s\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/dd2f00408d8c4e2f8333d2da1f3788ff : SUCCESS in 48m 51s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/0c87abf9a7e74561b8eed39b2cdf9213 : SUCCESS in 1h 47m 01s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-shard https://zuul.opendev.org/t/openstack/build/da442ec2175f430d9d80d6398c59bbfb : SUCCESS in 1h 25m 57s (non-voting)\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/08a0b6e6b7234da189cf60d2e8430180 : SUCCESS in 53m 54s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/33ed8ac6b647436eb659dc4e7332e553 : SUCCESS in 29m 58s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/eab2defbde0f426ab4d102024fec9931 : POST_FAILURE in 17m 14s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/e4c4c25764c3421fa1e986a6dfd440c7 : SUCCESS in 1h 32m 42s\n- metalsmith-integration-glance-centos9-uefi https://zuul.opendev.org/t/openstack/build/2289cd70accc4f009368cb10249238dd : FAILURE in 28m 05s (non-voting)\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/cee2897d7c9849c68104558c926f0b3e : SUCCESS in 1h 17m 11s (non-voting)\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/e1c45909dd9244c9a55385f3a44e79ee : SUCCESS in 1h 12m 33s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/144bc804afab4e9d96b549e423682250 : SUCCESS in 25m 27s (non-voting)\n- ironic-inspector-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/9e5151f5852d496eb085eddd068f6425 : SUCCESS in 48m 34s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/38f729ea3fdc450f9f124e641d02f4cf : POST_FAILURE in 17m 15s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/8e61f6aa98ec4159a72dd66ea3fb6e57 : SUCCESS in 24m 30s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/1bd8a1c8c0b74a80bdecabb89eb7dfbf : SUCCESS in 1h 00m 38s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/322d0fbf4177442ba82051796d73b96a : SUCCESS in 37m 26s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7dcd7e9d18394bd49b9f980eacc95a55 : SUCCESS in 43m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d73e6a47fe8315ed44cc4941e43a368d2cf6cbf1","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2026-07-23 15:07:56.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":1},{"id":"5ca79668c8ed022f5982e867feeca80635fcd46f","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2026-07-23 15:21:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"50a40cbd378a1d65fd9a465691a27dd94b34c664","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2026-07-23 15:21:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d38dc0cb2b3ee7a1b8f7b1c6de89854484b6b060","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 17:26:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22b355fc5a594050a7f57f33ba1d63cf\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e85062cbc4943c5984d3b868c5ace4d : SUCCESS in 13m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a67022f749a4dfb96e0c9e1973c46f7 : SUCCESS in 6m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f59ee51318214ad4b7edb1d70069e08a : SUCCESS in 9m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c301c57ed6de4ff3823c1e6dadc7738f : SUCCESS in 9m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f491bc4d189c4a53a667d23035a49927 : SUCCESS in 12m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87a8f14bd4af4bfdbc55ddb6da671b03 : SUCCESS in 16m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2b0b02851794c81be6d582b6e8d76f2 : SUCCESS in 21m 51s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/cc7eaeec6b13481e9f27c2a51efbe9b4 : SUCCESS in 6m 40s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/7ed3eec42f6544bc93f6b6365df676aa : SUCCESS in 12m 23s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/d0ca2d7d2a1f4f62a6af2146324f9c72 : SUCCESS in 11m 14s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/5b0b90a22e2a40839ce9ac0e1a5b5295 : SUCCESS in 38m 11s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/1db918cb00464694aa3037745fe62078 : SUCCESS in 49m 00s (non-voting)\n- ironic-tempest-standalone-advanced https://zuul.opendev.org/t/openstack/build/b9829cd6ca79433f898eb1efda5afd67 : SUCCESS in 1h 03m 09s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/d61b88b9664f4114af593e0426b0078d : SUCCESS in 49m 42s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/3669b13984cf464f8b4e132537a65531 : SUCCESS in 59m 33s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/1c173891f825405498dbdaa677e775ec : FAILURE in 32m 38s (non-voting)\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/1bc8e081da4c45509952514fee9c3e05 : SUCCESS in 46m 10s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/a1af673f2d11426f94ef3eb9e5013f86 : SUCCESS in 42m 44s\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/2ecedabd78fb419a8dd2911da1838899 : SUCCESS in 1h 05m 44s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/3d040419eebc4516a2c95d86a6ef8229 : SUCCESS in 1h 58m 16s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-shard https://zuul.opendev.org/t/openstack/build/0511714ad6ac4f5b889b5575fbd6cb3d : FAILURE in 1h 36m 03s (non-voting)\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/b4b3ebaec12748079cbb66f729199a0b : SUCCESS in 1h 15m 14s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/8111df6b1edc4a6fa3196406679bdcd4 : SUCCESS in 39m 47s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/08830e9f66484f3e80c20f19356e8eeb : SUCCESS in 41m 40s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/78b43d95b5764707b439636d82e637fa : SUCCESS in 1h 05m 47s\n- metalsmith-integration-glance-centos9-uefi https://zuul.opendev.org/t/openstack/build/3658cad84e5f4f1ca26cdb54cc7947e7 : FAILURE in 24m 41s (non-voting)\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/bb82652e696b40809c3c40ea1e0fe2ef : SUCCESS in 1h 00m 18s (non-voting)\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/89cea046f0734ff49bd5f20c50648ebe : SUCCESS in 1h 21m 54s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/a45187e3dc8249798a7af32ba845e6c3 : SUCCESS in 49m 14s (non-voting)\n- ironic-inspector-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/7303cf3df5f443f99ae929c0099622f4 : SUCCESS in 45m 30s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/99f6aa120d454e259de0a0fd4797fe72 : FAILURE in 35m 06s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/13d9445e7cf04339871522f1f5b8b9c9 : SUCCESS in 1h 00m 27s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/24b8d1437bf04edaaac1dacf27891089 : SUCCESS in 58m 17s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a52acf70326c40939ff24ee430e0d797 : SUCCESS in 48m 40s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c42edfc7e6b146f3a3849d5efa4f8f41 : SUCCESS in 43m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4c9c08c16623894ee6c95bce82a4294c11100363","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2026-07-28 22:03:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b827943477ca84b641bf9069680a05617b12addc","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2026-07-29 13:50:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1a0720c7aa536a7f149e8c935cb52832829c28f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 13:51:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7005cddd55dfb390f206e2b2974675ae13393e26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 15:20:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/452f58a22b044b5cbbb98f4330fa2b42\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2db067f9031478bb0934e9e2235c7c7 : SUCCESS in 7m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a23ece5b507440d2a008050261fd4285 : SUCCESS in 9m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c940e4538ae74eb6ab8e9a099b738e92 : SUCCESS in 10m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a7589d38c8d460089dc771ea4437a38 : SUCCESS in 8m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/43c4467d76e7436bbf2da0b6315f975e : SUCCESS in 20m 47s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/9d7edca84406480da8a042c7a0b3762b : SUCCESS in 8m 42s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/02741d3f9fb945149c2df5a9f0cc502d : SUCCESS in 11m 16s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/5cad3ae158f541f4b753e08bc34de29d : SUCCESS in 40m 02s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/e23f148733c64379bd3870298f8a71df : SUCCESS in 46m 52s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/ffdaa8d601ff4af49e934f7e83ef104e : SUCCESS in 49m 27s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/3606443dd5024025867415e1d984ebbe : SUCCESS in 48m 17s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/26ca268f2d5047c6abfd5a94a7d0feab : SUCCESS in 36m 55s\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/5540236f324d49f9b9c2f33f59573624 : SUCCESS in 54m 48s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/f069b2982b28481cba7667a103026c53 : SUCCESS in 31m 36s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/72866206478448f0a4c82b8609ff81f6 : SUCCESS in 52m 05s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/d5e4361acabc4f44a7ef6f51d16772c6 : SUCCESS in 52m 02s","accounts_in_message":[],"_revision_number":1},{"id":"0a9c17b2eff2fb8de3dc693cd1dd355a118c3c72","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 15:20:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2535b37f08d83c044842fe0efff1c88b9d1b172b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 15:24:46.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ededdb16696496ea994f8b7c7bccb03\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/455c7d9082334ed6bf9efc28e77e9ee6 : SUCCESS in 1m 19s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/32b25a4c5749424da73cb5ad79eaacda : FAILURE in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cd4dd34b0dab41a0f7669aa72b00438d36ad2c17","revisions":{"cd4dd34b0dab41a0f7669aa72b00438d36ad2c17":{"kind":"REWORK","_number":1,"created":"2026-07-20 16:09:56.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/19/998019/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/19/998019/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/19/998019/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/19/998019/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/19/998019/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/19/998019/1"}}},"commit":{"parents":[{"commit":"4782ffd098fa5b90c3da1987e95091629d3ffe3d","subject":"Use service auth for Neutron port operations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4782ffd098fa5b90c3da1987e95091629d3ffe3d"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2026-07-13 18:14:07.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2026-07-20 16:09:56.000000000","tz":0},"subject":"security: Fix poisoning of the allowed origin list","message":"security: Fix poisoning of the allowed origin list\n\nRecently a bug was identified in nova (bug 2158919) where it was possible\nto poison the origin list because of the way which oslo.config works along\nwith the resulting code pattern in the service. This could lead to memory\nexhaustion or other additional origins being added to the list in memory.\n\nIn any event, this issue was relatively trivial to fix and is being\ntreated as a hardening opportunity.\n\nRelated-Bug: 2160548\nAssisted-By: Claude Opus 4.6\nChange-Id: I9d48a6827455d115e9e1176f3dee41c9e07251b8\nSigned-off-by: Julia Kreger \u003cjuliaashleykreger@gmail.com\u003e\n(cherry picked from commit 762cde0a03e7f2aebb95a4934cc13249689ee2b7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/cd4dd34b0dab41a0f7669aa72b00438d36ad2c17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/cd4dd34b0dab41a0f7669aa72b00438d36ad2c17"}]},"branch":"refs/heads/stable/2025.2","conflicts":{"base":"91e5754d6e7858b49a61455d2f2a8b781b540e4d","ours":"4782ffd098fa5b90c3da1987e95091629d3ffe3d","theirs":"6214f526b0b049fc3c3e5f85b09155ce51f7042f","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
