)]}'
{"id":"openstack%2Fironic~999765","triplet_id":"openstack%2Fironic~bugfix%2F34.0~Ie446481a97c8c30eaee7104e46701aff5670b89a","project":"openstack/ironic","branch":"bugfix/34.0","attention_set":{},"removed_from_attention_set":{"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2026-08-06 16:58:35.000000000","reason":"\u003cGERRIT_ACCOUNT_10342\u003e replied on the change","reason_account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}}},"hashtags":["CVE-2026-71201","ironic-week-prio"],"change_id":"Ie446481a97c8c30eaee7104e46701aff5670b89a","subject":"Portgroup shard filter bypasses scope to project","status":"MERGED","created":"2026-08-04 20:34:34.000000000","updated":"2026-08-06 19:17:24.000000000","submitted":"2026-08-06 19:16:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"999765","meta_rev_id":"47ae80309074ac619ff41a69e5f799ba52550b46","_number":999765,"virtual_id_number":999765,"owner":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-06 19:16:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":2,"date":"2026-08-04 23:14:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":2,"date":"2026-08-06 16:58:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":1,"date":"2026-08-04 23:14:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":1,"date":"2026-08-06 16:58:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-04 22:11:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-04 23:14:53.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"}],"messages":[{"id":"38b3a47b35f96e1350ac1205dce16ec6db6a18a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2026-08-04 20:34:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c5a86bd9e32883d8dfafc53da1154a25c4758190","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2026-08-04 20:44:26.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":1},{"id":"1c5863e268d7c887ebaaefd2d7e4c4fa51b14036","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 22:11:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c97bde10b5a0479ebb49835ba519e2c8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b230548d9a8f4bbc83e7aeb90b6ef654 : SUCCESS in 8m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73f1bb069ff54e5aa6415ece603c2e94 : SUCCESS in 19m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aa6333e90b684b3c99331942a85bbcdf : SUCCESS in 4m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3005c5c088224ef69659a4ee0171c6b4 : SUCCESS in 10m 03s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/6960b425f8f8488ca4b4e618f528dd6a : SUCCESS in 6m 26s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/d085ed03b91b4fe0a4d28a629748f9fa : SUCCESS in 9m 40s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/3b108975ffd541e0a8c64ba79f361f3f : SUCCESS in 42m 09s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/bffdc64a703b4eaa96daea1459df50a8 : SUCCESS in 46m 10s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/44be30dfa89942b88727869d1235ec04 : SUCCESS in 51m 42s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/ea23cbf3256f4ae282e8d6b4fdbba3ca : SUCCESS in 57m 46s\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/78e04e6bdb0b47dda963ebb73e65eb66 : SUCCESS in 1h 28m 22s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/3b0d97880a8e4809adf29eebe5b7f16b : SUCCESS in 31m 49s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/eec7d11d7aac4d018ca6f814f0ce1c3b : SUCCESS in 45m 16s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/862ff60ca55b439bb735e2c0d6f5924a : SUCCESS in 45m 39s","accounts_in_message":[],"_revision_number":1},{"id":"08a733375b1c30469e615f5bf86fe47eae66ebbb","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-08-04 23:14:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bd7deff80923b67fe324f1faefbd8c37c336c9df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 23:15:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3c444820bab36dede95bb28220b47d56f86b93e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 00:25:19.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bfcd86a4e114479cad59c74b09bc0e5f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d204de9fcaf5473690a6fce162f3d35a : SUCCESS in 14m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e518c5251b73497e981ca287ad0781e6 : SUCCESS in 24m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/590e9459fff44ca39bce32fe4ef2608e : SUCCESS in 8m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7aa907284bba40c5918fcba8f94bbce0 : SUCCESS in 7m 38s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/b53a7ad5ad484f4f87b3431921604c0d : SUCCESS in 6m 04s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/d471f22d6d0244bf9907916248611f02 : SUCCESS in 9m 23s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/00e7f37825604db7ac1b603db3d0912b : SUCCESS in 41m 53s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/28698843fa6c46a286321567c5b1f1dc : SUCCESS in 42m 22s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/2c8e501b1985462fb5564575f85c87a9 : SUCCESS in 51m 17s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/25ec4a37f9494a3991cffc874160eb3f : SUCCESS in 47m 16s\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/b908aba48a494664903754397300f5cb : SUCCESS in 55m 38s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/08ffb9795e2449eabd2ceb81cfa7f6d8 : SUCCESS in 40m 33s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/b6ca51b2880a4d62ae5ec50089f1f7ce : SUCCESS in 52m 00s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/39e93890a8f94a1e8abf7a808287e1b7 : FAILURE in 32m 46s","accounts_in_message":[],"_revision_number":1},{"id":"b7c85055633d1e1bf915f1c6979edb9ed3495c8b","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-05 15:47:22.000000000","message":"Hashtag added: CVE-2026-71201","accounts_in_message":[],"_revision_number":1},{"id":"4c3042d32cc46a1a52a41d5f8b75149943b91901","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2026-08-06 16:58:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"91aa899bd461477e8c06cced46e83a7623fa90e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 18:00:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edb70001ea01477aac50e3557381118f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7ed786322fd4712956bde0b8d39fdf3 : SUCCESS in 13m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/369fb99bd36a46df8b49072114cc78f2 : SUCCESS in 8m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b592cc8cf4584b6897e8b25cf3ba7a65 : SUCCESS in 8m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/15e09191f6464b1da925f5a0a3563f9d : SUCCESS in 7m 54s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/524103b19f1747578e217c42a994204d : SUCCESS in 5m 45s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/3440236485984b689704244c9d61fde2 : SUCCESS in 8m 28s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/c9d9ec8602274fbb9df2cbf86faba8c4 : SUCCESS in 40m 53s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/e46138395c2d406f92ae3983a5962946 : SUCCESS in 35m 43s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/54a649ff30ce4782b3f51c3a472fcf17 : SUCCESS in 38m 01s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/db77f55614444b7fb3b86bfeaf0f0892 : SUCCESS in 37m 53s\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/aa55580ead754135bba39767a2940773 : SUCCESS in 57m 43s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/c217f4615c98457c90977fb01320d641 : SUCCESS in 19m 18s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/67e74cf4548c4493b1b8609f2a3f11af : SUCCESS in 52m 33s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/78e970f9362a416599a59a64a478f66a : SUCCESS in 28m 38s","accounts_in_message":[],"_revision_number":1},{"id":"9a16e6973dae022e3190c3ff573d4f0cce536081","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 18:00:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"751ac1abe2ecfc6185b86e0a35a137ec1d94efa6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 19:16:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/366ff4c650274faea17d28beee717585\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db38d41d30a04bc583591ce654a3c83e : SUCCESS in 14m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a6bc7e5e76d43a89630bf70e2b4d317 : SUCCESS in 20m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2bf5bce52ce54d22b0325e2a76095c63 : SUCCESS in 9m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b5e17de17e2745e790b966a4be9693b6 : SUCCESS in 6m 28s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/31210b512a4c48ed82c9f5e419ea70b7 : SUCCESS in 5m 14s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/ab916cdc6e754196b15b565eb3e90102 : SUCCESS in 8m 27s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/13da8b12d3234e29af4083b73cc4fa73 : SUCCESS in 24m 57s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/d1a9c748e3ec41cbb5ba6894fae67004 : SUCCESS in 47m 31s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/b40fd1da3616444ca53a6136d82be02e : SUCCESS in 53m 02s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/c5da38a3c1db4badac35d752bf2b79b1 : SUCCESS in 33m 37s\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/70e6ea154d8c49d99fd2ae564ac6f077 : SUCCESS in 31m 01s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/8c0498b00d334d25a980a790d89a9727 : SUCCESS in 20m 47s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/b0e5f8de701346c8b48bedf24b042f5c : SUCCESS in 30m 54s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/313ddf065ef1473d9c4623d2a0b8d648 : SUCCESS in 1h 13m 27s","accounts_in_message":[],"_revision_number":1},{"id":"25da01d823796a89ccf55f6e4d479f5866fab40c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 19:16:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"47ae80309074ac619ff41a69e5f799ba52550b46","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 19:17:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/336681bbf88847f4a70ecdf9d946db1c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f742007cd41749f0abb8d459f23e987c : SUCCESS in 1m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53581c111ea64d509e51922914203bbc : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5f56bddec9ee81dde50a453926005ca8b8cd0b9c","revisions":{"5f56bddec9ee81dde50a453926005ca8b8cd0b9c":{"kind":"REWORK","_number":1,"created":"2026-08-04 20:34:34.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/65/999765/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/65/999765/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/65/999765/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/65/999765/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/65/999765/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/65/999765/1"}}},"commit":{"parents":[{"commit":"ae176160753955741c8b0722da84a3c6e041c4c1","subject":"security: Fix poisoning of the allowed origin list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/ae176160753955741c8b0722da84a3c6e041c4c1"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2026-08-03 20:02:38.000000000","tz":-420},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2026-08-04 20:34:29.000000000","tz":-420},"subject":"Portgroup shard filter bypasses scope to project","message":"Portgroup shard filter bypasses scope to project\n\nPrior to this fix, a project reader could query for portgroups by shard\nand recieve all portgroups belonging to that shard, not just the ones\nthey are authorized to see.\n\nCloses-bug: #2162715\nChange-Id: Ie446481a97c8c30eaee7104e46701aff5670b89a\nSigned-off-by: Jay Faulkner \u003cjay@jvf.cc\u003e\n(cherry picked from commit 00625d5dd89195d6658899f6e7454741ca166f73)\n(cherry picked from commit 0ca2b0f6580a4a6ef1ef0b5e7b81c316043402f5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/5f56bddec9ee81dde50a453926005ca8b8cd0b9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/5f56bddec9ee81dde50a453926005ca8b8cd0b9c"}]},"branch":"refs/heads/bugfix/34.0"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
