)]}'
{"id":"openstack%2Fkarbor-dashboard~654937","triplet_id":"openstack%2Fkarbor-dashboard~master~I98af21d8c2b70592f9b131a8e2ea193fb80ca62b","project":"openstack/karbor-dashboard","branch":"master","hashtags":[],"change_id":"I98af21d8c2b70592f9b131a8e2ea193fb80ca62b","subject":"Replace git.openstack.org URLs with opendev.org URLs","status":"MERGED","created":"2019-04-23 10:22:53.000000000","updated":"2019-09-12 07:33:29.000000000","submitted":"2019-09-12 07:33:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654937-1568273609166-4cbec2ef","meta_rev_id":"a242c88f8f03ba5d33803bf4270b017dbfb517c1","_number":654937,"virtual_id_number":654937,"owner":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":2,"date":"2019-09-12 07:33:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":2,"date":"2019-09-12 07:23:05.000000000","_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":1,"date":"2019-09-12 07:23:05.000000000","_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-13 08:23:35.000000000","updated_by":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"reviewer":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"state":"REVIEWER"},{"updated":"2019-06-14 13:58:35.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2019-06-27 12:59:38.000000000","updated_by":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"reviewer":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"state":"REVIEWER"},{"updated":"2019-08-11 14:14:05.000000000","updated_by":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"reviewer":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"state":"REVIEWER"},{"updated":"2019-08-12 22:24:58.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-08-13 00:49:04.000000000","updated_by":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"reviewer":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"state":"REVIEWER"},{"updated":"2019-08-14 15:23:56.000000000","updated_by":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"reviewer":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"state":"REVIEWER"},{"updated":"2019-09-12 07:23:05.000000000","updated_by":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"reviewer":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"state":"REVIEWER"},{"updated":"2019-09-12 07:33:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"da056cf44e62ad68fe26f957150eb7466e65c50a","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-04-23 10:22:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8775ee22a53be198606af72d4d11baac7a25d51e","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-04-23 10:29:10.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"144605f576113f8f0051cc4027a7a1550e990995","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-04-23 10:29:46.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"08a59a7337109db441ef4bd3877d5efe78043247","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 10:41:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/37/654937/3/check/test-release-openstack/c4fc669/ : SUCCESS in 3m 30s\n- openstack-tox-pep8 http://logs.openstack.org/37/654937/3/check/openstack-tox-pep8/4176ef3/ : SUCCESS in 4m 45s\n- openstack-tox-py27 http://logs.openstack.org/37/654937/3/check/openstack-tox-py27/49c6ab2/ : SUCCESS in 4m 41s\n- openstack-tox-py35 http://logs.openstack.org/37/654937/3/check/openstack-tox-py35/5bb0792/ : SUCCESS in 5m 01s\n- openstack-tox-py36 http://logs.openstack.org/37/654937/3/check/openstack-tox-py36/6164fb9/ : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":3},{"id":"b29c8987528526dccea2fa425f7431e80516ca25","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-06-04 08:54:17.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"7e08bcd0d9c40835589ee37ec1b7f445fd8662c5","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-06-04 10:08:48.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"8ac35eb2df49949ec7673080cd01f40386f127cb","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2019-06-13 08:23:35.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"387db4ac52249fb41ae8bdbb16c1edf51a1e5755","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-06-14 05:18:40.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\nthanks for the patch. please one comment inline.","accounts_in_message":[],"_revision_number":3},{"id":"309faafd57d760fec79f9233b513c4381a094ebc","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-06-14 08:07:57.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"3968b723ecd75ee8cb850e81c0834c8e210b6204","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 08:20:44.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/37/654937/4/check/test-release-openstack/e1d3e9a/ : SUCCESS in 4m 16s\n- openstack-tox-pep8 http://logs.openstack.org/37/654937/4/check/openstack-tox-pep8/0e188ec/ : SUCCESS in 6m 32s\n- openstack-tox-py27 http://logs.openstack.org/37/654937/4/check/openstack-tox-py27/0e00c6e/ : SUCCESS in 5m 33s\n- openstack-tox-py35 http://logs.openstack.org/37/654937/4/check/openstack-tox-py35/488ba67/ : SUCCESS in 5m 22s\n- openstack-tox-py36 http://logs.openstack.org/37/654937/4/check/openstack-tox-py36/8076c3a/ : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":4},{"id":"c27bc98f63e99df441616cae6cc8b2ebece3867e","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-06-14 13:58:35.000000000","message":"Patch Set 4: Code-Review+1\n\nlgtm","accounts_in_message":[],"_revision_number":4},{"id":"c99bb4a77ae28fdebe9eff4206f1f0d0cad03cd5","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-06-27 12:59:38.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"28e373b44e631c1f5b53fe3dfd8f1454248d9007","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2019-07-03 09:38:11.000000000","message":"Patch Set 4: Code-Review-1\n\nduplicate patch","accounts_in_message":[],"_revision_number":4},{"id":"618e0e96933dfee5337fea56a4fff897238017b5","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2019-07-03 09:39:07.000000000","message":"Patch Set 4:\n\nhttps://review.opendev.org/#/c/668818/1","accounts_in_message":[],"_revision_number":4},{"id":"b9021f0b46c932204adaf671f86aa56cd2c8a86c","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2019-07-03 09:39:51.000000000","message":"Patch Set 4:\n\nhttps://review.opendev.org/#/c/568759/","accounts_in_message":[],"_revision_number":4},{"id":"baa5f534618a5761f730cc48096339791d1837d7","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-07-03 15:34:15.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"83cc8560f7cf1219e6bacd5465a6e38fbe0608ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 15:46:46.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/37/654937/5/check/test-release-openstack/6a7caaa/ : SUCCESS in 2m 01s\n- openstack-tox-pep8 http://logs.openstack.org/37/654937/5/check/openstack-tox-pep8/77e8938/ : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":5},{"id":"9eeee61672f091fee5a0fa13b9a9a831594e66d4","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-07-29 09:59:35.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"325ae3f750bda0eb463fd835e4eab6a0a39d73f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 10:08:50.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/37/654937/6/check/test-release-openstack/ec42201/ : SUCCESS in 3m 26s\n- openstack-tox-pep8 http://logs.openstack.org/37/654937/6/check/openstack-tox-pep8/119f674/ : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":6},{"id":"c0489639ff5dd9529950fe3f35e18aa1c3a91921","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-08-11 14:14:05.000000000","message":"Patch Set 6: Code-Review-1\n\ndo in another ps, please abondon this ps","accounts_in_message":[],"_revision_number":6},{"id":"e78402f1c978c437ee7d63c2cba1befd103502dd","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-08-12 22:24:58.000000000","message":"Patch Set 6: Code-Review-1\n\nkindly abandon it. Done in #568759","accounts_in_message":[],"_revision_number":6},{"id":"7f841888dc18077b3eca638295ce33b96b727390","author":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"date":"2019-08-13 00:49:04.000000000","message":"Patch Set 6: Code-Review-1","accounts_in_message":[],"_revision_number":6},{"id":"7f8887b0da9750ca553a6b5fd300233a6f82b6f6","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-08-14 15:23:56.000000000","message":"Patch Set 6: Code-Review-1","accounts_in_message":[],"_revision_number":6},{"id":"bc18f779c8defdc5b92498ff166f074fa77680aa","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-08-16 04:48:36.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"a93f37d6cab2dcb982304c2208805d14aa439fae","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-08-16 04:49:58.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"91c5cdfc08e59415839f58b8e7397ad0bf5ee718","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 05:00:02.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/12fb5cd647c845889b3136cfa17cc33c : SUCCESS in 2m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a5dbe5e70774b638e64c8cd46dba42e : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":8},{"id":"c9416c7ea9465a4c4865a2349d4b270e3bdd21eb","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2019-09-12 07:07:04.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"3da84105827c55bdea156a42e3a10197da8e07b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 07:16:46.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5a0b1520c4bd48918af9c248a6fc27cb : SUCCESS in 1m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc7e35bfdcf54cb79fa8a43a5e30354d : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":9},{"id":"e46517dcf195f7caa1adbb1c30d76045fc9f06c8","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2019-09-12 07:23:05.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"b6bb2787b8d61c2d0fe7880a3a6af052698a01e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 07:23:18.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"4fb97debfce1f08d680e84781fbcb8e28ab34dcf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 07:33:28.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7b68301dff284ea0987cf1b46c8b3b0d : SUCCESS in 2m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d41ca9f0b3e4d6bb2a0b043bc2c7d53 : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":9},{"id":"a242c88f8f03ba5d33803bf4270b017dbfb517c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 07:33:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"374898665173c73dc97bea05ce1a7e83f09de094","revisions":{"ece090961654ce873414782175f6cc55159f23eb":{"kind":"REWORK","_number":1,"created":"2019-04-23 10:22:53.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/37/654937/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor-dashboard","ref":"refs/changes/37/654937/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/37/654937/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/37/654937/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/37/654937/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor-dashboard refs/changes/37/654937/1"}}},"commit":{"parents":[{"commit":"056e51066eac787fbdd3c552271264a9704cc352","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/056e51066eac787fbdd3c552271264a9704cc352"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 10:22:28.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 10:22:48.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I98af21d8c2b70592f9b131a8e2ea193fb80ca62b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/ece090961654ce873414782175f6cc55159f23eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/ece090961654ce873414782175f6cc55159f23eb"}]},"branch":"refs/heads/master"},"f41d982e8488ef414a6417bb0a0a9362f6b50841":{"kind":"REWORK","_number":2,"created":"2019-04-23 10:29:10.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/37/654937/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor-dashboard","ref":"refs/changes/37/654937/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/37/654937/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/37/654937/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/37/654937/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor-dashboard refs/changes/37/654937/2"}}},"commit":{"parents":[{"commit":"056e51066eac787fbdd3c552271264a9704cc352","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/056e51066eac787fbdd3c552271264a9704cc352"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 10:22:28.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 10:29:07.000000000","tz":0},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I98af21d8c2b70592f9b131a8e2ea193fb80ca62b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/f41d982e8488ef414a6417bb0a0a9362f6b50841"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/f41d982e8488ef414a6417bb0a0a9362f6b50841"}]},"branch":"refs/heads/master"},"0eb21f71416d43299b51108cbddfecf3e52af20b":{"kind":"REWORK","_number":3,"created":"2019-04-23 10:29:46.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/37/654937/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor-dashboard","ref":"refs/changes/37/654937/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/37/654937/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/37/654937/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/37/654937/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor-dashboard refs/changes/37/654937/3"}}},"commit":{"parents":[{"commit":"056e51066eac787fbdd3c552271264a9704cc352","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/056e51066eac787fbdd3c552271264a9704cc352"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 10:22:28.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 10:29:42.000000000","tz":0},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I98af21d8c2b70592f9b131a8e2ea193fb80ca62b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/0eb21f71416d43299b51108cbddfecf3e52af20b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/0eb21f71416d43299b51108cbddfecf3e52af20b"}]},"branch":"refs/heads/master"},"81cc98d6b60d76292c6f0b2dd8128d134e271528":{"kind":"REWORK","_number":4,"created":"2019-06-14 08:07:57.000000000","uploader":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"ref":"refs/changes/37/654937/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor-dashboard","ref":"refs/changes/37/654937/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/37/654937/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/37/654937/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/37/654937/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor-dashboard refs/changes/37/654937/4"}}},"commit":{"parents":[{"commit":"056e51066eac787fbdd3c552271264a9704cc352","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/056e51066eac787fbdd3c552271264a9704cc352"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 10:22:28.000000000","tz":480},"committer":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-06-14 08:07:46.000000000","tz":0},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I98af21d8c2b70592f9b131a8e2ea193fb80ca62b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/81cc98d6b60d76292c6f0b2dd8128d134e271528"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/81cc98d6b60d76292c6f0b2dd8128d134e271528"}]},"branch":"refs/heads/master"},"d77ff2693a6c0527669fab32ec1716a6c827db89":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-07-03 15:34:15.000000000","uploader":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"ref":"refs/changes/37/654937/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor-dashboard","ref":"refs/changes/37/654937/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/37/654937/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/37/654937/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/37/654937/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor-dashboard refs/changes/37/654937/5"}}},"commit":{"parents":[{"commit":"e06dbf28252ba1fa27f032381de81500a70f2c8e","subject":"Switch to the new canonical constraints URL on master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/e06dbf28252ba1fa27f032381de81500a70f2c8e"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 10:22:28.000000000","tz":480},"committer":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2019-07-03 15:34:15.000000000","tz":0},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I98af21d8c2b70592f9b131a8e2ea193fb80ca62b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/d77ff2693a6c0527669fab32ec1716a6c827db89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/d77ff2693a6c0527669fab32ec1716a6c827db89"}]},"branch":"refs/heads/master"},"1735f45d4ccd5d55c9aefd4f8381c8b7bf243782":{"kind":"REWORK","_number":6,"created":"2019-07-29 09:59:35.000000000","uploader":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"ref":"refs/changes/37/654937/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor-dashboard","ref":"refs/changes/37/654937/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/37/654937/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/37/654937/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/37/654937/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor-dashboard refs/changes/37/654937/6"}}},"commit":{"parents":[{"commit":"492072f15692093246bee72e14e9bf8d24e6a9ad","subject":"Update links in README","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/492072f15692093246bee72e14e9bf8d24e6a9ad"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 10:22:28.000000000","tz":480},"committer":{"name":"gujin","email":"gu.jin@99cloud.net","date":"2019-07-29 09:59:28.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I98af21d8c2b70592f9b131a8e2ea193fb80ca62b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/1735f45d4ccd5d55c9aefd4f8381c8b7bf243782"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/1735f45d4ccd5d55c9aefd4f8381c8b7bf243782"}]},"branch":"refs/heads/master"},"6c052d9280f30d3777e32959050562104e4e10a9":{"kind":"REWORK","_number":7,"created":"2019-08-16 04:48:36.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/37/654937/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor-dashboard","ref":"refs/changes/37/654937/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/37/654937/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/37/654937/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/37/654937/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor-dashboard refs/changes/37/654937/7"}}},"commit":{"parents":[{"commit":"4a7fbd3ad81e88bd6ba011a0af97b3b13c3a06e3","subject":"Sync Sphinx requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/4a7fbd3ad81e88bd6ba011a0af97b3b13c3a06e3"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 10:22:28.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-08-16 04:48:17.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I98af21d8c2b70592f9b131a8e2ea193fb80ca62b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/6c052d9280f30d3777e32959050562104e4e10a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/6c052d9280f30d3777e32959050562104e4e10a9"}]},"branch":"refs/heads/master"},"541a6d693f6afc67a2ffa8bd52eaa48271394e87":{"kind":"REWORK","_number":8,"created":"2019-08-16 04:49:58.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/37/654937/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor-dashboard","ref":"refs/changes/37/654937/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/37/654937/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/37/654937/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/37/654937/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor-dashboard refs/changes/37/654937/8"}}},"commit":{"parents":[{"commit":"4a7fbd3ad81e88bd6ba011a0af97b3b13c3a06e3","subject":"Sync Sphinx requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/4a7fbd3ad81e88bd6ba011a0af97b3b13c3a06e3"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 10:22:28.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-08-16 04:49:52.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I98af21d8c2b70592f9b131a8e2ea193fb80ca62b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/541a6d693f6afc67a2ffa8bd52eaa48271394e87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/541a6d693f6afc67a2ffa8bd52eaa48271394e87"}]},"branch":"refs/heads/master"},"374898665173c73dc97bea05ce1a7e83f09de094":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2019-09-12 07:07:04.000000000","uploader":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"ref":"refs/changes/37/654937/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor-dashboard","ref":"refs/changes/37/654937/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/37/654937/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/37/654937/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/37/654937/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor-dashboard refs/changes/37/654937/9"}}},"commit":{"parents":[{"commit":"0609a0e996891aef575e25730f1690d95c1216c0","subject":"Merge \"Add Python 3 Train unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/0609a0e996891aef575e25730f1690d95c1216c0"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 10:22:28.000000000","tz":480},"committer":{"name":"Jiao Pengju","email":"jiaopengju@cmss.chinamobile.com","date":"2019-09-12 07:07:04.000000000","tz":0},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I98af21d8c2b70592f9b131a8e2ea193fb80ca62b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/374898665173c73dc97bea05ce1a7e83f09de094"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/374898665173c73dc97bea05ce1a7e83f09de094"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
