)]}'
{"id":"openstack%2Fkarbor-dashboard~701330","triplet_id":"openstack%2Fkarbor-dashboard~master~Ifc49a69d4b34b32d64a95cca488c3a1ab7eb2ef6","project":"openstack/karbor-dashboard","branch":"master","topic":"drop-django111-support","hashtags":[],"change_id":"Ifc49a69d4b34b32d64a95cca488c3a1ab7eb2ef6","subject":"Drop Django 1.11 support","status":"MERGED","created":"2020-01-07 07:03:38.000000000","updated":"2020-02-05 13:28:11.000000000","submitted":"2020-02-05 13:28:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701330-1580909291005-5f1f21a0","meta_rev_id":"ca354a8e134a83fe65da91ccc0b6189d137204a5","_number":701330,"virtual_id_number":701330,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-05 13:28:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":2,"date":"2020-02-05 13:14:29.000000000","_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"value":1,"date":"2020-01-21 08:00:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":1,"date":"2020-02-05 13:14:29.000000000","_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-21 08:00:45.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-02-05 13:14:29.000000000","updated_by":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"reviewer":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"state":"REVIEWER"},{"updated":"2020-02-05 13:28:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5b6705f873e0bacfcf8a258fce0c4aebf96c3a76","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-01-07 07:03:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d0443be0fae9523e10527869ad76dba235e7945e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 07:24:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- horizon-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/c0967c3baa7c44099988a8a89d188010 : RETRY_LIMIT in 2m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bf8815f5bd2428fb7d27ab8d14d36ed : FAILURE in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ff23190ade64a3aae513585d23219eb : FAILURE in 4m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d5d4db40c2bd442489059170363e8ed8 : FAILURE in 4m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/39bfff3787ab41968c6fba73795efdca : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":1},{"id":"d992d8648b59e56f96e8b71cd334c237d29c0541","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-01-07 07:50:54.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"4f6c4e437003c6c09735f26036262304ed348eb1","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-01-07 07:51:26.000000000","message":"Topic changed from drop-py27-support to drop-django111-support","accounts_in_message":[],"_revision_number":2},{"id":"2082859b2d7fd45da6bb894c4ad22fc92a167bc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 08:05:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- horizon-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/82537749c32a4d1b999614b9d4d10d97 : RETRY_LIMIT in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c4641bef6d744e59f7eb2aec55355ae : FAILURE in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a366d38442340a9a58cd1db2387cdba : FAILURE in 2m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e0090788b5e64a878dc101e0c21b3028 : FAILURE in 4m 44s\n- requirements-check https://zuul.opendev.org/t/openstack/build/01118f95067a4c15bac55fe89832a0e7 : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":2},{"id":"e2f192a14f048b8e265e7fff9c5cb1456654f6b3","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-01-07 08:12:51.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"4f6785e90137c022520ef92a7a91e2239a62a9f5","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-01-07 08:25:29.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9432b36115ec0eed894cf02eba65f4a05b7b8e53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 08:34:30.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- horizon-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/8354f904e57749c69e6d263dc128a78e : RETRY_LIMIT in 2m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f31b6c81f84d4788a1bbb819af191066 : FAILURE in 2m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8168d4fd90ad41769d97f21327fe89fb : FAILURE in 2m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/09340ae247b044948f8e1f71674247b7 : FAILURE in 2m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/626ce2a447e84d399d4d0c2b241f6173 : SUCCESS in 2m 18s","accounts_in_message":[],"_revision_number":4},{"id":"62d9358959e14c31c0ea3018388e1fceb63446ef","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-01-08 00:06:48.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"726d0ae4ebd7c08dd8ea7863be33dfdc14d15bf7","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-01-08 00:07:18.000000000","message":"Topic changed from drop-py27-support to drop-django111-support","accounts_in_message":[],"_revision_number":5},{"id":"8b3c1a9f57e9e5b8ba88ecc2feea029355c6cf2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 00:32:11.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- horizon-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/02b9eefd160c40eeb56cb090fb3ebd91 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44408c16dc72418a991c67a194186027 : FAILURE in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5745d61696547d1ab68b7ad00f3688c : SUCCESS in 3m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bd2b887cd15a404b9064b763bdcc24a3 : SUCCESS in 4m 20s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d43c6419448047e3bf719b1c83d705dd : FAILURE in 3m 01s","accounts_in_message":[],"_revision_number":5},{"id":"93abf6ef1720eb083acfa59857ec5193e43fd725","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-01-08 07:47:22.000000000","message":"Patch Set 5:\n\nGate is fixed in [1].\n[1] https://review.opendev.org/#/c/700627/","accounts_in_message":[],"_revision_number":5},{"id":"93fb97d418247074bdd6c311c12cb0969c3fbf8d","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-01-21 06:33:37.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ba3eade1996da15dc3b3fa790141563ae586e0e8","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-21 06:52:43.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4c4765fab0e4a5edc5a4ca48afc039cf43ace105","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-01-21 06:54:58.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"f9f7bf9a8b69c824c3db0e9f782d703264150c22","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-21 06:58:52.000000000","message":"Patch Set 7:\n\n(1 comment)\n\nNeeds to be rebased on top of the latest patch set of https://review.opendev.org/#/c/694442/","accounts_in_message":[],"_revision_number":7},{"id":"5ab00a221c7d8ae4a48ba92edd4c36831e762a65","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-21 06:59:40.000000000","message":"Patch Set 7: Code-Review-1","accounts_in_message":[],"_revision_number":7},{"id":"a107d76e05fd515ca500673b814fce9736669c12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 07:04:30.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/99796b1d992c42acb874b44c87201be8 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a149c22ae6844a19ba904a9b887976e : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dba45d1727c44e0fac717cd2c1d012f4 : SUCCESS in 3m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c4f36aacd50042cb83f0d631b5f9ff6c : SUCCESS in 4m 06s\n- requirements-check https://zuul.opendev.org/t/openstack/build/01e7a16bea024ffb8c5d90e63f7ef30e : SUCCESS in 2m 20s","accounts_in_message":[],"_revision_number":7},{"id":"998b61643fa3aacbac9e1564ee067b9a1b9013a4","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-01-21 07:08:17.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"cbb107306d12cccaa6cec9e56ad8c66fc7036cab","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-21 07:15:06.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"899053cdcab9c91f7551fd723d495be4065c72e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 07:15:51.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/e2f1e76345124fc89de821d5971a25ec : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b28404e8e054513b0d3fc4691757864 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/249379ec7ab8491d84da5becd9364229 : SUCCESS in 5m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a7e53609463e4860b2efef29e21d399d : SUCCESS in 5m 16s\n- requirements-check https://zuul.opendev.org/t/openstack/build/57999fe2d02044559a083c314009bd0e : SUCCESS in 2m 38s","accounts_in_message":[],"_revision_number":8},{"id":"6b3e6dcb2f71e01558802d1f7daf656ec92b628d","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-01-21 07:20:06.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"2a9e9c683ed4f414487bd63ab5f46af330111265","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-21 07:32:52.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"40519938e33d315cdc9cbe25622d7039d1f20290","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-21 07:33:39.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"7a5c081528d2a5f1078b05160081650ee8d21fee","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-21 07:38:16.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"14dd48d9bffe88b2fbceafbc7a4814f32a39c4ac","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-01-21 07:38:40.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"03b4eb62655fbf9071485cbbddd27a4e273d6e1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 07:42:02.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/038f231ef3f3486f9e494c17e58bb7d6 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91a26c8857604ee3ab811f1758be9fb9 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc2498dc1bfa42bd86c3183d38ecfc9d : SUCCESS in 5m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4469e8a1250a4e5db4b928d828217fd2 : SUCCESS in 4m 09s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2144d615c73d4518b4d9a7bf51452325 : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":9},{"id":"6a08446f2356b41f73fedfdf6eb67b9fcf024d79","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-01-21 07:45:25.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"8f22ae4eaf28e9b2d4377af641f460719eb4a58a","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-01-21 07:46:26.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"9c790cbf651fe29fede487221090cd3c674270ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 07:53:08.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/124dc92e49c2413bb475124982e5be26 : SUCCESS in 7m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/faa032708d1d49f0b92575d82ecac402 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b373ee1ae3fc43f8b52fdd68c4162d4a : SUCCESS in 4m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/db92e8110887478ba55bda68e088d53a : SUCCESS in 4m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0bafaf74129b4dbeb13fc5f3074cdfd4 : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":10},{"id":"4c6c7c7cfda2ba2dd5c4270078d67c898b9a9e43","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-21 08:00:45.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"19c025ea4c569e08baa0d150a55558c9093fa922","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2020-02-05 13:14:29.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"a9a64d5bfb0d7a2673a92afc66c3b18a1b552820","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 13:14:42.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"8d72d1b84f3c3a7733a38428fba374e78603399d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 13:28:10.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aae06085957e457b82c01eb3fd7bb3e5 : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56fe39106b544d888c694be48a0ec1f9 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/287de922b7c64d41806a34f06be74a9a : SUCCESS in 3m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/619f20d43ec848b08eae3afd2c5b7a02 : SUCCESS in 2m 25s","accounts_in_message":[],"_revision_number":10},{"id":"ca354a8e134a83fe65da91ccc0b6189d137204a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 13:28:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"af0f54f7e490bc88505754d30be90eeb2aa5792e","revisions":{"dd05f93f9f17e15c8e6c2a099385333482ccb076":{"kind":"REWORK","_number":1,"created":"2020-01-07 07:03:38.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/30/701330/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor-dashboard","ref":"refs/changes/30/701330/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/30/701330/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/30/701330/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/30/701330/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor-dashboard refs/changes/30/701330/1"}}},"commit":{"parents":[{"commit":"95fb9d2a4938453ef303acaf9f19a424d55be51b","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/95fb9d2a4938453ef303acaf9f19a424d55be51b"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-01-07 07:00:42.000000000","tz":0},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-01-07 07:03:15.000000000","tz":0},"subject":"Drop Django 1.11 support","message":"Drop Django 1.11 support\n\nDjango 1.11 ends its extended support in April 2020 (which is before\nUssuri release), so horizon drops Django 1.11 support in Ussuri.\n\ntox envs for non-primary Django versions are no longer needed in tox.ini\nas testing environments for non-primary Django versions are setup in\nthe zuul jobs now.\n\nhorizon\u003e\u003d17.1.0 is required to use Django 2.2.\nrequirements.txt is updated accordingly.\nFor more info. please refer [1], [2].\n\n[1] https://review.opendev.org/#/c/700733/\n[2] https://etherpad.openstack.org/p/Enable_Django22_support_in_Horizon_Plugin\n\nChange-Id: Ifc49a69d4b34b32d64a95cca488c3a1ab7eb2ef6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/dd05f93f9f17e15c8e6c2a099385333482ccb076"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/dd05f93f9f17e15c8e6c2a099385333482ccb076"}]},"branch":"refs/heads/master"},"5876edfe1213feb64c0670bcf40982b100448602":{"kind":"REWORK","_number":2,"created":"2020-01-07 07:50:54.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/30/701330/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor-dashboard","ref":"refs/changes/30/701330/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/30/701330/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/30/701330/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/30/701330/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor-dashboard refs/changes/30/701330/2"}}},"commit":{"parents":[{"commit":"95fb9d2a4938453ef303acaf9f19a424d55be51b","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/95fb9d2a4938453ef303acaf9f19a424d55be51b"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-01-07 07:00:42.000000000","tz":0},"committer":{"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","date":"2020-01-07 07:50:51.000000000","tz":0},"subject":"Drop Django 1.11 support","message":"Drop Django 1.11 support\n\nDjango 1.11 ends its extended support in April 2020 (which is before\nUssuri release), so horizon drops Django 1.11 support in Ussuri.\n\ntox envs for non-primary Django versions are no longer needed in tox.ini\nas testing environments for non-primary Django versions are setup in\nthe zuul jobs now.\n\nhorizon\u003e\u003d17.1.0 is required to use Django 2.2.\nrequirements.txt is updated accordingly.\nFor more info. please refer [1], [2].\n\n[1] https://review.opendev.org/#/c/700733/\n[2] https://etherpad.openstack.org/p/Enable_Django22_support_in_Horizon_Plugin\n\nChange-Id: Ifc49a69d4b34b32d64a95cca488c3a1ab7eb2ef6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/5876edfe1213feb64c0670bcf40982b100448602"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/5876edfe1213feb64c0670bcf40982b100448602"}]},"branch":"refs/heads/master"},"d9adf89dd4a3a390571d3aa96a4d57d8ad0cb469":{"kind":"REWORK","_number":3,"created":"2020-01-07 08:12:51.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/30/701330/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor-dashboard","ref":"refs/changes/30/701330/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/30/701330/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/30/701330/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/30/701330/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor-dashboard refs/changes/30/701330/3"}}},"commit":{"parents":[{"commit":"95fb9d2a4938453ef303acaf9f19a424d55be51b","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/95fb9d2a4938453ef303acaf9f19a424d55be51b"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-01-07 07:00:42.000000000","tz":0},"committer":{"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","date":"2020-01-07 08:12:47.000000000","tz":0},"subject":"Drop Django 1.11 support","message":"Drop Django 1.11 support\n\nDjango 1.11 ends its extended support in April 2020 (which is before\nUssuri release), so horizon drops Django 1.11 support in Ussuri.\n\ntox envs for non-primary Django versions are no longer needed in tox.ini\nas testing environments for non-primary Django versions are setup in\nthe zuul jobs now.\n\nhorizon\u003e\u003d17.1.0 is required to use Django 2.2.\nrequirements.txt is updated accordingly.\nFor more info. please refer [1], [2].\n\n[1] https://review.opendev.org/#/c/700733/\n[2] https://etherpad.openstack.org/p/Enable_Django22_support_in_Horizon_Plugin\n\nChange-Id: Ifc49a69d4b34b32d64a95cca488c3a1ab7eb2ef6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/d9adf89dd4a3a390571d3aa96a4d57d8ad0cb469"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/d9adf89dd4a3a390571d3aa96a4d57d8ad0cb469"}]},"branch":"refs/heads/master"},"1e972186a38ff264ae692b72fc9f84c1e8dd9970":{"kind":"REWORK","_number":4,"created":"2020-01-07 08:25:29.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/30/701330/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor-dashboard","ref":"refs/changes/30/701330/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/30/701330/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/30/701330/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/30/701330/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor-dashboard refs/changes/30/701330/4"}}},"commit":{"parents":[{"commit":"95fb9d2a4938453ef303acaf9f19a424d55be51b","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/95fb9d2a4938453ef303acaf9f19a424d55be51b"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-01-07 07:00:42.000000000","tz":0},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-01-07 08:25:08.000000000","tz":0},"subject":"Drop Django 1.11 support","message":"Drop Django 1.11 support\n\nDjango 1.11 ends its extended support in April 2020 (which is before\nUssuri release), so horizon drops Django 1.11 support in Ussuri.\n\ntox envs for non-primary Django versions are no longer needed in tox.ini\nas testing environments for non-primary Django versions are setup in\nthe zuul jobs now.\n\nhorizon\u003e\u003d17.1.0 is required to use Django 2.2.\nrequirements.txt is updated accordingly.\nFor more info. please refer [1], [2].\n\n[1] https://review.opendev.org/#/c/700733/\n[2] https://etherpad.openstack.org/p/Enable_Django22_support_in_Horizon_Plugin\n\nChange-Id: Ifc49a69d4b34b32d64a95cca488c3a1ab7eb2ef6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/1e972186a38ff264ae692b72fc9f84c1e8dd9970"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/1e972186a38ff264ae692b72fc9f84c1e8dd9970"}]},"branch":"refs/heads/master"},"add7353c376a8b95465035dadd66e95e3daccb17":{"kind":"REWORK","_number":5,"created":"2020-01-08 00:06:48.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/30/701330/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor-dashboard","ref":"refs/changes/30/701330/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/30/701330/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/30/701330/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/30/701330/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor-dashboard refs/changes/30/701330/5"}}},"commit":{"parents":[{"commit":"95fb9d2a4938453ef303acaf9f19a424d55be51b","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/95fb9d2a4938453ef303acaf9f19a424d55be51b"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-01-07 07:00:42.000000000","tz":0},"committer":{"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","date":"2020-01-08 00:06:41.000000000","tz":0},"subject":"Drop Django 1.11 support","message":"Drop Django 1.11 support\n\nDjango 1.11 ends its extended support in April 2020 (which is before\nUssuri release), so horizon drops Django 1.11 support in Ussuri.\n\ntox envs for non-primary Django versions are no longer needed in tox.ini\nas testing environments for non-primary Django versions are setup in\nthe zuul jobs now.\n\nhorizon\u003e\u003d17.1.0 is required to use Django 2.2.\nrequirements.txt is updated accordingly.\nFor more info. please refer [1], [2].\n\n[1] https://review.opendev.org/#/c/700733/\n[2] https://etherpad.openstack.org/p/Enable_Django22_support_in_Horizon_Plugin\n\nChange-Id: Ifc49a69d4b34b32d64a95cca488c3a1ab7eb2ef6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/add7353c376a8b95465035dadd66e95e3daccb17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/add7353c376a8b95465035dadd66e95e3daccb17"}]},"branch":"refs/heads/master"},"db5ee220acb382924947caa515a9f8de37165e62":{"kind":"REWORK","_number":6,"created":"2020-01-21 06:33:37.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/30/701330/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor-dashboard","ref":"refs/changes/30/701330/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/30/701330/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/30/701330/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/30/701330/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor-dashboard refs/changes/30/701330/6"}}},"commit":{"parents":[{"commit":"4630547b8d24a51c0111fdd9ca5bbbeec15fdc80","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/4630547b8d24a51c0111fdd9ca5bbbeec15fdc80"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-01-07 07:00:42.000000000","tz":0},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-01-21 06:33:18.000000000","tz":0},"subject":"Drop Django 1.11 support","message":"Drop Django 1.11 support\n\nDjango 1.11 ends its extended support in April 2020 (which is before\nUssuri release), so horizon drops Django 1.11 support in Ussuri.\n\ntox envs for non-primary Django versions are no longer needed in tox.ini\nas testing environments for non-primary Django versions are setup in\nthe zuul jobs now.\n\nhorizon\u003e\u003d17.1.0 is required to use Django 2.2.\nrequirements.txt is updated accordingly.\nFor more info. please refer [1], [2].\n\n[1] https://review.opendev.org/#/c/700733/\n[2] https://etherpad.openstack.org/p/Enable_Django22_support_in_Horizon_Plugin\n\nChange-Id: Ifc49a69d4b34b32d64a95cca488c3a1ab7eb2ef6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/db5ee220acb382924947caa515a9f8de37165e62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/db5ee220acb382924947caa515a9f8de37165e62"}]},"branch":"refs/heads/master"},"167fedd54df85e97c859a9213080fb45baa97f6f":{"kind":"REWORK","_number":7,"created":"2020-01-21 06:54:58.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/30/701330/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor-dashboard","ref":"refs/changes/30/701330/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/30/701330/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/30/701330/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/30/701330/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor-dashboard refs/changes/30/701330/7"}}},"commit":{"parents":[{"commit":"4630547b8d24a51c0111fdd9ca5bbbeec15fdc80","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/4630547b8d24a51c0111fdd9ca5bbbeec15fdc80"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-01-07 07:00:42.000000000","tz":0},"committer":{"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","date":"2020-01-21 06:54:54.000000000","tz":0},"subject":"Drop Django 1.11 support","message":"Drop Django 1.11 support\n\nDjango 1.11 ends its extended support in April 2020 (which is before\nUssuri release), so horizon drops Django 1.11 support in Ussuri.\n\ntox envs for non-primary Django versions are no longer needed in tox.ini\nas testing environments for non-primary Django versions are setup in\nthe zuul jobs now.\n\nhorizon\u003e\u003d17.1.0 is required to use Django 2.2.\nrequirements.txt is updated accordingly.\nFor more info. please refer [1], [2].\n\n[1] https://review.opendev.org/#/c/700733/\n[2] https://etherpad.openstack.org/p/Enable_Django22_support_in_Horizon_Plugin\n\nChange-Id: Ifc49a69d4b34b32d64a95cca488c3a1ab7eb2ef6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/167fedd54df85e97c859a9213080fb45baa97f6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/167fedd54df85e97c859a9213080fb45baa97f6f"}]},"branch":"refs/heads/master"},"26f714e5f5d94d062caf250a448ac2b06b0cf74d":{"kind":"REWORK","_number":8,"created":"2020-01-21 07:08:17.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/30/701330/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor-dashboard","ref":"refs/changes/30/701330/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/30/701330/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/30/701330/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/30/701330/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor-dashboard refs/changes/30/701330/8"}}},"commit":{"parents":[{"commit":"4630547b8d24a51c0111fdd9ca5bbbeec15fdc80","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/4630547b8d24a51c0111fdd9ca5bbbeec15fdc80"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-01-07 07:00:42.000000000","tz":0},"committer":{"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","date":"2020-01-21 07:06:54.000000000","tz":0},"subject":"Drop Django 1.11 support","message":"Drop Django 1.11 support\n\nDjango 1.11 ends its extended support in April 2020 (which is before\nUssuri release), so horizon drops Django 1.11 support in Ussuri.\n\ntox envs for non-primary Django versions are no longer needed in tox.ini\nas testing environments for non-primary Django versions are setup in\nthe zuul jobs now.\n\nhorizon\u003e\u003d17.1.0 is required to use Django 2.2.\nrequirements.txt is updated accordingly.\nFor more info. please refer [1], [2].\n\n[1] https://review.opendev.org/#/c/700733/\n[2] https://etherpad.openstack.org/p/Enable_Django22_support_in_Horizon_Plugin\n\nChange-Id: Ifc49a69d4b34b32d64a95cca488c3a1ab7eb2ef6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/26f714e5f5d94d062caf250a448ac2b06b0cf74d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/26f714e5f5d94d062caf250a448ac2b06b0cf74d"}]},"branch":"refs/heads/master"},"37229ca873a5a0701acd6a5aeaa56c08ee655be6":{"kind":"REWORK","_number":9,"created":"2020-01-21 07:32:52.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/30/701330/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor-dashboard","ref":"refs/changes/30/701330/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/30/701330/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/30/701330/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/30/701330/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor-dashboard refs/changes/30/701330/9"}}},"commit":{"parents":[{"commit":"f2977418ec25302792185843bb4802aec473f836","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/f2977418ec25302792185843bb4802aec473f836"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-01-07 07:00:42.000000000","tz":0},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-01-21 07:32:09.000000000","tz":540},"subject":"Drop Django 1.11 support","message":"Drop Django 1.11 support\n\nDjango 1.11 ends its extended support in April 2020 (which is before\nUssuri release), so horizon drops Django 1.11 support in Ussuri.\n\ntox envs for non-primary Django versions are no longer needed in tox.ini\nas testing environments for non-primary Django versions are setup in\nthe zuul jobs now.\n\nhorizon\u003e\u003d17.1.0 is required to use Django 2.2.\nrequirements.txt is updated accordingly.\nFor more info. please refer [1], [2].\n\n[1] https://review.opendev.org/#/c/700733/\n[2] https://etherpad.openstack.org/p/Enable_Django22_support_in_Horizon_Plugin\n\nChange-Id: Ifc49a69d4b34b32d64a95cca488c3a1ab7eb2ef6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/37229ca873a5a0701acd6a5aeaa56c08ee655be6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/37229ca873a5a0701acd6a5aeaa56c08ee655be6"}]},"branch":"refs/heads/master"},"af0f54f7e490bc88505754d30be90eeb2aa5792e":{"kind":"REWORK","_number":10,"created":"2020-01-21 07:45:25.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/30/701330/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor-dashboard","ref":"refs/changes/30/701330/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/30/701330/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/30/701330/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor-dashboard refs/changes/30/701330/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor-dashboard refs/changes/30/701330/10"}}},"commit":{"parents":[{"commit":"f2977418ec25302792185843bb4802aec473f836","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/f2977418ec25302792185843bb4802aec473f836"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-01-07 07:00:42.000000000","tz":0},"committer":{"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","date":"2020-01-21 07:45:12.000000000","tz":0},"subject":"Drop Django 1.11 support","message":"Drop Django 1.11 support\n\nDjango 1.11 ends its extended support in April 2020 (which is before\nUssuri release), so horizon drops Django 1.11 support in Ussuri.\n\ntox envs for non-primary Django versions are no longer needed in tox.ini\nas testing environments for non-primary Django versions are setup in\nthe zuul jobs now.\n\nhorizon\u003e\u003d17.1.0 is required to use Django 2.2.\nrequirements.txt is updated accordingly.\nFor more info. please refer [1], [2].\n\n[1] https://review.opendev.org/#/c/700733/\n[2] https://etherpad.openstack.org/p/Enable_Django22_support_in_Horizon_Plugin\n\nChange-Id: Ifc49a69d4b34b32d64a95cca488c3a1ab7eb2ef6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/af0f54f7e490bc88505754d30be90eeb2aa5792e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor-dashboard/commit/af0f54f7e490bc88505754d30be90eeb2aa5792e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
