)]}'
{"id":"openstack%2Fkarbor~658453","triplet_id":"openstack%2Fkarbor~master~I0d7ffaa0873d192aeb24c17191683d705044644c","project":"openstack/karbor","branch":"master","topic":"bug/1797462","hashtags":[],"change_id":"I0d7ffaa0873d192aeb24c17191683d705044644c","subject":"Implements karbor events notifications","status":"MERGED","created":"2019-05-10 18:58:43.000000000","updated":"2019-07-12 01:47:32.000000000","submitted":"2019-07-12 01:47:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658453-1562896052084-cc061412","meta_rev_id":"8dd8d23adcfa768ec914ed97df85ef0da348f480","_number":658453,"virtual_id_number":658453,"owner":{"_account_id":29299,"name":"Genevieve Nantel","email":"genevieve.nantel@ormuco.com","username":"genevieve.nantel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28721,"name":"Manuel Rodriguez","email":"manu.rodri.hernan@gmail.com","username":"manu.rodriguez"},{"value":0,"_account_id":29299,"name":"Genevieve Nantel","email":"genevieve.nantel@ormuco.com","username":"genevieve.nantel"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":30661,"name":"Jorge Franco","email":"jorgefrancoibanez@gmail.com","username":"JorgeFrancoIbanez"},{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":0,"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"value":0,"_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"value":2,"date":"2019-07-12 01:47:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26925,"name":"Mohamed El Gindi","email":"mohamed.gindi@ormuco.com","username":"mogindi"},{"value":0,"_account_id":30047,"name":"Bruca Lock","email":"lockshi@gmail.com"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"recommended":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"all":[{"value":0,"_account_id":28721,"name":"Manuel Rodriguez","email":"manu.rodri.hernan@gmail.com","username":"manu.rodriguez"},{"value":0,"_account_id":29299,"name":"Genevieve Nantel","email":"genevieve.nantel@ormuco.com","username":"genevieve.nantel"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":1,"date":"2019-07-11 15:46:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30661,"name":"Jorge Franco","email":"jorgefrancoibanez@gmail.com","username":"JorgeFrancoIbanez"},{"value":1,"date":"2019-07-03 15:02:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":2,"date":"2019-07-04 02:27:27.000000000","_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"value":0,"_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},{"value":1,"date":"2019-07-10 13:29:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26925,"name":"Mohamed El Gindi","email":"mohamed.gindi@ormuco.com","username":"mogindi"},{"value":1,"date":"2019-07-08 20:04:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30047,"name":"Bruca Lock","email":"lockshi@gmail.com"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"all":[{"value":0,"_account_id":28721,"name":"Manuel Rodriguez","email":"manu.rodri.hernan@gmail.com","username":"manu.rodriguez"},{"value":0,"_account_id":29299,"name":"Genevieve Nantel","email":"genevieve.nantel@ormuco.com","username":"genevieve.nantel"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":30661,"name":"Jorge Franco","email":"jorgefrancoibanez@gmail.com","username":"JorgeFrancoIbanez"},{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":1,"date":"2019-07-12 01:38:02.000000000","_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"value":0,"_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26925,"name":"Mohamed El Gindi","email":"mohamed.gindi@ormuco.com","username":"mogindi"},{"value":0,"_account_id":30047,"name":"Bruca Lock","email":"lockshi@gmail.com"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26925,"name":"Mohamed El Gindi","email":"mohamed.gindi@ormuco.com","username":"mogindi"},{"_account_id":28721,"name":"Manuel Rodriguez","email":"manu.rodri.hernan@gmail.com","username":"manu.rodriguez"},{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"_account_id":29299,"name":"Genevieve Nantel","email":"genevieve.nantel@ormuco.com","username":"genevieve.nantel"},{"_account_id":30047,"name":"Bruca Lock","email":"lockshi@gmail.com"},{"_account_id":30661,"name":"Jorge Franco","email":"jorgefrancoibanez@gmail.com","username":"JorgeFrancoIbanez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-10 18:59:17.000000000","updated_by":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"reviewer":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"state":"REVIEWER"},{"updated":"2019-06-18 21:13:16.000000000","updated_by":{"_account_id":26925,"name":"Mohamed El Gindi","email":"mohamed.gindi@ormuco.com","username":"mogindi"},"reviewer":{"_account_id":26925,"name":"Mohamed El Gindi","email":"mohamed.gindi@ormuco.com","username":"mogindi"},"state":"REVIEWER"},{"updated":"2019-06-18 21:15:04.000000000","updated_by":{"_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},"reviewer":{"_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},"state":"REVIEWER"},{"updated":"2019-06-18 21:16:40.000000000","updated_by":{"_account_id":28721,"name":"Manuel Rodriguez","email":"manu.rodri.hernan@gmail.com","username":"manu.rodriguez"},"reviewer":{"_account_id":28721,"name":"Manuel Rodriguez","email":"manu.rodri.hernan@gmail.com","username":"manu.rodriguez"},"state":"REVIEWER"},{"updated":"2019-07-03 15:02:23.000000000","updated_by":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"reviewer":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"state":"REVIEWER"},{"updated":"2019-07-08 20:04:36.000000000","updated_by":{"_account_id":30047,"name":"Bruca Lock","email":"lockshi@gmail.com"},"reviewer":{"_account_id":30047,"name":"Bruca Lock","email":"lockshi@gmail.com"},"state":"REVIEWER"},{"updated":"2019-07-10 13:29:34.000000000","updated_by":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"reviewer":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"state":"REVIEWER"},{"updated":"2019-07-11 15:46:05.000000000","updated_by":{"_account_id":30661,"name":"Jorge Franco","email":"jorgefrancoibanez@gmail.com","username":"JorgeFrancoIbanez"},"reviewer":{"_account_id":30661,"name":"Jorge Franco","email":"jorgefrancoibanez@gmail.com","username":"JorgeFrancoIbanez"},"state":"REVIEWER"},{"updated":"2019-07-12 01:38:02.000000000","updated_by":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"reviewer":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"state":"REVIEWER"},{"updated":"2019-07-12 01:47:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4e4db25bc298cf943bff6f8c62c7c69a2435e0fb","author":{"_account_id":29299,"name":"Genevieve Nantel","email":"genevieve.nantel@ormuco.com","username":"genevieve.nantel"},"date":"2019-05-10 18:58:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ece9247d8f7fa166fc9c5bcd09eafcf3f53b7c26","author":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"date":"2019-05-10 18:59:17.000000000","message":"Patch Set 1:\n\nThank you for your first contribution to OpenStack.\n\nYour patch will now be tested automatically by OpenStack testing frameworks\nand once the automatic tests pass, it will be reviewed by other friendly\ndevelopers. They will give you feedback and may require you to refine it.\n\nPeople seldom get their patch approved on the first try, so don\u0027t be\nconcerned if requested to make corrections. Feel free to modify your patch\nand resubmit a new change-set.\n\nPatches usually take 3 to 7 days to be reviewed so be patient and be\navailable on IRC to ask and answer questions about your work. Also it\ntakes generally at least a couple of weeks for cores to get around to\nreviewing code. The more you participate in the community the more\nrewarding it is for you. You may also notice that the more you get to know\npeople and get to be known, the faster your patches will be reviewed and\neventually approved. Get to know others and become known by doing code\nreviews: anybody can do it, and it\u0027s a great way to learn the code base.\n\nThanks again for supporting OpenStack, we look forward to working with you.\n\nIRC: https://wiki.openstack.org/wiki/IRC\nWorkflow: https://docs.openstack.org/infra/manual/developers.html\nCommit Messages: https://wiki.openstack.org/wiki/GitCommitMessages","accounts_in_message":[],"_revision_number":1},{"id":"049e0e8ec5071ec01902b93279aa16162313211a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 20:15:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/53/658453/1/check/openstack-tox-cover/ce996f4/cover/ : SUCCESS in 5m 56s\n- openstack-tox-lower-constraints http://logs.openstack.org/53/658453/1/check/openstack-tox-lower-constraints/52b3841/ : SUCCESS in 6m 39s\n- openstack-tox-pep8 http://logs.openstack.org/53/658453/1/check/openstack-tox-pep8/2fb9dcc/ : SUCCESS in 4m 26s\n- openstack-tox-py27 http://logs.openstack.org/53/658453/1/check/openstack-tox-py27/7cc0d6a/ : SUCCESS in 4m 46s\n- openstack-tox-py35 http://logs.openstack.org/53/658453/1/check/openstack-tox-py35/4c7494c/ : SUCCESS in 5m 42s\n- openstack-tox-py36 http://logs.openstack.org/53/658453/1/check/openstack-tox-py36/ba0ad3a/ : SUCCESS in 5m 39s\n- openstack-tox-docs http://logs.openstack.org/53/658453/1/check/openstack-tox-docs/993b57b/html/ : SUCCESS in 5m 35s\n- karbor-dsvm-fullstack http://logs.openstack.org/53/658453/1/check/karbor-dsvm-fullstack/c22b919/ : SUCCESS in 1h 12m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"94a153f15b327bd2feef99fa41b3789ce166f49d","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2019-05-21 06:00:37.000000000","message":"Patch Set 1:\n\nThanks for your contribution, genevieve nantel\nThis patch should be updated：\n1. Git log message format is not correct, you can refer to this link: https://wiki.openstack.org/wiki/GitCommitMessages\n2. No unit tests were added\n3. It\u0027s better to write a spec file to tell other devs better understand this patch (IMO, this is not a bug, but a new feature, so we should consider it as a blueprint)","accounts_in_message":[],"_revision_number":1},{"id":"d39b5bc54f2fbbb95b246664d2f0458086770ba4","author":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"date":"2019-05-21 15:50:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"363b9901e607e6387a6f6e9455df9985caa2a607","author":{"_account_id":30047,"name":"Bruca Lock","email":"lockshi@gmail.com"},"date":"2019-06-05 14:53:00.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a2e55fffbca51032aa19aad1c60a098c9b56a7c8","author":{"_account_id":29299,"name":"Genevieve Nantel","email":"genevieve.nantel@ormuco.com","username":"genevieve.nantel"},"date":"2019-06-12 18:56:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"330062e46ebdd6c143e2adb3c20a6baffb651b0a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 20:05:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/53/658453/2/check/openstack-tox-cover/4d8491f/cover/ : SUCCESS in 5m 54s\n- openstack-tox-lower-constraints http://logs.openstack.org/53/658453/2/check/openstack-tox-lower-constraints/3c9573e/ : SUCCESS in 6m 50s\n- openstack-tox-pep8 http://logs.openstack.org/53/658453/2/check/openstack-tox-pep8/81dcb68/ : SUCCESS in 5m 33s\n- openstack-tox-py27 http://logs.openstack.org/53/658453/2/check/openstack-tox-py27/fe6f86c/ : SUCCESS in 6m 14s\n- openstack-tox-py35 http://logs.openstack.org/53/658453/2/check/openstack-tox-py35/1ac01ad/ : SUCCESS in 5m 23s\n- openstack-tox-py36 http://logs.openstack.org/53/658453/2/check/openstack-tox-py36/81767ad/ : SUCCESS in 7m 34s\n- openstack-tox-docs http://logs.openstack.org/53/658453/2/check/openstack-tox-docs/78ad09d/html/ : SUCCESS in 5m 38s\n- karbor-dsvm-fullstack http://logs.openstack.org/53/658453/2/check/karbor-dsvm-fullstack/0d737a0/ : FAILURE in 1h 01m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"331bd6f90680c0eb35b5028d746af4bdee634dd0","author":{"_account_id":29299,"name":"Genevieve Nantel","email":"genevieve.nantel@ormuco.com","username":"genevieve.nantel"},"date":"2019-06-12 20:28:14.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"7cb1a0113b031b110539846b0b19d0dffe17dfcd","author":{"_account_id":29299,"name":"Genevieve Nantel","email":"genevieve.nantel@ormuco.com","username":"genevieve.nantel"},"date":"2019-06-12 20:35:33.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"5c5e04e93dbaebe154029aebd486484d51266a3e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 21:58:16.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/53/658453/4/check/openstack-tox-cover/22f1355/cover/ : SUCCESS in 5m 48s\n- openstack-tox-lower-constraints http://logs.openstack.org/53/658453/4/check/openstack-tox-lower-constraints/cd69d4f/ : SUCCESS in 6m 25s\n- openstack-tox-pep8 http://logs.openstack.org/53/658453/4/check/openstack-tox-pep8/c4d5901/ : SUCCESS in 8m 03s\n- openstack-tox-py27 http://logs.openstack.org/53/658453/4/check/openstack-tox-py27/d8092ed/ : SUCCESS in 8m 16s\n- openstack-tox-py35 http://logs.openstack.org/53/658453/4/check/openstack-tox-py35/9bd21e6/ : SUCCESS in 6m 28s\n- openstack-tox-py36 http://logs.openstack.org/53/658453/4/check/openstack-tox-py36/b2c2fed/ : SUCCESS in 7m 36s\n- openstack-tox-docs http://logs.openstack.org/53/658453/4/check/openstack-tox-docs/d9c0da2/html/ : SUCCESS in 6m 02s\n- karbor-dsvm-fullstack http://logs.openstack.org/53/658453/4/check/karbor-dsvm-fullstack/5b7f5b9/ : SUCCESS in 1h 19m 25s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6bdacf65643045c4b3de749e2640a5f57073761e","author":{"_account_id":29299,"name":"Genevieve Nantel","email":"genevieve.nantel@ormuco.com","username":"genevieve.nantel"},"date":"2019-06-14 21:47:34.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"31dbbf6ab752aeab849418bf1a59c24f462ced2c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 22:56:29.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/53/658453/5/check/openstack-tox-cover/c22946d/cover/ : SUCCESS in 7m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/53/658453/5/check/openstack-tox-lower-constraints/814f6f6/ : SUCCESS in 5m 06s\n- openstack-tox-pep8 http://logs.openstack.org/53/658453/5/check/openstack-tox-pep8/5e3bcdd/ : SUCCESS in 5m 23s\n- openstack-tox-py27 http://logs.openstack.org/53/658453/5/check/openstack-tox-py27/3e57524/ : SUCCESS in 5m 52s\n- openstack-tox-py35 http://logs.openstack.org/53/658453/5/check/openstack-tox-py35/093a91d/ : SUCCESS in 6m 17s\n- openstack-tox-py36 http://logs.openstack.org/53/658453/5/check/openstack-tox-py36/78fc68f/ : SUCCESS in 5m 35s\n- openstack-tox-docs http://logs.openstack.org/53/658453/5/check/openstack-tox-docs/ca8b841/html/ : SUCCESS in 5m 56s\n- karbor-dsvm-fullstack http://logs.openstack.org/53/658453/5/check/karbor-dsvm-fullstack/6432479/ : SUCCESS in 56m 21s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"af3117066a7687df5c9ff0b4d6f47ac334beca2a","author":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"date":"2019-06-17 19:03:42.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"93862e26bdee0a3764e20f5ecd9004352c85cf01","author":{"_account_id":26925,"name":"Mohamed El Gindi","email":"mohamed.gindi@ormuco.com","username":"mogindi"},"date":"2019-06-18 21:13:16.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"4a41d500fcf2b33daad2e1d952a0d87f7c673a48","author":{"_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},"date":"2019-06-18 21:15:04.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"821a8cf0c7a898a200b1485ff58238263192a820","author":{"_account_id":28721,"name":"Manuel Rodriguez","email":"manu.rodri.hernan@gmail.com","username":"manu.rodriguez"},"date":"2019-06-18 21:16:40.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"d8760da5255d67cc53ff9e0584a809f5ffe1d379","author":{"_account_id":30047,"name":"Bruca Lock","email":"lockshi@gmail.com"},"date":"2019-06-20 03:16:49.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"c88727d33a4c5092d62973a28e0158bf2d6aaea5","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2019-06-24 07:15:47.000000000","message":"Patch Set 5:\n\n(3 comments)\n\nall the \u0027\u0027\u0027 should be instead with \"\"\" in notification.py.\nIf someone want disable the notifications, how?","accounts_in_message":[],"_revision_number":5},{"id":"7d95348e7570f62e216719144504221e68791242","author":{"_account_id":29299,"name":"Genevieve Nantel","email":"genevieve.nantel@ormuco.com","username":"genevieve.nantel"},"date":"2019-07-03 14:41:41.000000000","message":"Patch Set 5:\n\n\u003e (3 comments)\n \u003e \n \u003e all the \u0027\u0027\u0027 should be instead with \"\"\" in notification.py.\n \u003e If someone want disable the notifications, how?\n\nOk , changing the \u0027\u0027\u0027 to \"\"\" in next patch set . For disabling the notifications completely you just need to set in the config for oslo_messaging_notifications the\ndriver to noop https://github.com/openstack/oslo.messaging/blob/e45dfa7ea207b8f95ab09eeb0c863840db02ff7f/oslo_messaging/notify/notifier.py#L427-L433","accounts_in_message":[],"_revision_number":5},{"id":"3fbb5bcaae0dab4a6c22675c277d5adbbccd1eb0","author":{"_account_id":29299,"name":"Genevieve Nantel","email":"genevieve.nantel@ormuco.com","username":"genevieve.nantel"},"date":"2019-07-03 14:53:53.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"80968cd9e4240aec259b97ba9d67b96df7aec606","author":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"date":"2019-07-03 15:02:23.000000000","message":"Patch Set 6: Code-Review+1\n\nLGTM, good job!","accounts_in_message":[],"_revision_number":6},{"id":"37382ac90846ccb7be42dafbca88d8e488f45640","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 16:08:48.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/53/658453/6/check/openstack-tox-cover/f0d6a98/cover/ : SUCCESS in 4m 01s\n- openstack-tox-lower-constraints http://logs.openstack.org/53/658453/6/check/openstack-tox-lower-constraints/7c63bed/ : SUCCESS in 3m 59s\n- openstack-tox-pep8 http://logs.openstack.org/53/658453/6/check/openstack-tox-pep8/e8f4110/ : SUCCESS in 4m 52s\n- openstack-tox-py27 http://logs.openstack.org/53/658453/6/check/openstack-tox-py27/c5e1c87/ : SUCCESS in 4m 52s\n- openstack-tox-py35 http://logs.openstack.org/53/658453/6/check/openstack-tox-py35/f55fe53/ : SUCCESS in 6m 10s\n- openstack-tox-py36 http://logs.openstack.org/53/658453/6/check/openstack-tox-py36/7f53f93/ : SUCCESS in 5m 12s\n- openstack-tox-docs http://logs.openstack.org/53/658453/6/check/openstack-tox-docs/51aa37f/html/ : SUCCESS in 4m 24s\n- karbor-dsvm-fullstack http://logs.openstack.org/53/658453/6/check/karbor-dsvm-fullstack/b9ba19b/ : SUCCESS in 1h 00m 11s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d4107d8c4f8455fc4284a457742ffb8d5e972790","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2019-07-04 02:27:27.000000000","message":"Patch Set 6: Code-Review+2\n\nThanks! Good job.","accounts_in_message":[],"_revision_number":6},{"id":"29afb0de7d834b92547b797a99d1beba238fffad","author":{"_account_id":30047,"name":"Bruca Lock","email":"lockshi@gmail.com"},"date":"2019-07-08 20:04:36.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"15798d11dc3b84ed5fcaa79f80eb603767cd5f99","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-07-10 13:29:34.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"837f48a74ba38503324b0cdd573d16ce1d4369cc","author":{"_account_id":30661,"name":"Jorge Franco","email":"jorgefrancoibanez@gmail.com","username":"JorgeFrancoIbanez"},"date":"2019-07-11 15:46:05.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"9c43c1455717819a2fa3dfc9b1a00a55c6575878","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2019-07-12 01:38:02.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"41be1cd9a3a3b35c0e27bf4cc6a34c350a7fb377","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 01:38:14.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"d09597dc36685c9c469f72ea5eeada1339d855f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 01:47:31.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/53/658453/6/gate/openstack-tox-lower-constraints/3f41591/ : SUCCESS in 3m 37s\n- openstack-tox-pep8 http://logs.openstack.org/53/658453/6/gate/openstack-tox-pep8/9f5ec9a/ : SUCCESS in 3m 10s\n- openstack-tox-py27 http://logs.openstack.org/53/658453/6/gate/openstack-tox-py27/66e43fe/ : SUCCESS in 3m 55s\n- openstack-tox-py35 http://logs.openstack.org/53/658453/6/gate/openstack-tox-py35/1d0fc61/ : SUCCESS in 3m 52s\n- openstack-tox-py36 http://logs.openstack.org/53/658453/6/gate/openstack-tox-py36/bd87206/ : SUCCESS in 3m 31s\n- openstack-tox-docs http://logs.openstack.org/53/658453/6/gate/openstack-tox-docs/0f16741/html/ : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":6},{"id":"bbc494974b1d58db8f1e6a78ee45ca5d9afa5afe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 01:47:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"e96ec2726caabef7d6e1b064f8dcc2cb0c00d938","revisions":{"2d0f3383bc1d211a64d162bfae5079e934439bb6":{"kind":"REWORK","_number":1,"created":"2019-05-10 18:58:43.000000000","uploader":{"_account_id":29299,"name":"Genevieve Nantel","email":"genevieve.nantel@ormuco.com","username":"genevieve.nantel"},"ref":"refs/changes/53/658453/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor","ref":"refs/changes/53/658453/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor refs/changes/53/658453/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor refs/changes/53/658453/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor refs/changes/53/658453/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor refs/changes/53/658453/1"}}},"commit":{"parents":[{"commit":"39bdc2a89166d7b4e8b5cff2e154cc358491bcfd","subject":"Merge \"Add unit test for plan api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/39bdc2a89166d7b4e8b5cff2e154cc358491bcfd"}]}],"author":{"name":"genevieve-nantel","email":"genevieve.nantel@ormuco.com","date":"2019-05-10 18:58:12.000000000","tz":-240},"committer":{"name":"genevieve-nantel","email":"genevieve.nantel@ormuco.com","date":"2019-05-10 18:58:12.000000000","tz":-240},"subject":"Closes-Bug: 1797462","message":"Closes-Bug: 1797462\n\nChange-Id: I0d7ffaa0873d192aeb24c17191683d705044644c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/2d0f3383bc1d211a64d162bfae5079e934439bb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/2d0f3383bc1d211a64d162bfae5079e934439bb6"}]},"branch":"refs/heads/master"},"d7091379f771cd02ee4b4b9f89389e0871276088":{"kind":"REWORK","_number":2,"created":"2019-06-12 18:56:31.000000000","uploader":{"_account_id":29299,"name":"Genevieve Nantel","email":"genevieve.nantel@ormuco.com","username":"genevieve.nantel"},"ref":"refs/changes/53/658453/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor","ref":"refs/changes/53/658453/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor refs/changes/53/658453/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor refs/changes/53/658453/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor refs/changes/53/658453/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor refs/changes/53/658453/2"}}},"commit":{"parents":[{"commit":"20b9f9e3cd32c05742c7b3d09b6a2cea5f391c64","subject":"Merge \"Add quota support for checkpoints resource\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/20b9f9e3cd32c05742c7b3d09b6a2cea5f391c64"}]}],"author":{"name":"genevieve-nantel","email":"genevieve.nantel@ormuco.com","date":"2019-05-10 18:58:12.000000000","tz":-240},"committer":{"name":"genevieve-nantel","email":"genevieve.nantel@ormuco.com","date":"2019-06-12 18:55:40.000000000","tz":-240},"subject":"Closes-Bug: 1797462","message":"Closes-Bug: 1797462\n\nChange-Id: I0d7ffaa0873d192aeb24c17191683d705044644c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/d7091379f771cd02ee4b4b9f89389e0871276088"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/d7091379f771cd02ee4b4b9f89389e0871276088"}]},"branch":"refs/heads/master"},"7410ae2f78730a6df486e636615e340ca7a36e44":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-06-12 20:28:14.000000000","uploader":{"_account_id":29299,"name":"Genevieve Nantel","email":"genevieve.nantel@ormuco.com","username":"genevieve.nantel"},"ref":"refs/changes/53/658453/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor","ref":"refs/changes/53/658453/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor refs/changes/53/658453/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor refs/changes/53/658453/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor refs/changes/53/658453/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor refs/changes/53/658453/3"}}},"commit":{"parents":[{"commit":"20b9f9e3cd32c05742c7b3d09b6a2cea5f391c64","subject":"Merge \"Add quota support for checkpoints resource\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/20b9f9e3cd32c05742c7b3d09b6a2cea5f391c64"}]}],"author":{"name":"genevieve-nantel","email":"genevieve.nantel@ormuco.com","date":"2019-05-10 18:58:12.000000000","tz":-240},"committer":{"name":"genevieve-nantel","email":"genevieve.nantel@ormuco.com","date":"2019-06-12 20:26:22.000000000","tz":-240},"subject":"Implements:  blueprint karbor-event-notifications Closes-Bug: 1797462","message":"Implements:  blueprint karbor-event-notifications\nCloses-Bug: 1797462\n\nGetting the messages from Karbor like:\n  - Plan create/update/delete\n  - Checkpoint create/update/available\n  - Restore create\n  - Scheduled Operations create/delete\n  - Triggers create/update/delete\n\nThis should add and send the desired messages to RabbitMQ with the start and end identifiers for each message (e.g. karbor.plan_create.start, karbor.plan_create.end) . The notification code was done in a similar fashion than the notifications from the Trove project.\nChange-Id: I0d7ffaa0873d192aeb24c17191683d705044644c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/7410ae2f78730a6df486e636615e340ca7a36e44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/7410ae2f78730a6df486e636615e340ca7a36e44"}]},"branch":"refs/heads/master"},"a98982f711e858ab47512ec588ef24e0da76a781":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-06-12 20:35:33.000000000","uploader":{"_account_id":29299,"name":"Genevieve Nantel","email":"genevieve.nantel@ormuco.com","username":"genevieve.nantel"},"ref":"refs/changes/53/658453/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor","ref":"refs/changes/53/658453/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor refs/changes/53/658453/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor refs/changes/53/658453/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor refs/changes/53/658453/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor refs/changes/53/658453/4"}}},"commit":{"parents":[{"commit":"20b9f9e3cd32c05742c7b3d09b6a2cea5f391c64","subject":"Merge \"Add quota support for checkpoints resource\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/20b9f9e3cd32c05742c7b3d09b6a2cea5f391c64"}]}],"author":{"name":"genevieve-nantel","email":"genevieve.nantel@ormuco.com","date":"2019-05-10 18:58:12.000000000","tz":-240},"committer":{"name":"genevieve-nantel","email":"genevieve.nantel@ormuco.com","date":"2019-06-12 20:33:41.000000000","tz":-240},"subject":"Implements karbor events notifications","message":"Implements karbor events notifications\n\nGetting the messages from Karbor like:\n  - Plan create/update/delete\n  - Checkpoint create/update/available\n  - Restore create\n  - Scheduled Operations create/delete\n  - Triggers create/update/delete\n\nThis should add and send the desired messages to RabbitMQ with the\nstart and end identifiers for each message\n(e.g. karbor.plan_create.start, karbor.plan_create.end). The\nnotification code was done in a similar fashion than the\nnotifications from the Trove project.\n\nImplements:  blueprint karbor-event-notifications\nCloses-Bug: 1797462\nChange-Id: I0d7ffaa0873d192aeb24c17191683d705044644c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/a98982f711e858ab47512ec588ef24e0da76a781"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/a98982f711e858ab47512ec588ef24e0da76a781"}]},"branch":"refs/heads/master"},"862023a269ab0dcfb28f5a7092ccb33fc25e2728":{"kind":"REWORK","_number":5,"created":"2019-06-14 21:47:34.000000000","uploader":{"_account_id":29299,"name":"Genevieve Nantel","email":"genevieve.nantel@ormuco.com","username":"genevieve.nantel"},"ref":"refs/changes/53/658453/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor","ref":"refs/changes/53/658453/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor refs/changes/53/658453/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor refs/changes/53/658453/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor refs/changes/53/658453/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor refs/changes/53/658453/5"}}},"commit":{"parents":[{"commit":"20b9f9e3cd32c05742c7b3d09b6a2cea5f391c64","subject":"Merge \"Add quota support for checkpoints resource\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/20b9f9e3cd32c05742c7b3d09b6a2cea5f391c64"}]}],"author":{"name":"genevieve-nantel","email":"genevieve.nantel@ormuco.com","date":"2019-05-10 18:58:12.000000000","tz":-240},"committer":{"name":"genevieve-nantel","email":"genevieve.nantel@ormuco.com","date":"2019-06-14 21:46:47.000000000","tz":-240},"subject":"Implements karbor events notifications","message":"Implements karbor events notifications\n\nGetting the messages from Karbor like:\n  - Plan create/update/delete\n  - Checkpoint create/update/available\n  - Restore create\n  - Scheduled Operations create/delete\n  - Triggers create/update/delete\n\nThis should add and send the desired messages to RabbitMQ with the\nstart and end identifiers for each message\n(e.g. karbor.plan_create.start, karbor.plan_create.end). The\nnotification code was done in a similar fashion than the\nnotifications from the Trove project.\n\nImplements:  blueprint karbor-event-notifications\nCloses-Bug: 1797462\nChange-Id: I0d7ffaa0873d192aeb24c17191683d705044644c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/862023a269ab0dcfb28f5a7092ccb33fc25e2728"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/862023a269ab0dcfb28f5a7092ccb33fc25e2728"}]},"branch":"refs/heads/master"},"e96ec2726caabef7d6e1b064f8dcc2cb0c00d938":{"kind":"REWORK","_number":6,"created":"2019-07-03 14:53:53.000000000","uploader":{"_account_id":29299,"name":"Genevieve Nantel","email":"genevieve.nantel@ormuco.com","username":"genevieve.nantel"},"ref":"refs/changes/53/658453/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor","ref":"refs/changes/53/658453/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor refs/changes/53/658453/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor refs/changes/53/658453/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor refs/changes/53/658453/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor refs/changes/53/658453/6"}}},"commit":{"parents":[{"commit":"8632a6cfcda741c2a057a26911fea2736a53628b","subject":"Update using storyboard in README","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/8632a6cfcda741c2a057a26911fea2736a53628b"}]}],"author":{"name":"genevieve-nantel","email":"genevieve.nantel@ormuco.com","date":"2019-05-10 18:58:12.000000000","tz":-240},"committer":{"name":"genevieve-nantel","email":"genevieve.nantel@ormuco.com","date":"2019-07-03 14:52:56.000000000","tz":-240},"subject":"Implements karbor events notifications","message":"Implements karbor events notifications\n\nGetting the messages from Karbor like:\n  - Plan create/update/delete\n  - Checkpoint create/update/available\n  - Restore create\n  - Scheduled Operations create/delete\n  - Triggers create/update/delete\n\nThis should add and send the desired messages to RabbitMQ with the\nstart and end identifiers for each message\n(e.g. karbor.plan_create.start, karbor.plan_create.end). The\nnotification code was done in a similar fashion than the\nnotifications from the Trove project.\n\nImplements:  blueprint karbor-event-notifications\nCloses-Bug: 1797462\nChange-Id: I0d7ffaa0873d192aeb24c17191683d705044644c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/e96ec2726caabef7d6e1b064f8dcc2cb0c00d938"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/e96ec2726caabef7d6e1b064f8dcc2cb0c00d938"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
