)]}'
{"id":"openstack%2Fkarbor~668817","triplet_id":"openstack%2Fkarbor~master~Ibf8ee07aa34096887e8e3a9b62176b7d3537055f","project":"openstack/karbor","branch":"master","hashtags":[],"change_id":"Ibf8ee07aa34096887e8e3a9b62176b7d3537055f","subject":"Switch to the new canonical constraints URL on master","status":"MERGED","created":"2019-07-03 07:56:13.000000000","updated":"2019-09-09 11:57:50.000000000","submitted":"2019-09-09 11:56:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668817-1568030200090-98feb54a","meta_rev_id":"0262ab0e796476c2b6a7e6b22f3f1cb8764b699c","_number":668817,"virtual_id_number":668817,"owner":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"value":0,"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"value":2,"date":"2019-09-09 11:56:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"all":[{"value":2,"date":"2019-09-09 09:30:41.000000000","_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"value":0,"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},{"value":1,"date":"2019-09-09 09:30:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"all":[{"value":1,"date":"2019-09-09 11:44:09.000000000","_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"value":0,"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-03 08:01:39.000000000","updated_by":{"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},"reviewer":{"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},"state":"REVIEWER"},{"updated":"2019-09-09 11:44:09.000000000","updated_by":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"reviewer":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"state":"REVIEWER"},{"updated":"2019-09-09 11:56:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2fa0f8d64d10a03c67efa96f54586acc0f66900b","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-07-03 07:56:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ba12e100a94cc7914b7d2bdf65931092227e512","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 09:14:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/17/668817/1/check/build-openstack-api-ref/fe1a91f/html/ : SUCCESS in 3m 56s\n- openstack-tox-cover http://logs.openstack.org/17/668817/1/check/openstack-tox-cover/05c8f17/cover/ : SUCCESS in 4m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/17/668817/1/check/openstack-tox-lower-constraints/299f5e7/ : SUCCESS in 4m 03s\n- openstack-tox-pep8 http://logs.openstack.org/17/668817/1/check/openstack-tox-pep8/3930c00/ : SUCCESS in 5m 10s\n- openstack-tox-py27 http://logs.openstack.org/17/668817/1/check/openstack-tox-py27/f99bb5e/ : SUCCESS in 5m 05s\n- openstack-tox-py35 http://logs.openstack.org/17/668817/1/check/openstack-tox-py35/a0dd9f8/ : SUCCESS in 6m 11s\n- openstack-tox-py36 http://logs.openstack.org/17/668817/1/check/openstack-tox-py36/f6526e5/ : SUCCESS in 3m 51s\n- openstack-tox-docs http://logs.openstack.org/17/668817/1/check/openstack-tox-docs/3d1af49/html/ : SUCCESS in 4m 10s\n- karbor-dsvm-fullstack http://logs.openstack.org/17/668817/1/check/karbor-dsvm-fullstack/8790256/ : SUCCESS in 1h 13m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3583349eb6f6dad29e3a98a74d9b49eb4cda1c3e","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2019-07-03 09:28:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f5e81e0f3a4392bf73ddcc1ef40adadee8791b3c","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-07-17 06:18:52.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"00b4a0767ce6cb9052df9f0520f8bbba67c3508b","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-07-20 03:20:06.000000000","message":"Patch Set 1:\n\nHi PengJu, liushuai, This patch needs your review if you have some free time. Thanks.","accounts_in_message":[],"_revision_number":1},{"id":"4a922dc744e8df1249da41022deb58f4bfc84375","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2019-07-23 11:43:09.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"a52408216cf0fe045745b96b3c18660773001f13","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 12:13:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/17/668817/2/check/build-openstack-api-ref/f37de5e/html/ : SUCCESS in 5m 49s\n- openstack-tox-cover http://logs.openstack.org/17/668817/2/check/openstack-tox-cover/6f3e380/cover/ : SUCCESS in 5m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/17/668817/2/check/openstack-tox-lower-constraints/4c58009/ : SUCCESS in 6m 18s\n- openstack-tox-pep8 http://logs.openstack.org/17/668817/2/check/openstack-tox-pep8/140779c/ : SUCCESS in 4m 24s\n- openstack-tox-py27 http://logs.openstack.org/17/668817/2/check/openstack-tox-py27/56e083b/ : SUCCESS in 4m 37s\n- openstack-tox-py35 http://logs.openstack.org/17/668817/2/check/openstack-tox-py35/04c41f4/ : SUCCESS in 5m 23s\n- openstack-tox-py36 http://logs.openstack.org/17/668817/2/check/openstack-tox-py36/39651ec/ : SUCCESS in 5m 12s\n- openstack-tox-docs http://logs.openstack.org/17/668817/2/check/openstack-tox-docs/1437fa7/html/ : SUCCESS in 5m 07s\n- karbor-dsvm-fullstack http://logs.openstack.org/17/668817/2/check/karbor-dsvm-fullstack/c19d2df/ : FAILURE in 26m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2b3334e0b21e8d0370b05698901d18d2444e9ee1","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2019-07-23 12:36:02.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4354228cea695e4dd87b1836afd2d5d00948f119","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 14:23:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/17/668817/2/check/build-openstack-api-ref/7320718/html/ : SUCCESS in 4m 54s\n- openstack-tox-cover http://logs.openstack.org/17/668817/2/check/openstack-tox-cover/265a9e4/cover/ : SUCCESS in 4m 33s\n- openstack-tox-lower-constraints http://logs.openstack.org/17/668817/2/check/openstack-tox-lower-constraints/d74eb58/ : SUCCESS in 5m 39s\n- openstack-tox-pep8 http://logs.openstack.org/17/668817/2/check/openstack-tox-pep8/b4f062d/ : SUCCESS in 4m 23s\n- openstack-tox-py27 http://logs.openstack.org/17/668817/2/check/openstack-tox-py27/1a7dbe5/ : SUCCESS in 3m 55s\n- openstack-tox-py35 http://logs.openstack.org/17/668817/2/check/openstack-tox-py35/0abfafe/ : SUCCESS in 5m 55s\n- openstack-tox-py36 http://logs.openstack.org/17/668817/2/check/openstack-tox-py36/2a27dd4/ : SUCCESS in 3m 47s\n- openstack-tox-docs http://logs.openstack.org/17/668817/2/check/openstack-tox-docs/9d1af6e/html/ : SUCCESS in 6m 22s\n- karbor-dsvm-fullstack http://logs.openstack.org/17/668817/2/check/karbor-dsvm-fullstack/c1a1d79/ : FAILURE in 1h 40m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"02172f4d027c92310fbc753c3033d1cfe5cba0f1","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-08-19 11:51:43.000000000","message":"Patch Set 2:\n\nHi, Pengju. This patch needs to +1.","accounts_in_message":[],"_revision_number":2},{"id":"34d95411218b67560aab857fe895885caf60427c","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-09-09 02:09:25.000000000","message":"Patch Set 2:\n\n@Pengju. Do you think this patch needs to continue to exist?","accounts_in_message":[],"_revision_number":2},{"id":"134c6a4e144a4099282f4f7538e482e25f0d31ad","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2019-09-09 09:30:41.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"8c6fbe3f19258ae42e73e326e973d43723a98cec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 11:41:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2974fda8ca2642b796b7e9906c1b5ab0 : SUCCESS in 3m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ddd2c4a74f9e449281272a09fe718e93 : SUCCESS in 5m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7767d498895946198ada6941dfed9fce : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e6ed23419324f1ea1d035621ddbc7f6 : SUCCESS in 3m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d52e675126c243ca8a2d1a8e38e24829 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/609c872f8a0d43ef8f1b6255166913b3 : SUCCESS in 5m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9b1e16d690bc40ae89255e0a4e23260a : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd60e5658884412ca076f194f63e148f : SUCCESS in 5m 15s\n- karbor-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/01bad6c65410473488346b8afd4c9a19 : SUCCESS in 1h 05m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"13d616d2591e5149f8249239e4c5fd45512315b9","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2019-09-09 11:44:09.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"2f86ebf8354092ba3aee1d5ad556ddb1b237a85b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 11:44:21.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7efc0c95226733c4ec20e8e3cb46806b055e5c03","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 11:56:39.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/db310f8a48bc498d9021a5e0555b4356 : SUCCESS in 4m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6c30b6b0778a470a90de94d5a8c1235e : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e4b1881db6b46cdbc1a7bf4949631e7 : SUCCESS in 4m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d56e16ba386440a6af64f914ff6df66f : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/131422c27a30431e8f78454ba85176cf : SUCCESS in 4m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/215a9d2297a047e3bbe6487bf99989f6 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecb5fe82b1514c7ab4977d7181020712 : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":3},{"id":"ad5ec819a2ad28dfee2cbb5618376a8bbb906601","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 11:56:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"0262ab0e796476c2b6a7e6b22f3f1cb8764b699c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 11:57:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/f0994c81896b4946917a8666ca8852f8 : SUCCESS in 43s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/431d5d0e33c542d9b710b75db670b645 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4108ed826096bc6c74e873aa87fda1a761ec2ee4","revisions":{"28e1d0b3db53b83bba5a81cef939a1dfc5bb2910":{"kind":"REWORK","_number":1,"created":"2019-07-03 07:56:13.000000000","uploader":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"ref":"refs/changes/17/668817/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor","ref":"refs/changes/17/668817/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor refs/changes/17/668817/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor refs/changes/17/668817/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor refs/changes/17/668817/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor refs/changes/17/668817/1"}}},"commit":{"parents":[{"commit":"8632a6cfcda741c2a057a26911fea2736a53628b","subject":"Update using storyboard in README","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/8632a6cfcda741c2a057a26911fea2736a53628b"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2019-07-03 07:55:07.000000000","tz":480},"committer":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2019-07-03 07:55:59.000000000","tz":480},"subject":"Switch to the new canonical constraints URL on master","message":"Switch to the new canonical constraints URL on master\n\nReference:\n1. http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006478.html\n2. https://github.com/openstack/nova/blob/master/tox.ini#L17\n\nChange-Id: Ibf8ee07aa34096887e8e3a9b62176b7d3537055f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/28e1d0b3db53b83bba5a81cef939a1dfc5bb2910"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/28e1d0b3db53b83bba5a81cef939a1dfc5bb2910"}]},"branch":"refs/heads/master"},"8f3c25c5a29f54cefbe469906e1a35e8e237552d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-07-23 11:43:09.000000000","uploader":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"ref":"refs/changes/17/668817/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor","ref":"refs/changes/17/668817/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor refs/changes/17/668817/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor refs/changes/17/668817/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor refs/changes/17/668817/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor refs/changes/17/668817/2"}}},"commit":{"parents":[{"commit":"2dd7ede588bfb9cd5c15e0d77e3500f793d36890","subject":"Merge \"Summarize output of sample configuration generator\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/2dd7ede588bfb9cd5c15e0d77e3500f793d36890"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2019-07-03 07:55:07.000000000","tz":480},"committer":{"name":"Jiao Pengju","email":"jiaopengju@cmss.chinamobile.com","date":"2019-07-23 11:43:09.000000000","tz":0},"subject":"Switch to the new canonical constraints URL on master","message":"Switch to the new canonical constraints URL on master\n\nReference:\n1. http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006478.html\n2. https://github.com/openstack/nova/blob/master/tox.ini#L17\n\nChange-Id: Ibf8ee07aa34096887e8e3a9b62176b7d3537055f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/8f3c25c5a29f54cefbe469906e1a35e8e237552d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/8f3c25c5a29f54cefbe469906e1a35e8e237552d"}]},"branch":"refs/heads/master"},"4108ed826096bc6c74e873aa87fda1a761ec2ee4":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-09-09 09:30:41.000000000","uploader":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"ref":"refs/changes/17/668817/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor","ref":"refs/changes/17/668817/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor refs/changes/17/668817/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor refs/changes/17/668817/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor refs/changes/17/668817/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor refs/changes/17/668817/3"}}},"commit":{"parents":[{"commit":"799e5fe24d0998025f83a6fc732ff434487a11dd","subject":"Add Python 3 Train unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/799e5fe24d0998025f83a6fc732ff434487a11dd"}]}],"author":{"name":"chenke","email":"chen.ke14@zte.com.cn","date":"2019-07-03 07:55:07.000000000","tz":480},"committer":{"name":"Jiao Pengju","email":"jiaopengju@cmss.chinamobile.com","date":"2019-09-09 09:30:41.000000000","tz":0},"subject":"Switch to the new canonical constraints URL on master","message":"Switch to the new canonical constraints URL on master\n\nReference:\n1. http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006478.html\n2. https://github.com/openstack/nova/blob/master/tox.ini#L17\n\nChange-Id: Ibf8ee07aa34096887e8e3a9b62176b7d3537055f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/4108ed826096bc6c74e873aa87fda1a761ec2ee4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/4108ed826096bc6c74e873aa87fda1a761ec2ee4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
