)]}'
{"id":"openstack%2Fkarbor~681661","triplet_id":"openstack%2Fkarbor~master~I1bd70cea69cc54bea34de4b655f6d7caecfdf0e2","project":"openstack/karbor","branch":"master","hashtags":[],"change_id":"I1bd70cea69cc54bea34de4b655f6d7caecfdf0e2","subject":"Fix unit tests failed","status":"MERGED","created":"2019-09-12 07:59:56.000000000","updated":"2019-09-16 01:13:48.000000000","submitted":"2019-09-16 01:12:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681661-1568596355328-600a17fd","meta_rev_id":"a45981463d049858194dab5618b58926c721bcf5","_number":681661,"virtual_id_number":681661,"owner":{"_account_id":29729,"name":"wangyu","email":"wangyu_yewu@cmss.chinamobile.com","username":"wangyu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},{"value":0,"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"value":2,"date":"2019-09-16 01:12:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},"all":[{"value":2,"date":"2019-09-16 01:00:54.000000000","_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},{"value":2,"date":"2019-09-16 01:01:35.000000000","_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"all":[{"value":0,"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},{"value":1,"date":"2019-09-16 01:01:35.000000000","_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-16 01:00:54.000000000","updated_by":{"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},"reviewer":{"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},"state":"REVIEWER"},{"updated":"2019-09-16 01:01:35.000000000","updated_by":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"reviewer":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"state":"REVIEWER"},{"updated":"2019-09-16 01:12:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bc21b636a836fedd6f9562fda522186a6409a578","author":{"_account_id":29729,"name":"wangyu","email":"wangyu_yewu@cmss.chinamobile.com","username":"wangyu"},"date":"2019-09-12 07:59:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c3526524ac0520e21f2f27c892063fb997df118","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 08:52:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d91beaa6171c4c029246d5a4c716f8be : SUCCESS in 5m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1362dfa51a1249fea4c05bb6952b30e7 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f536909903544fa8a34c87606692b9f : SUCCESS in 5m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6a82ed9290f94275843946aa3e0e8a29 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d580261ee78c4aedac3a62ed8ccca748 : SUCCESS in 5m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ae8ae247bc564db7994c1123f1e6dfeb : SUCCESS in 6m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54fad042850c40f89b9213f1c50a473b : SUCCESS in 7m 14s\n- karbor-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/e03721b92a854025802e5b9d1994f321 : FAILURE in 42m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dfba98644478bcf4cd5907160cf3fd98f9624098","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2019-09-12 12:40:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1b5ebdc8091c397d33303c9e3ac00ccc14c2e540","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 13:52:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/062be2f56a30411a850d835c943ebb77 : SUCCESS in 4m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/656d10c84aea452187964f74b7e57f72 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db06184652f94735abf7c0e57aa417ce : SUCCESS in 4m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f9830331f6d04af8bb256534f5738e48 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04be277455d74ddb9a0509d47a339dba : SUCCESS in 3m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5addf7ccb33d4893bad166894e1a66ee : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1de2c998a9c43c6bc280e28182fe128 : SUCCESS in 5m 31s\n- karbor-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/56c30ddd56154aaeba33aafc8f527760 : SUCCESS in 56m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"23cb874d92be0a48a810b3226f540d5d30cd9481","author":{"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},"date":"2019-09-16 01:00:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"50c82bb5085d5d49d6ae50debf03d827627eccea","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2019-09-16 01:01:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"449ae3917156f38cc6512301d0e23434401735d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 01:01:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9ead1826a64dd461fe2f3fe4d9ea587a6cc1e04b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 01:12:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d82844c97c06482d8fe634a02be9c649 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b5c2d8e44bc447bb3a513b9183db4af : SUCCESS in 3m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bfb0867d24c34abea46cf6336fd432f3 : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb4b4b16fee6422bb402fd5630d7b804 : SUCCESS in 3m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a997a5fe30fb4d32b01d8cd8f6777e4d : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b07787e8125a4d19ad5bf149ea533a3b : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":1},{"id":"176e28ce568d0111310d8e1e0b59cb1678996084","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 01:12:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a45981463d049858194dab5618b58926c721bcf5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 01:13:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f39228d586df44dbbcfba9c1102275ff : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a3fcd6d6be65b88d8b8240eac700192f33262d4e","revisions":{"a3fcd6d6be65b88d8b8240eac700192f33262d4e":{"kind":"REWORK","_number":1,"created":"2019-09-12 07:59:56.000000000","uploader":{"_account_id":29729,"name":"wangyu","email":"wangyu_yewu@cmss.chinamobile.com","username":"wangyu"},"ref":"refs/changes/61/681661/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor","ref":"refs/changes/61/681661/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor refs/changes/61/681661/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor refs/changes/61/681661/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor refs/changes/61/681661/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor refs/changes/61/681661/1"}}},"commit":{"parents":[{"commit":"4108ed826096bc6c74e873aa87fda1a761ec2ee4","subject":"Switch to the new canonical constraints URL on master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/4108ed826096bc6c74e873aa87fda1a761ec2ee4"}]}],"author":{"name":"wangyu","email":"wangyu@cmss.chinamobile.com","date":"2019-09-12 07:51:51.000000000","tz":480},"committer":{"name":"wangyu","email":"wangyu@cmss.chinamobile.com","date":"2019-09-12 08:01:10.000000000","tz":480},"subject":"Fix unit tests failed","message":"Fix unit tests failed\n\nif flow is None for add_tasks when running unit tests,\nthe method \u0027add\u0027 for flow  will be failed. this patch\nwill fix it.\n\nChange-Id: I1bd70cea69cc54bea34de4b655f6d7caecfdf0e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/a3fcd6d6be65b88d8b8240eac700192f33262d4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/a3fcd6d6be65b88d8b8240eac700192f33262d4e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
