)]}'
{"id":"openstack%2Fkarbor~686920","triplet_id":"openstack%2Fkarbor~master~Ice5723615d87481bc47641cc9dbf44fc477290a3","project":"openstack/karbor","branch":"master","hashtags":[],"change_id":"Ice5723615d87481bc47641cc9dbf44fc477290a3","subject":"Add Wiki and admin guide links to readme","status":"MERGED","created":"2019-10-06 07:00:39.000000000","updated":"2019-10-08 12:51:41.000000000","submitted":"2019-10-08 12:50:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686920-1570539017180-0f5a3271","meta_rev_id":"ffb69803543553b69df8dbf8c257c71a5f1c48c8","_number":686920,"virtual_id_number":686920,"owner":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-08 12:50:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"all":[{"value":2,"date":"2019-10-08 12:36:36.000000000","_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"all":[{"value":1,"date":"2019-10-08 12:36:36.000000000","_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-08 12:36:36.000000000","updated_by":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"reviewer":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"state":"REVIEWER"},{"updated":"2019-10-08 12:50:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ae822a731fd70be279bdfdb392e6c193d65cbfab","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2019-10-06 07:00:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c4c60ab4df9e827f01418c697f38734ef78f9ce9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-06 08:02:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/774df4e6f84c4cf5b00b57b845a638fe : SUCCESS in 2m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a76f35af749048568999eb5d566a0c69 : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc2c7d50959f4581a07a5bca50214777 : SUCCESS in 4m 21s\n- karbor-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/f90258b1c75e4bbc8cf496ff7f57fd23 : SUCCESS in 56m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dd03de507f157188a61e6720d5ba9c74ef5f6b1f","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2019-10-08 12:36:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1111eb7416bad68dce810c4275436bdad1ce5045","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 12:36:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2981d34236684a57ca24e126e0546c78b6676951","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 12:50:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/83bb721f69c84974947da5be02f55777 : SUCCESS in 2m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9dd6b7852cc943a09b370d3bc69fef6a : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a1ab00e1a5d4029ba84a3b2a946d3fc : SUCCESS in 6m 36s","accounts_in_message":[],"_revision_number":1},{"id":"6e0c979dc1bddb7c7b01e3598ff271c9f2ca54ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 12:50:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ffb69803543553b69df8dbf8c257c71a5f1c48c8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 12:51:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8af9e0b16a164034835135241609609a : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"09d5cc91c3ac50cc4937693491ee305497271b2a","revisions":{"09d5cc91c3ac50cc4937693491ee305497271b2a":{"kind":"REWORK","_number":1,"created":"2019-10-06 07:00:39.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/20/686920/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor","ref":"refs/changes/20/686920/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor refs/changes/20/686920/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor refs/changes/20/686920/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor refs/changes/20/686920/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor refs/changes/20/686920/1"}}},"commit":{"parents":[{"commit":"8d3c3b52598bd5a1f5c416155e2daf8bab73dc5c","subject":"Merge \"Replace use of log.warn with log.warning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/8d3c3b52598bd5a1f5c416155e2daf8bab73dc5c"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2019-10-06 06:54:41.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2019-10-06 06:54:41.000000000","tz":480},"subject":"Add Wiki and admin guide links to readme","message":"Add Wiki and admin guide links to readme\n\nChange-Id: Ice5723615d87481bc47641cc9dbf44fc477290a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/09d5cc91c3ac50cc4937693491ee305497271b2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/09d5cc91c3ac50cc4937693491ee305497271b2a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
