)]}'
{"id":"openstack%2Fkarbor~692692","triplet_id":"openstack%2Fkarbor~master~I8be62bb00e70dacd174e7fa3bd65489fb63488ab","project":"openstack/karbor","branch":"master","topic":"defautl-service","hashtags":[],"change_id":"I8be62bb00e70dacd174e7fa3bd65489fb63488ab","subject":"Use trustee project_name instead of fixed value","status":"MERGED","created":"2019-11-03 03:17:37.000000000","updated":"2019-11-27 00:25:49.000000000","submitted":"2019-11-14 10:43:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692692-1573728223951-64a7fc44","meta_rev_id":"dfb2ada7f4292915a5c63c80ebcedf50fc274192","_number":692692,"virtual_id_number":692692,"owner":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"value":0,"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-14 10:43:43.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},"all":[{"value":0,"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"value":2,"date":"2019-11-14 10:30:25.000000000","_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},"all":[{"value":0,"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"value":1,"date":"2019-11-14 10:30:25.000000000","_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-14 10:30:25.000000000","updated_by":{"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},"reviewer":{"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},"state":"REVIEWER"},{"updated":"2019-11-14 10:43:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f47498bd9a224bbdb6e4f3351d386f577352c5e4","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2019-11-03 03:17:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b8a5c2b75ab39c2c244b420f50cdf5a2cacf335","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-03 04:14:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb15d9b51dd6466b82b3d78acc794043 : SUCCESS in 4m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/11716fd64a8249fc8fb8d553d860dc6c : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65ea246e7f0648de82f913f913e4ff8b : SUCCESS in 3m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ef6793679f304df79fcedae3ed5ee818 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a64d75d4668e439baf7e686c4536c593 : SUCCESS in 3m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cdd6c8d030bf4204984949f3c007daa6 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99de035377134507b61a87ec13c95ff8 : SUCCESS in 4m 49s\n- karbor-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/954ccfc55b5d4af08ebf2d62f4b7eb70 : FAILURE in 50m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7b41f0f98589fb9e74f836b57f7cb32c4ab13d83","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2019-11-04 07:43:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"45e6e800326492d4170c6e337c47796704de900e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 08:41:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9ec2913baea445db2ccced97a289945 : SUCCESS in 5m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db950a93ba044e10bb5639d4463e284f : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f716fe8a1c904fb7bfc73bc4308f90dd : SUCCESS in 4m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eb8198e7a3e343989729bd9f47a81c6d : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b44f80153ad34d379043f5f4cb3c5150 : SUCCESS in 5m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5278a94aae2a436f8118cd3d9cbf5232 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f81b2a3dd2346679281ec83ca7c9f2a : SUCCESS in 5m 30s\n- karbor-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/31803b039fb64e6bb38c6ed301b3cbe7 : FAILURE in 52m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fde3c1678a4c953c075d492840d7d14708b7d1c9","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2019-11-14 06:24:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b489b76330c61de9691093d860f050f2afc1f484","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 08:13:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f4bf913e58b4fe98ab4ff56285896ae : SUCCESS in 6m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/887490cee35d4c9c8cc441f0cbd9efd9 : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaa6329c2f784623bd89c7ebea34222e : SUCCESS in 6m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/15be660b5db04541807ed67ea69786f6 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3aed6401551742c889288bac74f50058 : SUCCESS in 6m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f3e4cfb8467146b7801cb4181b164ffe : SUCCESS in 8m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32eccf9d372f4cce85fd3a377063ea56 : SUCCESS in 7m 04s\n- karbor-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/66141a08f2f04a7088aeb441f03d60fe : FAILURE in 1h 25m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"90d5d02aeeec0f0529f7c38130fe073e62a3db58","author":{"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},"date":"2019-11-14 10:30:25.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9b26ceb2bda1df04652af339e819ff82931ffc79","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 10:30:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"23542dde72e4a0a12448bf4490b6f56f34ab1ccc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 10:43:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e7a22f4cfc446ceba91fc2d09a90af0 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66ab3d5151b547ca8e2175800f27db41 : SUCCESS in 3m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c38e854a337b46a4a0a1826777db536f : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8677ac48a3143a29d5a9b3b1724ca05 : SUCCESS in 4m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/71328face5d04d7ab4f16597a2e72012 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b90a053ac7a14d8daf6d1e36068c96c1 : SUCCESS in 8m 10s","accounts_in_message":[],"_revision_number":2},{"id":"4b3421dc0e980e278ad5ffa961bb5ba641eb2f85","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 10:43:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"cadef0fad43d381ca808b8762dcc54584b6e71de","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2019-11-27 00:25:49.000000000","message":"Created a revert of this change as I4a8ada268c295505c6159b4eaf71fe77fb2b56eb","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"82f07c14b6a1c5a9303e37001e85cc30df796dcc","revisions":{"c4f84e19c8c1a6452b70c7bba9df7da0253f3bf3":{"kind":"REWORK","_number":1,"created":"2019-11-03 03:17:37.000000000","uploader":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"ref":"refs/changes/92/692692/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor","ref":"refs/changes/92/692692/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor refs/changes/92/692692/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor refs/changes/92/692692/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor refs/changes/92/692692/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor refs/changes/92/692692/1"}}},"commit":{"parents":[{"commit":"08cef4fae621fa0f29ce80393a8ba9fce091cb80","subject":"Merge \"Remove invalid none value judgment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/08cef4fae621fa0f29ce80393a8ba9fce091cb80"}]}],"author":{"name":"jiaopengju","email":"jiaopengju@cmss.chinamobile.com","date":"2019-11-03 03:14:33.000000000","tz":480},"committer":{"name":"jiaopengju","email":"jiaopengju@cmss.chinamobile.com","date":"2019-11-03 03:16:58.000000000","tz":480},"subject":"Use trustee project_name instead of fixed value","message":"Use trustee project_name instead of fixed value\n\nCurrently, when the default project name for openstack\nservices is not \"service\", it will fail when listing\nprotectable resources. This patch using trustee.project_name\nto replace it.\n\nChange-Id: I8be62bb00e70dacd174e7fa3bd65489fb63488ab\nStory: 2006797\nTask: 37343\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/c4f84e19c8c1a6452b70c7bba9df7da0253f3bf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/c4f84e19c8c1a6452b70c7bba9df7da0253f3bf3"}]},"branch":"refs/heads/master"},"82f07c14b6a1c5a9303e37001e85cc30df796dcc":{"kind":"REWORK","_number":2,"created":"2019-11-14 06:24:58.000000000","uploader":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"ref":"refs/changes/92/692692/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor","ref":"refs/changes/92/692692/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor refs/changes/92/692692/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor refs/changes/92/692692/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor refs/changes/92/692692/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor refs/changes/92/692692/2"}}},"commit":{"parents":[{"commit":"08cef4fae621fa0f29ce80393a8ba9fce091cb80","subject":"Merge \"Remove invalid none value judgment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/08cef4fae621fa0f29ce80393a8ba9fce091cb80"}]}],"author":{"name":"jiaopengju","email":"jiaopengju@cmss.chinamobile.com","date":"2019-11-03 03:14:33.000000000","tz":480},"committer":{"name":"jiaopengju","email":"jiaopengju@cmss.chinamobile.com","date":"2019-11-14 06:24:29.000000000","tz":480},"subject":"Use trustee project_name instead of fixed value","message":"Use trustee project_name instead of fixed value\n\nCurrently, when the default project name for openstack\nservices is not \"service\", it will fail when listing\nprotectable resources. This patch using trustee.project_name\nto replace it.\n\nStory: 2006797\nTask: 37343\n\nChange-Id: I8be62bb00e70dacd174e7fa3bd65489fb63488ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/82f07c14b6a1c5a9303e37001e85cc30df796dcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/82f07c14b6a1c5a9303e37001e85cc30df796dcc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
