)]}'
{"id":"openstack%2Fkarbor~693956","triplet_id":"openstack%2Fkarbor~master~Id84f40f5786d96438fc7fb84dafc96372a28d634","project":"openstack/karbor","branch":"master","topic":"api-doc-verifications","hashtags":[],"change_id":"Id84f40f5786d96438fc7fb84dafc96372a28d634","subject":"Add verifications api doc","status":"MERGED","created":"2019-11-13 01:15:48.000000000","updated":"2019-11-14 10:46:04.000000000","submitted":"2019-11-14 10:43:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693956-1573728227826-e0020214","meta_rev_id":"e3a0a3dc254d2fe80295435e21f69a1032467bbb","_number":693956,"virtual_id_number":693956,"owner":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"value":0,"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-14 10:43:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"all":[{"value":2,"date":"2019-11-14 01:46:39.000000000","_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"value":2,"date":"2019-11-14 10:32:08.000000000","_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},{"value":1,"date":"2019-11-14 03:05:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},"all":[{"value":0,"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"value":1,"date":"2019-11-14 10:32:08.000000000","_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-14 03:05:46.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-11-14 10:32:08.000000000","updated_by":{"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},"reviewer":{"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},"state":"REVIEWER"},{"updated":"2019-11-14 10:43:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2920f146c9b566c14d85702af20c21375e13e106","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2019-11-13 01:15:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"66f568c087117c9b6a16913c2337752ddc71346d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 02:09:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/899c802da444424fbacb24c21860b103 : FAILURE in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c06ec468baa40abbb64e6cf289109dd : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/631549b439e741db9716ffee61065d37 : SUCCESS in 4m 34s\n- karbor-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/ca8177e0a485465ba22ada98d866c465 : SUCCESS in 52m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ed32bf6c3f62ba27bcf69fb1479161264a0e5bee","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2019-11-13 02:49:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7f59f9907c3c3f8efed53cf35a72e24d74e9a430","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 03:39:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0acfd712071c4254bf57ff083cce53f3 : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61430aa85fbb44e39f70d11a4b54d6c4 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/878f176e0aeb4303a5ac25b6688ec255 : SUCCESS in 5m 15s\n- karbor-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/12f0a7dd34204f8d83fcadb787308d18 : SUCCESS in 48m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9268510b940dce84a1b0a49cf33a6f1870b6fc6f","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2019-11-14 01:46:39.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6a1ca3ee3ae37cb6516912c241a69143bdc604be","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-11-14 03:05:46.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"1d1186d7e6178c3a22bb67a32fae6b812b53b2fb","author":{"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},"date":"2019-11-14 10:32:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8fc609d8939f038a8dad5fb778352a4c6a2d0a52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 10:32:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"674734ebb194e0e02da15df8a5e5e7203a3bfd7f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 10:43:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"4a3aea3525da6827dfc7b1265d0e7f67150e773c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 10:43:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/27914891a848452b99d08a0af5834154 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f72d8bfe8c524f11b1c44bc9b0c1b21c : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0308db6d4c174950bc8606f9a45d2deb : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":2},{"id":"b8ee6f37e5c6a2e15dde6f7841a88bbacb702f25","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 10:46:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/cbca37d18d214951b88ee231b1ca3e05 : SUCCESS in 1m 09s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7c75324a5764e439a5ec97f836c4ce9 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0970daeb5d6e1985a4162a773538f9eddacca137","revisions":{"06405a9d2b017cb13763da2c5848a5305c30cc3a":{"kind":"REWORK","_number":1,"created":"2019-11-13 01:15:48.000000000","uploader":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"ref":"refs/changes/56/693956/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor","ref":"refs/changes/56/693956/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor refs/changes/56/693956/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor refs/changes/56/693956/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor refs/changes/56/693956/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor refs/changes/56/693956/1"}}},"commit":{"parents":[{"commit":"c5c16090a60134fadaf51754be10d5fca16d36c6","subject":"Merge \"Add services api doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/c5c16090a60134fadaf51754be10d5fca16d36c6"}]}],"author":{"name":"jiaopengju","email":"jiaopengju@cmss.chinamobile.com","date":"2019-11-13 01:14:44.000000000","tz":480},"committer":{"name":"jiaopengju","email":"jiaopengju@cmss.chinamobile.com","date":"2019-11-13 01:14:44.000000000","tz":480},"subject":"Add verifications api doc","message":"Add verifications api doc\n\nChange-Id: Id84f40f5786d96438fc7fb84dafc96372a28d634\nStory: 2006807\nTask: 37357\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/06405a9d2b017cb13763da2c5848a5305c30cc3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/06405a9d2b017cb13763da2c5848a5305c30cc3a"}]},"branch":"refs/heads/master"},"0970daeb5d6e1985a4162a773538f9eddacca137":{"kind":"REWORK","_number":2,"created":"2019-11-13 02:49:38.000000000","uploader":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"ref":"refs/changes/56/693956/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor","ref":"refs/changes/56/693956/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor refs/changes/56/693956/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor refs/changes/56/693956/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor refs/changes/56/693956/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor refs/changes/56/693956/2"}}},"commit":{"parents":[{"commit":"c5c16090a60134fadaf51754be10d5fca16d36c6","subject":"Merge \"Add services api doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/c5c16090a60134fadaf51754be10d5fca16d36c6"}]}],"author":{"name":"jiaopengju","email":"jiaopengju@cmss.chinamobile.com","date":"2019-11-13 01:14:44.000000000","tz":480},"committer":{"name":"jiaopengju","email":"jiaopengju@cmss.chinamobile.com","date":"2019-11-13 02:49:26.000000000","tz":480},"subject":"Add verifications api doc","message":"Add verifications api doc\n\nChange-Id: Id84f40f5786d96438fc7fb84dafc96372a28d634\nStory: 2006807\nTask: 37357\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/0970daeb5d6e1985a4162a773538f9eddacca137"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/0970daeb5d6e1985a4162a773538f9eddacca137"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
