)]}'
{"id":"openstack%2Fkarbor~694190","triplet_id":"openstack%2Fkarbor~master~I9ae08550f74347fc8f8436eca3cf4320ec97dd1f","project":"openstack/karbor","branch":"master","topic":"master-time-interval","hashtags":[],"change_id":"I9ae08550f74347fc8f8436eca3cf4320ec97dd1f","subject":"Modify the time interval configurations in devstack","status":"MERGED","created":"2019-11-14 02:50:14.000000000","updated":"2019-11-14 05:52:45.000000000","submitted":"2019-11-14 05:51:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694190-1573710698120-a3a83055","meta_rev_id":"92f3137742e799ccdd5b67afc992ad0333d8b8dd","_number":694190,"virtual_id_number":694190,"owner":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"value":0,"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-14 05:51:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"all":[{"value":2,"date":"2019-11-14 05:42:25.000000000","_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"value":2,"date":"2019-11-14 05:43:33.000000000","_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},"all":[{"value":0,"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"value":1,"date":"2019-11-14 05:43:33.000000000","_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-14 05:43:33.000000000","updated_by":{"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},"reviewer":{"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},"state":"REVIEWER"},{"updated":"2019-11-14 05:51:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e33c0ffb03c03cbbc49c6f76696605dfcb25b6a7","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2019-11-14 02:50:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2637911c3b5675b3233ce1aa580a9ae866032e95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 03:42:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b698d1643ed249718dac8e6add1a7c9c : SUCCESS in 4m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f390dddeea2405a99158ca1023caa65 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cad39a0639b54d9ab58a33d02e6870a7 : SUCCESS in 3m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f6980a1438214a1bbbf24cce1b9e8afa : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a3ee34e311e447bb6d5fd4e1a4ae626 : SUCCESS in 3m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fa28ae571d16487a8bf4141201670076 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e25d3fc70f2146f6885edce06b6d8eab : SUCCESS in 4m 37s\n- karbor-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/35d0b1b908b64e12970d25f9fe965e06 : SUCCESS in 49m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ca15f2fa25c84728ef2797d7b4cceed91700c606","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2019-11-14 05:42:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"313de92a94ae09e78b127862b5575a3cae6b53d6","author":{"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},"date":"2019-11-14 05:43:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4876ed32e056b6ca10ac373abcd4a29da287de45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 05:43:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7d0359e3774cf0b49bb9ddf321b3fdba560e9531","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 05:51:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dfb5a473a86a41f7955c6f8f404c99a0 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a84a27d2d1154671a30ff64f3bef4458 : SUCCESS in 3m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d89dbfda37b24a248b3e2c332d2711e7 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93f1d77c55de4e87bb71356498a0f96f : SUCCESS in 3m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ddbfad1c652e4d9c8a1e38b746a43862 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7a30f07a307473397c9d9c8c54886dd : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":1},{"id":"70d7421fe833773465b66274e802c65d6bebffd8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 05:51:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0b9dd65dd0a28c9fcef703a0664bed3b13cd3c4e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 05:52:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2039227b503d455381ff8a45bb150383 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6093b5b11e23b984250fd21c268632276440d60e","revisions":{"6093b5b11e23b984250fd21c268632276440d60e":{"kind":"REWORK","_number":1,"created":"2019-11-14 02:50:14.000000000","uploader":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"ref":"refs/changes/90/694190/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor","ref":"refs/changes/90/694190/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor refs/changes/90/694190/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor refs/changes/90/694190/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor refs/changes/90/694190/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor refs/changes/90/694190/1"}}},"commit":{"parents":[{"commit":"28fabef6ec5a88f9ee7da0f3a56628efc71347fb","subject":"Remove invalid exception catching","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/28fabef6ec5a88f9ee7da0f3a56628efc71347fb"}]}],"author":{"name":"jiaopengju","email":"jiaopengju@cmss.chinamobile.com","date":"2019-11-14 02:48:36.000000000","tz":480},"committer":{"name":"jiaopengju","email":"jiaopengju@cmss.chinamobile.com","date":"2019-11-14 02:49:52.000000000","tz":480},"subject":"Modify the time interval configurations in devstack","message":"Modify the time interval configurations in devstack\n\nStory: 2006848\nTask: 37436\n\nChange-Id: I9ae08550f74347fc8f8436eca3cf4320ec97dd1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/6093b5b11e23b984250fd21c268632276440d60e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/6093b5b11e23b984250fd21c268632276440d60e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
