)]}'
{"id":"openstack%2Fkarbor~696214","triplet_id":"openstack%2Fkarbor~master~I4a8ada268c295505c6159b4eaf71fe77fb2b56eb","project":"openstack/karbor","branch":"master","topic":"(HEAD","hashtags":[],"change_id":"I4a8ada268c295505c6159b4eaf71fe77fb2b56eb","subject":"Revert \"Use trustee project_name instead of fixed value\"","status":"ABANDONED","created":"2019-11-27 00:25:49.000000000","updated":"2021-06-09 22:49:54.000000000","total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1a0a77b2d45e1c5bb41673082b4b28f96ac55f47","_number":696214,"virtual_id_number":696214,"owner":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-09-16 10:52:55.000000000","_account_id":29729,"name":"wangyu","email":"wangyu_yewu@cmss.chinamobile.com","username":"wangyu"},{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-09-16 08:12:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29729,"name":"wangyu","email":"wangyu_yewu@cmss.chinamobile.com","username":"wangyu"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":29729,"name":"wangyu","email":"wangyu_yewu@cmss.chinamobile.com","username":"wangyu"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},{"_account_id":29729,"name":"wangyu","email":"wangyu_yewu@cmss.chinamobile.com","username":"wangyu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-27 00:25:49.000000000","updated_by":{"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},"reviewer":{"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},"state":"REVIEWER"},{"updated":"2020-09-16 08:12:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-16 10:52:55.000000000","updated_by":{"_account_id":29729,"name":"wangyu","email":"wangyu_yewu@cmss.chinamobile.com","username":"wangyu"},"reviewer":{"_account_id":29729,"name":"wangyu","email":"wangyu_yewu@cmss.chinamobile.com","username":"wangyu"},"state":"REVIEWER"}],"messages":[{"id":"c81242a4fca678debb1843eb76e00f1410b0415c","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2019-11-27 00:25:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab211939660d868623cc6b592b48a27069593562","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 01:29:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/405361368e544dc3a641176a5ca0f301 : SUCCESS in 4m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/077078e128a5465d870afe0eefbd5dbc : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05bd0859186642fe889040cad15b605d : SUCCESS in 3m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ad7e87385c8c48e0a7d21843352746e8 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d90e28452b814a2b99d668a233572494 : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d7760672d1774cf7affd2a4ad08d9ae4 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca6903fbab1044f18c7920346d8325c1 : SUCCESS in 4m 56s\n- karbor-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/19800895d26a489db00ddcefdd8179d4 : FAILURE in 57m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4dce88ccf130c577190ea4ffe93e37333e3eb2a4","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2019-11-27 01:37:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9a884ac662c38cc7b8652a8e25c36ed7982a39db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 02:29:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5a644cd02dd4778b9f77f92fe62765f : SUCCESS in 5m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7668758c1b084537a5d478215efadd45 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e13751056b54462abd88a0dd53af97c : SUCCESS in 3m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/21b3462696c4498c91110352e2392bf9 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a285a5263add45279f7ae398118bff28 : SUCCESS in 4m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6375faad96f346108e6af57bed1f068d : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54b26bcd860f405cbdd538ecd95a961e : SUCCESS in 5m 13s\n- karbor-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/612f193ffd254901b1e4c0e232900b96 : FAILURE in 49m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d8372e7667e08d34cdd12a1a4c987f9d43ba211d","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2019-11-29 05:55:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"df07caaaa1f53580cff17ee6372ed4ddf6662459","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-29 07:00:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62305c40057e415d90fbb42b170c9e6e : SUCCESS in 4m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/38ad57251e5f48e09f3f007d17fe242f : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/320b70b9b6bc4b6e90edd59469611573 : SUCCESS in 4m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/89404d9bc65e4e8397825dec7eb035b5 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f95935e849e6481f8fba348929614d7c : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5a7c12f4f41747629dfdad9ff734d9a8 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c91cbfd28a5445a8a134d61ecc549af : SUCCESS in 5m 17s\n- karbor-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/a4e749eb3c88444fac3ec670781a73f0 : FAILURE in 1h 01m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fa934aa3c75b78e4a3c3111111a55d27f7598905","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2019-12-02 11:32:29.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"cc594b1fe5a9894aa9986e1554937b7cbe8be386","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 12:25:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/023bcf7fced54cc189816b79050e387e : SUCCESS in 4m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3cdad13a15b74923bcb4836938d4d3da : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d9bf9dee9c049818ff97ba56e2bc6dd : SUCCESS in 3m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a9de9a5d30474cd1b9ad59c30a67d938 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/977e71c6273e42b9a4ac3f6ce252c5d4 : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c308d8573e064e298904309a48dbb6c4 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bd0eb196eed4add90e5ddd1e1c856fa : SUCCESS in 5m 23s\n- karbor-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/022cd022a5f84d3aac4ed478873acf95 : FAILURE in 51m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"64b3824c45f20004b4b2821e7585b5342a12f1b6","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2019-12-24 08:09:25.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4e6e6c1bd475779e21c97df1cab875e473a70fc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-24 09:06:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1dead30c8914486eb52f65fc90f48430 : SUCCESS in 4m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a459c1211dd54670a2d71e2ec699bb3f : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7401730221b4cc1b4d3ddfc04361d5b : SUCCESS in 4m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eccfe71765bf4eb1816a4d5d5ec62f45 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3aaf56ce0c949509d10d7bdf963c8a9 : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7f7e622c5c2848febb286a4c6cd05de9 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92409a124f28473e8a9474aed97c5e1a : SUCCESS in 5m 16s\n- karbor-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/70425427cc99424b8b7cccbd6ca1276a : FAILURE in 51m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e138e38bb3da4c5eb3e2c910b3ac68dc6381d10c","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2020-01-08 08:47:58.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0324f2d292217b2e3e1bf26555fef315cde5970d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 10:45:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13305569e28c4c869ca2bf65745afd90 : SUCCESS in 5m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87092f007aa44616b26c869b2ed3b78a : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35e7b8b2e9994c0aa4deaec3b8d39288 : SUCCESS in 6m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/768afc74ecd74a06bac025f602f160f5 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/beb337a3122c49c4a572613b1653ef41 : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/095e7bb587ed45dfa6fd7d7086270436 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c89dc5f47864e358db4eda37214cbaf : SUCCESS in 5m 26s\n- karbor-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/a8a1851f3dd547c896a96b965ab2a959 : FAILURE in 1h 21m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"74ed60008ae5c59280210789714c36ebd734610b","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2020-08-17 01:08:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"704792fc18707a328f5424874fc22f3c76c0fd43","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 01:08:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"17f5b980609e961105529e423e1c5bdf0134ff91","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 01:17:49.000000000","message":"Patch Set 2: Verified-2\n\n(2 comments)\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7cda34f306f3453daaaa0e4b471240c4 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c4753b155214bc993f1e4bcf100bad8 : FAILURE in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7faea4a3eb8449bad27043e5fd37fb6 : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3b9d544b37924627b88a22110b094753 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f10e5f41d2546b093deca3c16b70cb3 : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":2},{"id":"7456196fbc75efdf2c508e05dc87e771cc4e6331","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2020-08-17 08:11:13.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"79b3b2b1b410fdd54d6c78dc3c0954b35e894134","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 08:39:52.000000000","message":"Patch Set 3: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e33052589bc4bdcb7608e70002dc0bd : SUCCESS in 4m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f13332c10c445858884668715a2af66 : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef0dbfd16d164cc1b39509885a49cd19 : FAILURE in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1be4aa914d744e0a8b94bf9cf0c965ff : SUCCESS in 4m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ca65a77d56e64945b2f641c38553cf76 : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a423007c007458a80d3ea747a13429f : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71e21342dc0c430787c0eda4f02d7029 : SUCCESS in 5m 10s\n- karbor-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/b8510151b64e4c00af7961dca8649f07 : FAILURE in 25m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5cfec34f42913648649b06740f82eec26ebe615b","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2020-09-15 09:00:19.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a8ef48fef52c4a03bb156d398937500de41e3eb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 09:33:31.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0bce7afbc57b4ab38f07e788ff666038 : FAILURE in 2m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/334e42cba1e04697ad6b6a432584a84c : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47835d19c716484c84f3614c6bffeda5 : FAILURE in 2m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0278e2042dd41b488ae9b9bf243ca3e : FAILURE in 2m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/36f201eaa13a4bec8422acb81618b7c2 : POST_FAILURE in 2m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/585a6baea67942d0b7a6402e3f664fb4 : POST_FAILURE in 4m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a40723e95f954f2695b6e29610554b47 : FAILURE in 2m 16s\n- karbor-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/f241fd31486042b381bd8c908dc00d9b : FAILURE in 28m 11s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5379f4e258dcc45cc95f90eedb8da3f9e88799fe","author":{"_account_id":29729,"name":"wangyu","email":"wangyu_yewu@cmss.chinamobile.com","username":"wangyu"},"date":"2020-09-16 02:39:36.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"96228172ccff65db30e0377fe938b838c3c9f26b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 03:50:27.000000000","message":"Patch Set 4:\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7491d998e6b443a7abce96f0785f94eb : SUCCESS in 6m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a759f3f6ee2b4a05a327a6a553daccca : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bda2f1a41074bc690e2f9202a716351 : FAILURE in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b20afe2b140b4c94b8ab9f62dc1f9fdc : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ee1bedda7cf4f459861d83adc8bf57f : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe07e07c810e46909d289d8b896117f4 : SUCCESS in 4m 26s\n- karbor-fullstack https://zuul.opendev.org/t/openstack/build/7ed6d80c4c9d477fab7fb94ee91dd68f : POST_FAILURE in 33m 28s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"33688e2c35dee016997dd5c4691c29cbbc0f986f","author":{"_account_id":29729,"name":"wangyu","email":"wangyu_yewu@cmss.chinamobile.com","username":"wangyu"},"date":"2020-09-16 06:33:59.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"47ccb93cbaf8c9053587beff6b4ca923c2a12e8d","author":{"_account_id":29729,"name":"wangyu","email":"wangyu_yewu@cmss.chinamobile.com","username":"wangyu"},"date":"2020-09-16 06:37:31.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f086459227963dc2515a6861b003d41eb46d8002","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 08:12:49.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3d57cb9878a4b0b99921ea37d7bf960 : SUCCESS in 8m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd3f6d16dd52462cada22099cba10a22 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/278daa1b7771449393a45d1f555b989c : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d3e3ca8f86544813a89176cedfc82985 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e11e8c5c85446989c9a959a6cca649b : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/951b00d6658548e8808646a0e4749bc8 : SUCCESS in 5m 02s\n- karbor-fullstack https://zuul.opendev.org/t/openstack/build/e2aedba071954361bd943a8f289e2424 : POST_FAILURE in 1h 23m 57s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"714996543c56802ef21411d5b222d0488da80899","author":{"_account_id":29729,"name":"wangyu","email":"wangyu_yewu@cmss.chinamobile.com","username":"wangyu"},"date":"2020-09-16 10:52:55.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"15fa9d09f721867ee2e1aefd107b0c4d0be5a50b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 12:22:15.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b21c32083658438aa5dff79025f2aa54 : SUCCESS in 5m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c224862b45c4a73adbd96f1888683af : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/680f0d02b83d432f99bc47b483673b01 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/030641b1b5ff4e49b880e8360f648511 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d428630f3ce4e988438035f820e2dc6 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/117ed8ce9cad4c47a9de85d3f893b3a4 : SUCCESS in 6m 09s\n- karbor-fullstack https://zuul.opendev.org/t/openstack/build/c91254c1fcea4ab6bfa90309c231a19c : POST_FAILURE in 1h 24m 24s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1a0a77b2d45e1c5bb41673082b4b28f96ac55f47","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-06-09 22:49:54.000000000","message":"Abandoned\n\nkarbor is retired now http://lists.openstack.org/pipermail/openstack-discuss/2020-November/018643.html","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"795d70ade650e8156b3e7c465f89580d4b125478","revisions":{"b9da15b5a08366613a10a26105cf136c34d5b5d3":{"kind":"REWORK","_number":1,"created":"2019-11-27 00:25:49.000000000","uploader":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"ref":"refs/changes/14/696214/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor","ref":"refs/changes/14/696214/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor refs/changes/14/696214/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor refs/changes/14/696214/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor refs/changes/14/696214/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor refs/changes/14/696214/1"}}},"commit":{"parents":[{"commit":"82f07c14b6a1c5a9303e37001e85cc30df796dcc","subject":"Use trustee project_name instead of fixed value","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/82f07c14b6a1c5a9303e37001e85cc30df796dcc"}]}],"author":{"name":"Jiao Pengju","email":"jiaopengju@cmss.chinamobile.com","date":"2019-11-27 00:25:49.000000000","tz":0},"committer":{"name":"Jiao Pengju","email":"jiaopengju@cmss.chinamobile.com","date":"2019-11-27 00:25:49.000000000","tz":0},"subject":"Revert \"Use trustee project_name instead of fixed value\"","message":"Revert \"Use trustee project_name instead of fixed value\"\n\nThis patch will break the fullstack test, so revert it.\nThis reverts commit 82f07c14b6a1c5a9303e37001e85cc30df796dcc.\n\nChange-Id: I4a8ada268c295505c6159b4eaf71fe77fb2b56eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/b9da15b5a08366613a10a26105cf136c34d5b5d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/b9da15b5a08366613a10a26105cf136c34d5b5d3"}]},"branch":"refs/heads/master"},"aadf18ff2b488cbd6a107245866eeeae04c88a77":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-12-02 11:32:29.000000000","uploader":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"ref":"refs/changes/14/696214/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor","ref":"refs/changes/14/696214/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor refs/changes/14/696214/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor refs/changes/14/696214/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor refs/changes/14/696214/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor refs/changes/14/696214/2"}}},"commit":{"parents":[{"commit":"95c1a0e7e22c83f07eadd9c052651933038ccc08","subject":"Fix unit test error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/95c1a0e7e22c83f07eadd9c052651933038ccc08"}]}],"author":{"name":"Jiao Pengju","email":"jiaopengju@cmss.chinamobile.com","date":"2019-11-27 00:25:49.000000000","tz":0},"committer":{"name":"Jiao Pengju","email":"jiaopengju@cmss.chinamobile.com","date":"2019-12-02 11:32:29.000000000","tz":0},"subject":"Revert \"Use trustee project_name instead of fixed value\"","message":"Revert \"Use trustee project_name instead of fixed value\"\n\nThis patch will break the fullstack test, so revert it.\nThis reverts commit 82f07c14b6a1c5a9303e37001e85cc30df796dcc.\n\nChange-Id: I4a8ada268c295505c6159b4eaf71fe77fb2b56eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/aadf18ff2b488cbd6a107245866eeeae04c88a77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/aadf18ff2b488cbd6a107245866eeeae04c88a77"}]},"branch":"refs/heads/master"},"5e8c510d1fb94af743ec3b8f9e3493e6231a0d7f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-08-17 08:11:13.000000000","uploader":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"ref":"refs/changes/14/696214/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor","ref":"refs/changes/14/696214/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor refs/changes/14/696214/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor refs/changes/14/696214/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor refs/changes/14/696214/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor refs/changes/14/696214/3"}}},"commit":{"parents":[{"commit":"ee26961cfcf8e0abb221b7cdea2a8c79e0e28782","subject":"Cleanup py27 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/ee26961cfcf8e0abb221b7cdea2a8c79e0e28782"}]}],"author":{"name":"Jiao Pengju","email":"jiaopengju@cmss.chinamobile.com","date":"2019-11-27 00:25:49.000000000","tz":0},"committer":{"name":"jiaopengju","email":"jiaopengju@cmss.chinamobile.com","date":"2020-08-17 08:10:37.000000000","tz":480},"subject":"Revert \"Use trustee project_name instead of fixed value\"","message":"Revert \"Use trustee project_name instead of fixed value\"\n\nThis patch will break the fullstack test, so revert it.\nThis reverts commit 82f07c14b6a1c5a9303e37001e85cc30df796dcc.\n\nChange-Id: I4a8ada268c295505c6159b4eaf71fe77fb2b56eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/5e8c510d1fb94af743ec3b8f9e3493e6231a0d7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/5e8c510d1fb94af743ec3b8f9e3493e6231a0d7f"}]},"branch":"refs/heads/master"},"691532553bab37a42878005f43a1228b2505e4ca":{"kind":"REWORK","_number":4,"created":"2020-09-15 09:00:19.000000000","uploader":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"ref":"refs/changes/14/696214/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor","ref":"refs/changes/14/696214/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor refs/changes/14/696214/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor refs/changes/14/696214/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor refs/changes/14/696214/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor refs/changes/14/696214/4"}}},"commit":{"parents":[{"commit":"3df667582c20368a24a885bf4491f34d85ec194e","subject":"Use unittest.mock instead of third party mock","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/3df667582c20368a24a885bf4491f34d85ec194e"}]}],"author":{"name":"Jiao Pengju","email":"jiaopengju@cmss.chinamobile.com","date":"2019-11-27 00:25:49.000000000","tz":0},"committer":{"name":"jiaopengju","email":"jiaopengju@cmss.chinamobile.com","date":"2020-09-15 08:59:18.000000000","tz":480},"subject":"Revert \"Use trustee project_name instead of fixed value\"","message":"Revert \"Use trustee project_name instead of fixed value\"\n\nThis patch will break the fullstack test, so revert it.\nThis reverts commit 82f07c14b6a1c5a9303e37001e85cc30df796dcc.\n\nChange-Id: I4a8ada268c295505c6159b4eaf71fe77fb2b56eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/691532553bab37a42878005f43a1228b2505e4ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/691532553bab37a42878005f43a1228b2505e4ca"}]},"branch":"refs/heads/master"},"0cbfed7299de56d6281bd83fc238cd33626b6d8d":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-09-16 06:33:59.000000000","uploader":{"_account_id":29729,"name":"wangyu","email":"wangyu_yewu@cmss.chinamobile.com","username":"wangyu"},"ref":"refs/changes/14/696214/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor","ref":"refs/changes/14/696214/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor refs/changes/14/696214/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor refs/changes/14/696214/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor refs/changes/14/696214/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor refs/changes/14/696214/5"}}},"commit":{"parents":[{"commit":"fcca3935af26a8b1b78eabfdd345840a3a2dd8c0","subject":"Merge \"Switch legacy Zuul jobs to native Zuul v3 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/fcca3935af26a8b1b78eabfdd345840a3a2dd8c0"}]}],"author":{"name":"Jiao Pengju","email":"jiaopengju@cmss.chinamobile.com","date":"2019-11-27 00:25:49.000000000","tz":0},"committer":{"name":"wangyu","email":"wangyu_yewu@cmss.chinamobile.com","date":"2020-09-16 06:33:59.000000000","tz":0},"subject":"Revert \"Use trustee project_name instead of fixed value\"","message":"Revert \"Use trustee project_name instead of fixed value\"\n\nThis patch will break the fullstack test, so revert it.\nThis reverts commit 82f07c14b6a1c5a9303e37001e85cc30df796dcc.\n\nChange-Id: I4a8ada268c295505c6159b4eaf71fe77fb2b56eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/0cbfed7299de56d6281bd83fc238cd33626b6d8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/0cbfed7299de56d6281bd83fc238cd33626b6d8d"}]},"branch":"refs/heads/master"},"795d70ade650e8156b3e7c465f89580d4b125478":{"kind":"REWORK","_number":6,"created":"2020-09-16 06:37:31.000000000","uploader":{"_account_id":29729,"name":"wangyu","email":"wangyu_yewu@cmss.chinamobile.com","username":"wangyu"},"ref":"refs/changes/14/696214/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor","ref":"refs/changes/14/696214/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor refs/changes/14/696214/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor refs/changes/14/696214/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor refs/changes/14/696214/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor refs/changes/14/696214/6"}}},"commit":{"parents":[{"commit":"fcca3935af26a8b1b78eabfdd345840a3a2dd8c0","subject":"Merge \"Switch legacy Zuul jobs to native Zuul v3 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/fcca3935af26a8b1b78eabfdd345840a3a2dd8c0"}]}],"author":{"name":"Jiao Pengju","email":"jiaopengju@cmss.chinamobile.com","date":"2019-11-27 00:25:49.000000000","tz":0},"committer":{"name":"wangyu","email":"wangyu_yewu@cmss.chinamobile.com","date":"2020-09-16 06:37:58.000000000","tz":480},"subject":"Revert \"Use trustee project_name instead of fixed value\"","message":"Revert \"Use trustee project_name instead of fixed value\"\n\nThis patch will break the fullstack test, so revert it.\nThis reverts commit 82f07c14b6a1c5a9303e37001e85cc30df796dcc.\n\nChange-Id: I4a8ada268c295505c6159b4eaf71fe77fb2b56eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/795d70ade650e8156b3e7c465f89580d4b125478"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/795d70ade650e8156b3e7c465f89580d4b125478"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
