)]}'
{"id":"openstack%2Fkarbor~700297","triplet_id":"openstack%2Fkarbor~master~I333852356cbe61a4209fde06b10c73def3e5100c","project":"openstack/karbor","branch":"master","topic":"zanata/translations","hashtags":[],"change_id":"I333852356cbe61a4209fde06b10c73def3e5100c","subject":"Imported Translations from Zanata","status":"MERGED","created":"2019-12-22 06:43:18.000000000","updated":"2020-04-08 00:52:01.000000000","submitted":"2020-04-08 00:49:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700297-1586306950670-9e742fff","meta_rev_id":"be7e011eeac5c25abdcf4bd5ff7964cdfeb9b154","_number":700297,"virtual_id_number":700297,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28208,"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","username":"gorshunovr"},{"value":0,"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-08 00:49:10.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"recommended":{"_account_id":28208,"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","username":"gorshunovr"},"all":[{"value":1,"date":"2020-04-07 09:18:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28208,"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","username":"gorshunovr"},{"value":2,"date":"2020-04-08 00:41:10.000000000","_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"all":[{"value":0,"_account_id":28208,"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","username":"gorshunovr"},{"value":1,"date":"2020-04-08 00:41:10.000000000","_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28208,"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","username":"gorshunovr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-07 09:18:42.000000000","updated_by":{"_account_id":28208,"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","username":"gorshunovr"},"reviewer":{"_account_id":28208,"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","username":"gorshunovr"},"state":"REVIEWER"},{"updated":"2020-04-08 00:41:10.000000000","updated_by":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"reviewer":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"state":"REVIEWER"},{"updated":"2020-04-08 00:49:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8e750df43222fa46960bd23553ca90724c33c854","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-22 06:43:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b6aa44a850f486ace99d471f8595119ed2a48e1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-22 07:45:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c20d33d6467243079bcc76e403b556b9 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c89e3c65ac0747b09ea32f1f31cea5db : SUCCESS in 5m 02s\n- karbor-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/b7a3705906f94bd7813d66860eab63b6 : FAILURE in 55m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f4dccf01ce3693689a70dc60f3b961d504d65bc1","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-06 10:01:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e026a51aefaed118ddbca1dbc61109a17637e139","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 10:59:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d40774dfaddc4de9ba0cc8636d8ff4a4 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/555788b5d48f4764be0d25912eb1913d : SUCCESS in 5m 09s\n- karbor-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/e08d55db11804138b063a4d4922261d1 : FAILURE in 54m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"186de48dc561fcda2e8d0dfa88c5439d877eaf6d","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-03-20 07:31:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"68cc2b9d923fcb42e958c467cd597322843ef8f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 08:21:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/703c92d23c7148f4b12d1deebed81262 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc49248807d149a0a79a8728b21fadfa : SUCCESS in 5m 14s\n- karbor-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/9df3e5422e3045adbd70c2b62d493c5c : FAILURE in 47m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"75c52e78e86e138654067b901b7edbecac40323a","author":{"_account_id":28208,"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","username":"gorshunovr"},"date":"2020-04-07 09:18:42.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"220f782a01e9e7cbcec8b8dfbd88bbf335a690dd","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2020-04-08 00:41:10.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d98dad1fc9650ae4f514f273afdf12ce50d23a8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 00:41:22.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4eb6de766613bee883aa6047991102f13485f464","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 00:49:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"8ec3136ae32d3a6816dd36e8060bc661fbbc0464","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 00:49:10.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce3ccfb8fa6948ae90922ed0bee80cf4 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c3f99bddc5544548d6237e6393ef547 : SUCCESS in 7m 01s\n\nWarning:\n  Comments left for invalid file karbor/services/protection/bank_plugin.py\n  Comments left for invalid file karbor/api/common.py","accounts_in_message":[],"_revision_number":3},{"id":"41b52fc856f794c6610e75e9a11d05e2a85b77d9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 00:52:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c13fa204f2b400083d0e62c06da6bb2 : SUCCESS in 2m 21s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9c55d2046ef8e30a7f389cff4aeff43058c061ff","revisions":{"8f81ca661bf6a8b7ea45f34fdf634553af8800d8":{"kind":"REWORK","_number":1,"created":"2019-12-22 06:43:18.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/97/700297/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor","ref":"refs/changes/97/700297/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor refs/changes/97/700297/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor refs/changes/97/700297/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor refs/changes/97/700297/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor refs/changes/97/700297/1"}}},"commit":{"parents":[{"commit":"95c1a0e7e22c83f07eadd9c052651933038ccc08","subject":"Fix unit test error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/95c1a0e7e22c83f07eadd9c052651933038ccc08"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-22 06:43:16.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-22 06:43:16.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I333852356cbe61a4209fde06b10c73def3e5100c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/8f81ca661bf6a8b7ea45f34fdf634553af8800d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/8f81ca661bf6a8b7ea45f34fdf634553af8800d8"}]},"branch":"refs/heads/master"},"01391a19f577c975bf87cc20c5651e794a1f9146":{"kind":"REWORK","_number":2,"created":"2020-02-06 10:01:36.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/97/700297/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor","ref":"refs/changes/97/700297/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor refs/changes/97/700297/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor refs/changes/97/700297/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor refs/changes/97/700297/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor refs/changes/97/700297/2"}}},"commit":{"parents":[{"commit":"401c9bca56af62b5858b05dcaea9a065348aa58a","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/401c9bca56af62b5858b05dcaea9a065348aa58a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-06 10:01:34.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-06 10:01:34.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I333852356cbe61a4209fde06b10c73def3e5100c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/01391a19f577c975bf87cc20c5651e794a1f9146"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/01391a19f577c975bf87cc20c5651e794a1f9146"}]},"branch":"refs/heads/master"},"9c55d2046ef8e30a7f389cff4aeff43058c061ff":{"kind":"REWORK","_number":3,"created":"2020-03-20 07:31:11.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/97/700297/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor","ref":"refs/changes/97/700297/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor refs/changes/97/700297/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor refs/changes/97/700297/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor refs/changes/97/700297/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor refs/changes/97/700297/3"}}},"commit":{"parents":[{"commit":"7fc2b1055753b283d8bda21cb85adf560e58c476","subject":"Remove unsupported calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/7fc2b1055753b283d8bda21cb85adf560e58c476"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-20 07:31:07.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-20 07:31:07.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I333852356cbe61a4209fde06b10c73def3e5100c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/9c55d2046ef8e30a7f389cff4aeff43058c061ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/9c55d2046ef8e30a7f389cff4aeff43058c061ff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
