)]}'
{"id":"openstack%2Fkarbor~702318","triplet_id":"openstack%2Fkarbor~master~I1091ddc8a36244132883932a3a21dea5274eb34a","project":"openstack/karbor","branch":"master","topic":"bug/1859433","hashtags":[],"change_id":"I1091ddc8a36244132883932a3a21dea5274eb34a","subject":"Preserve request id in Karbor logs","status":"MERGED","created":"2020-01-14 01:06:05.000000000","updated":"2020-04-08 00:57:36.000000000","submitted":"2020-04-08 00:54:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":991,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702318-1586307261335-0a348a98","meta_rev_id":"f55664120fcb3a0ce4afbb67b0fe3562ecd84c14","_number":702318,"virtual_id_number":702318,"owner":{"_account_id":29729,"name":"wangyu","email":"wangyu_yewu@cmss.chinamobile.com","username":"wangyu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-03-26 00:38:39.000000000","_account_id":29729,"name":"wangyu","email":"wangyu_yewu@cmss.chinamobile.com","username":"wangyu"},{"value":0,"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"value":0,"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-08 00:54:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"all":[{"value":0,"_account_id":29729,"name":"wangyu","email":"wangyu_yewu@cmss.chinamobile.com","username":"wangyu"},{"value":2,"date":"2020-04-08 00:42:12.000000000","_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"value":0,"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"all":[{"value":0,"_account_id":29729,"name":"wangyu","email":"wangyu_yewu@cmss.chinamobile.com","username":"wangyu"},{"value":1,"date":"2020-04-08 00:42:12.000000000","_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"value":0,"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},{"_account_id":29729,"name":"wangyu","email":"wangyu_yewu@cmss.chinamobile.com","username":"wangyu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-14 05:34:16.000000000","updated_by":{"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},"reviewer":{"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},"state":"REVIEWER"},{"updated":"2020-04-08 00:42:12.000000000","updated_by":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"reviewer":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"state":"REVIEWER"},{"updated":"2020-04-08 00:54:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fa275adaaa32aca4eb58504f0b2a826671358f7c","author":{"_account_id":29729,"name":"wangyu","email":"wangyu_yewu@cmss.chinamobile.com","username":"wangyu"},"date":"2020-01-14 01:06:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"838c58562f38e4fe20df2733bcf8de46f575b134","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 02:20:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7c76bf2cf0749d3ae5ab7054b87458b : SUCCESS in 6m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cf8900bf8ba44978bac2195703b87973 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c66a7e3903947f498a68345642ba5e4 : SUCCESS in 5m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cc4260ce1316419f9910e4e74357b0ff : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9262f7788d1d425084ed7733639b8e9c : SUCCESS in 3m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/544678c815f64309b55f581ef9e4d6cc : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a5a24f8ee8549e2aa60c4aa9f4174e5 : SUCCESS in 6m 38s\n- karbor-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/ca3561c2241644f6bbf443597cc126e0 : FAILURE in 52m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7e7d48b49e730f6d41d4189241c2910f47d78cfb","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2020-03-19 09:44:59.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"dc366fa7cea2a6dddfd1f3c59f06f7a98aa5ef42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 10:45:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb042dcc6a3b404f9a3a58c77e284375 : FAILURE in 6m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed8ffb18f2f74f7f83b1f994654644ff : FAILURE in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c75c8a9399f8492588da254a26f531b3 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15faae547c9446f5a69e7b87868f9b61 : FAILURE in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/638eb959da414618bce6722d8fd9eebf : FAILURE in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5fb3ca91913d4cda9538630188db0f2a : FAILURE in 5m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e18d1dabaf54e1c99e7cbd4e10c01ad : SUCCESS in 7m 32s\n- karbor-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/7ee691f67dfe49d8bc0b67d17078f740 : FAILURE in 54m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a06d1d9632ef566f4f91432b274b3f152f9a1610","author":{"_account_id":29729,"name":"wangyu","email":"wangyu_yewu@cmss.chinamobile.com","username":"wangyu"},"date":"2020-03-23 02:06:55.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"1dee622b69e80feaaf3d4d2d576dbaa1c24aa38f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 02:55:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1dd4fde886a84d30800414af1c385fa0 : SUCCESS in 5m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d5eb6dbc22d042a3a08b6e3fb26f2061 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/067dcbe4b7764995a2aba7a25a5af659 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f841930ae0f246a8bd19d07413a64c26 : SUCCESS in 4m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/14896208c25347eeaf5580e9aacecea2 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7a33143de784db69e2be0806b864b3d : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de9773ed79f444a79068f8aa6512517f : SUCCESS in 5m 01s\n- karbor-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/f5247615b32f4b84ae18cd93e275b173 : FAILURE in 47m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"afb4aaf88d3f512785949c51b89b6389de639208","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2020-03-25 12:04:43.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f752f65694631081db2cd99ecc464b398ad816fa","author":{"_account_id":29729,"name":"wangyu","email":"wangyu_yewu@cmss.chinamobile.com","username":"wangyu"},"date":"2020-03-26 00:29:21.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3096388ed8653db64e3ad7f70b7530730708b1e6","author":{"_account_id":29729,"name":"wangyu","email":"wangyu_yewu@cmss.chinamobile.com","username":"wangyu"},"date":"2020-03-26 00:38:39.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d9b64c38e05223078c792ce73361867f01c1227b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 01:21:44.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a98103d326f4ad3aecb8b6754b6a509 : SUCCESS in 5m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba08e60124fa48bfabfdcdcac5b33291 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/469cb07f2079468393d01c5158b38899 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a3a4ae817573452da3913b23154c6ce7 : SUCCESS in 5m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ff346b358aa749b6a58c76e62ffd37c6 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f166d51ba5841849e97b8e11d55d946 : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7e82b7ac0394d44901d244209b38a4e : SUCCESS in 5m 22s\n- karbor-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/7c692c5cf3fc4754a7a4698a272910a5 : FAILURE in 51m 22s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"66a1d89eca7160e238ebd63b536a1150ca6ef44c","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2020-04-08 00:42:12.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"1f7aa7fb49ce0b8f0331c8a1e577fb0892baff18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 00:42:24.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"409d45a9a1d71a2290e8da0ce3dfd6757b6db3b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 00:54:20.000000000","message":"Patch Set 4: Verified+2\n\n(989 comments)\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c3682e909e7747f099e053de308f5f28 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ec88a83c5644784850c27a0ae0e5c62 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5581ca329524c7a9520636dc0653a6c : SUCCESS in 5m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fed0aaf839874211ada2504455eb4a65 : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3248e4ee400a46d9abd548d3b11aa386 : SUCCESS in 7m 37s\n\nWarning:\n  Comments left for invalid file karbor/api/common.py\n  Comments left for invalid file karbor/tests/unit/db/test_models.py\n  Comments left for invalid file karbor/services/protection/bank_plugin.py\n  Comments left for invalid file karbor/common/notification.py\n  Comments left for invalid file karbor/db/sqlalchemy/api.py","accounts_in_message":[],"_revision_number":4},{"id":"e75aed61f6179f5ad7dffe9f31cf7543a22a2c1a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 00:54:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"2db45d7813ad5475c7f008dc3b6d99c57efb044e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 00:57:36.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e366d045ddf48cfb76b2a49eaab5bee : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"435dc3c423b4159428e8d2b84ea3ee8e37277c58","revisions":{"17ecea5428edcf28f64ffb7b78630f255ae53c37":{"kind":"REWORK","_number":1,"created":"2020-01-14 01:06:05.000000000","uploader":{"_account_id":29729,"name":"wangyu","email":"wangyu_yewu@cmss.chinamobile.com","username":"wangyu"},"ref":"refs/changes/18/702318/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor","ref":"refs/changes/18/702318/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor refs/changes/18/702318/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor refs/changes/18/702318/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor refs/changes/18/702318/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor refs/changes/18/702318/1"}}},"commit":{"parents":[{"commit":"95c1a0e7e22c83f07eadd9c052651933038ccc08","subject":"Fix unit test error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/95c1a0e7e22c83f07eadd9c052651933038ccc08"}]}],"author":{"name":"wangyu","email":"wangyu@cmss.chinamobile.com","date":"2020-01-14 01:05:56.000000000","tz":480},"committer":{"name":"wangyu","email":"wangyu@cmss.chinamobile.com","date":"2020-01-14 01:05:56.000000000","tz":480},"subject":"Preserve request id in Karbor logs","message":"Preserve request id in Karbor logs\n\nWhen the admin context is retrieved the user context and its request\nID is lost and all subsequent log entries have different request IDs.\n\nThe fix is to pass the overwrite parameter in Karbor\u0027s RequestContext\n__init__ method to the parent oslo class.\n\nChange-Id: I1091ddc8a36244132883932a3a21dea5274eb34a\nPartial-Bug: #1859433\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/17ecea5428edcf28f64ffb7b78630f255ae53c37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/17ecea5428edcf28f64ffb7b78630f255ae53c37"}]},"branch":"refs/heads/master"},"ab566a05151aa44c159711582a20a7325224f82e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-03-19 09:44:59.000000000","uploader":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"ref":"refs/changes/18/702318/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor","ref":"refs/changes/18/702318/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor refs/changes/18/702318/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor refs/changes/18/702318/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor refs/changes/18/702318/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor refs/changes/18/702318/2"}}},"commit":{"parents":[{"commit":"401c9bca56af62b5858b05dcaea9a065348aa58a","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/401c9bca56af62b5858b05dcaea9a065348aa58a"}]}],"author":{"name":"wangyu","email":"wangyu@cmss.chinamobile.com","date":"2020-01-14 01:05:56.000000000","tz":480},"committer":{"name":"Jiao Pengju","email":"jiaopengju@cmss.chinamobile.com","date":"2020-03-19 09:44:59.000000000","tz":0},"subject":"Preserve request id in Karbor logs","message":"Preserve request id in Karbor logs\n\nWhen the admin context is retrieved the user context and its request\nID is lost and all subsequent log entries have different request IDs.\n\nThe fix is to pass the overwrite parameter in Karbor\u0027s RequestContext\n__init__ method to the parent oslo class.\n\nChange-Id: I1091ddc8a36244132883932a3a21dea5274eb34a\nPartial-Bug: #1859433\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/ab566a05151aa44c159711582a20a7325224f82e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/ab566a05151aa44c159711582a20a7325224f82e"}]},"branch":"refs/heads/master"},"8f5a821a3915d6e2d56c26f80d587c02e8e95ab4":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-03-23 02:06:55.000000000","uploader":{"_account_id":29729,"name":"wangyu","email":"wangyu_yewu@cmss.chinamobile.com","username":"wangyu"},"ref":"refs/changes/18/702318/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor","ref":"refs/changes/18/702318/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor refs/changes/18/702318/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor refs/changes/18/702318/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor refs/changes/18/702318/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor refs/changes/18/702318/3"}}},"commit":{"parents":[{"commit":"7fc2b1055753b283d8bda21cb85adf560e58c476","subject":"Remove unsupported calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/7fc2b1055753b283d8bda21cb85adf560e58c476"}]}],"author":{"name":"wangyu","email":"wangyu@cmss.chinamobile.com","date":"2020-01-14 01:05:56.000000000","tz":480},"committer":{"name":"wangyu","email":"wangyu@cmss.chinamobile.com","date":"2020-03-23 02:06:55.000000000","tz":0},"subject":"Preserve request id in Karbor logs","message":"Preserve request id in Karbor logs\n\nWhen the admin context is retrieved the user context and its request\nID is lost and all subsequent log entries have different request IDs.\n\nThe fix is to pass the overwrite parameter in Karbor\u0027s RequestContext\n__init__ method to the parent oslo class.\n\nChange-Id: I1091ddc8a36244132883932a3a21dea5274eb34a\nPartial-Bug: #1859433\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/8f5a821a3915d6e2d56c26f80d587c02e8e95ab4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/8f5a821a3915d6e2d56c26f80d587c02e8e95ab4"}]},"branch":"refs/heads/master"},"435dc3c423b4159428e8d2b84ea3ee8e37277c58":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":4,"created":"2020-03-26 00:29:21.000000000","uploader":{"_account_id":29729,"name":"wangyu","email":"wangyu_yewu@cmss.chinamobile.com","username":"wangyu"},"ref":"refs/changes/18/702318/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor","ref":"refs/changes/18/702318/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor refs/changes/18/702318/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor refs/changes/18/702318/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor refs/changes/18/702318/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor refs/changes/18/702318/4"}}},"commit":{"parents":[{"commit":"95c1a0e7e22c83f07eadd9c052651933038ccc08","subject":"Fix unit test error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/95c1a0e7e22c83f07eadd9c052651933038ccc08"}]}],"author":{"name":"wangyu","email":"wangyu@cmss.chinamobile.com","date":"2020-01-14 01:05:56.000000000","tz":480},"committer":{"name":"wangyu","email":"wangyu@cmss.chinamobile.com","date":"2020-03-26 00:28:00.000000000","tz":480},"subject":"Preserve request id in Karbor logs","message":"Preserve request id in Karbor logs\n\nWhen the admin context is retrieved the user context and its request\nID is lost and all subsequent log entries have different request IDs.\n\nThe fix is to pass the overwrite parameter in Karbor\u0027s RequestContext\n__init__ method to the parent oslo class.\n\nChange-Id: I1091ddc8a36244132883932a3a21dea5274eb34a\nCloses-Bug: #1859433\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/435dc3c423b4159428e8d2b84ea3ee8e37277c58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/435dc3c423b4159428e8d2b84ea3ee8e37277c58"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
