)]}'
{"id":"openstack%2Fkarbor~720924","triplet_id":"openstack%2Fkarbor~master~Ifd0d1b8aa419e39361e76a4f846cb4eeaf0d46ca","project":"openstack/karbor","branch":"master","topic":"unittest.mock","hashtags":[],"change_id":"Ifd0d1b8aa419e39361e76a4f846cb4eeaf0d46ca","subject":"Use unittest.mock instead of third party mock","status":"MERGED","created":"2020-04-18 16:53:44.000000000","updated":"2020-08-17 01:15:59.000000000","submitted":"2020-08-17 01:14:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"720924-1597626873646-de949bf5","meta_rev_id":"51759587041c1ca6c7e74aff766aa8fda0025094","_number":720924,"virtual_id_number":720924,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-17 01:14:33.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"all":[{"value":2,"date":"2020-08-17 01:06:50.000000000","_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"all":[{"value":1,"date":"2020-08-17 01:06:50.000000000","_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-19 13:44:50.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-31 15:47:26.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2020-08-17 01:06:50.000000000","updated_by":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"reviewer":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"state":"REVIEWER"},{"updated":"2020-08-17 01:14:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"84841441d1ae6cde7b896aa3a9d0e309cc7394d4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-18 16:53:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f0308898569b9e79682b967aa91d68b3d63d9359","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-18 18:02:45.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/201492736747418493fb793645ed1bf9 : FAILURE in 6m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/204ccb3968bd409e84ccc9a1916a3b91 : FAILURE in 7m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5acc9e2350a741c9a6dcc61087bd1216 : FAILURE in 7m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f39a9b5dac9940f6b600d55b37f9b399 : FAILURE in 8m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bb1010f33aaa4819a285474548bda930 : FAILURE in 8m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96211354232440a689e8b147df2ae782 : FAILURE in 7m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfe8bae426d84fda9d02640699a28c2f : SUCCESS in 7m 19s\n- karbor-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/f37cb168845546aab2014d2e0e0c9754 : FAILURE in 1h 02m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ff66a14791f45020e1d49812f01bdb74a41e797a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-20 15:05:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d37e63a6743a7d35d996002946464de129fd4b6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 17:08:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/30695fc9a3bf4497b40b20995ee37e8a : SUCCESS in 6m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f42e3ce0b6644dcbb7d12b2e7fc08f3b : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60feed2800b448369e15d784052563ec : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51d8f404d867492597f2e6b77c57b28d : SUCCESS in 5m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0f26f0223c344361bb610576e8a54dd0 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6d598a2b97742a6bc007650b32e2a48 : SUCCESS in 9m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95038a56197e461d98f83471f931ed13 : SUCCESS in 6m 25s\n- karbor-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/382a77d4c6fe411a819e74c9bc31e986 : FAILURE in 51m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d5bcd447318bd5afb12ec52be9b512e5e5712490","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-19 13:44:50.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"0c42dfb8fab99af433834a98ac7d98cc54fe38b9","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-07-31 15:47:26.000000000","message":"Patch Set 2: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ac4d00a658f605a88f1214ced3c2e739b78c84da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 16:29:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4b565ac3a114fa0b0f98e8e7f314eac : FAILURE in 3m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/09fc39071f6c4aae8bee394b0e538aa7 : FAILURE in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/952da89eac274c54a830eeabefc5ef5b : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b1187165e9e4ef8954c4add1bf6f401 : FAILURE in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6e1a51cbe36b4eed9b401c3eb3d7ff25 : FAILURE in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b18bb819422d4ed89c73fb06612b2fe3 : FAILURE in 3m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbf8433607c64d22aa82b864456c2e20 : SUCCESS in 4m 21s\n- karbor-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/3179cb50fe984d50a447fc78cd440d45 : FAILURE in 31m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0cbcfb47605caf2d22cdbd952d34f5fd680f89b9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-13 13:28:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"09f81e1b768211beecba0ec02bd17b13b749afba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 14:12:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/904cb079a65142c4b0165cc9c4241ad4 : SUCCESS in 10m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b82accf35e6545feb72fdee21da66def : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a60e0ca347574e77a98d7bfe7e47ff8e : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ac0c25697d943c9bf0dc30d7335feb0 : SUCCESS in 4m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4c7bf88d665f499dae553ceecc914cfb : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f617eceed2044c1b346a16d7bd7fb17 : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b266077c106245b8bacd17eda4be4109 : SUCCESS in 13m 03s\n- karbor-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/370e157cdbe546cd95f83fdbf5cd9e99 : FAILURE in 34m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8a9490b76d378f67bd1f1206f429e404a327a750","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2020-08-17 01:06:50.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nthanks Sean McGinnis！","accounts_in_message":[],"_revision_number":3},{"id":"c09e85e18db712230b74627e70d9bd823881ecd9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 01:07:01.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5280e650241f0f2416c5930a9f2f44b107629049","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 01:14:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"fb3a06d1f6ffdcd610d026bd62d8ee7bcc4c3697","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 01:14:33.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/93fc50f9198f472b9002d3835e0709e2 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47fe583e66914068b50ef9b80c17411b : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d03df5ec79b4ef2b17677b397af039a : SUCCESS in 4m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/06bdcff971e14862932736c12b5e840c : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd92ed5f70224d6cb33a0fe2d5597381 : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":3},{"id":"1fde8b83fb5fa4828707ad8271ed47b70231277f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 01:15:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0369370dab2e467bac3dc3e7d4089d25 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3df667582c20368a24a885bf4491f34d85ec194e","revisions":{"6cb3ff9a880dcd33cf4afa50384ac9f04236fd39":{"kind":"REWORK","_number":1,"created":"2020-04-18 16:53:44.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/24/720924/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor","ref":"refs/changes/24/720924/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor refs/changes/24/720924/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor refs/changes/24/720924/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor refs/changes/24/720924/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor refs/changes/24/720924/1"}}},"commit":{"parents":[{"commit":"297700ee2b3d27f5ad7061106e561b0df3642af4","subject":"Merge \"Add unit tests for trigger update API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/297700ee2b3d27f5ad7061106e561b0df3642af4"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-18 16:53:33.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-18 16:53:33.000000000","tz":-300},"subject":"Use unittest.mock instead of third party mock","message":"Use unittest.mock instead of third party mock\n\nNow that we no longer support py27, we can use the standard library\nunittest.mock module instead of the third party mock lib.\n\nChange-Id: Ifd0d1b8aa419e39361e76a4f846cb4eeaf0d46ca\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/6cb3ff9a880dcd33cf4afa50384ac9f04236fd39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/6cb3ff9a880dcd33cf4afa50384ac9f04236fd39"}]},"branch":"refs/heads/master"},"85a16bab320c801f93fb57e1c0c07bfcb71cd2aa":{"kind":"REWORK","_number":2,"created":"2020-04-20 15:05:18.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/24/720924/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor","ref":"refs/changes/24/720924/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor refs/changes/24/720924/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor refs/changes/24/720924/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor refs/changes/24/720924/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor refs/changes/24/720924/2"}}},"commit":{"parents":[{"commit":"297700ee2b3d27f5ad7061106e561b0df3642af4","subject":"Merge \"Add unit tests for trigger update API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/297700ee2b3d27f5ad7061106e561b0df3642af4"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-18 16:53:33.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-20 15:05:06.000000000","tz":-300},"subject":"Use unittest.mock instead of third party mock","message":"Use unittest.mock instead of third party mock\n\nNow that we no longer support py27, we can use the standard library\nunittest.mock module instead of the third party mock lib.\n\nChange-Id: Ifd0d1b8aa419e39361e76a4f846cb4eeaf0d46ca\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/85a16bab320c801f93fb57e1c0c07bfcb71cd2aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/85a16bab320c801f93fb57e1c0c07bfcb71cd2aa"}]},"branch":"refs/heads/master"},"3df667582c20368a24a885bf4491f34d85ec194e":{"kind":"REWORK","_number":3,"created":"2020-08-13 13:28:14.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/24/720924/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor","ref":"refs/changes/24/720924/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor refs/changes/24/720924/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor refs/changes/24/720924/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor refs/changes/24/720924/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor refs/changes/24/720924/3"}}},"commit":{"parents":[{"commit":"0764f5dab12034efdc41d35ccd627f2a6618c459","subject":"Update hacking for Python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/0764f5dab12034efdc41d35ccd627f2a6618c459"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-18 16:53:33.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-08-13 13:28:07.000000000","tz":-300},"subject":"Use unittest.mock instead of third party mock","message":"Use unittest.mock instead of third party mock\n\nNow that we no longer support py27, we can use the standard library\nunittest.mock module instead of the third party mock lib.\n\nChange-Id: Ifd0d1b8aa419e39361e76a4f846cb4eeaf0d46ca\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/3df667582c20368a24a885bf4491f34d85ec194e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/3df667582c20368a24a885bf4491f34d85ec194e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
