)]}'
{"id":"openstack%2Fkarbor~729145","triplet_id":"openstack%2Fkarbor~master~I72f5c6ef6dc1430bf921309bf66e79a2554335e0","project":"openstack/karbor","branch":"master","topic":"rm-testrconf","hashtags":[],"change_id":"I72f5c6ef6dc1430bf921309bf66e79a2554335e0","subject":"Remove .testr.conf","status":"ABANDONED","created":"2020-05-19 07:09:47.000000000","updated":"2021-06-09 22:50:24.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ae869f14d63f7ab8fb8c17f788949c7b45a3332d","_number":729145,"virtual_id_number":729145,"owner":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-05-19 08:08:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":1,"date":"2020-05-25 14:35:58.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-19 08:08:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-25 14:35:58.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"52ec55fef0c5ea90c29ee248b6b7ca7cc2c966ae","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2020-05-19 07:09:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"785c0bd8018835fe4115e810648b5b16711eab79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 08:08:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b322e9c50b74637ab85bee2f30d5504 : SUCCESS in 7m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/885ee2a8e56045228349b74884c14a81 : SUCCESS in 7m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dd83e46f5424be8859f703112e2e68a : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c9782837c844f808db5b136141e4e9f : SUCCESS in 7m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/12d0469da1ee4f388b70b12c78da1519 : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80d97e5aa1014aa18ed4aad7abf99942 : SUCCESS in 10m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/014202f49173412b8f06f1fdd0061b3b : SUCCESS in 9m 12s\n- karbor-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/0715fec818db4e2fb8b15943025d4bf2 : FAILURE in 52m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"21e9169066812a4299418fffd90088d895c5cd03","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-25 14:35:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ae869f14d63f7ab8fb8c17f788949c7b45a3332d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-06-09 22:50:24.000000000","message":"Abandoned\n\nkarbor is retired now http://lists.openstack.org/pipermail/openstack-discuss/2020-November/018643.html","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2c9805ac7d94b3e36405042dc49e2ec4692016a3","revisions":{"2c9805ac7d94b3e36405042dc49e2ec4692016a3":{"kind":"REWORK","_number":1,"created":"2020-05-19 07:09:47.000000000","uploader":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"ref":"refs/changes/45/729145/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor","ref":"refs/changes/45/729145/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor refs/changes/45/729145/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor refs/changes/45/729145/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor refs/changes/45/729145/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor refs/changes/45/729145/1"}}},"commit":{"parents":[{"commit":"ee26961cfcf8e0abb221b7cdea2a8c79e0e28782","subject":"Cleanup py27 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/ee26961cfcf8e0abb221b7cdea2a8c79e0e28782"}]}],"author":{"name":"Masayuki Igawa","email":"masayuki@igawa.io","date":"2020-04-21 06:58:26.000000000","tz":540},"committer":{"name":"Masayuki Igawa","email":"masayuki@igawa.io","date":"2020-05-19 06:47:07.000000000","tz":540},"subject":"Remove .testr.conf","message":"Remove .testr.conf\n\nThis commit removes .testr.conf since this repo already has\n.stestr.conf. We just need .stestr.conf instead of .testr.conf.\n\nChange-Id: I72f5c6ef6dc1430bf921309bf66e79a2554335e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/2c9805ac7d94b3e36405042dc49e2ec4692016a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/2c9805ac7d94b3e36405042dc49e2ec4692016a3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
