)]}'
{"id":"openstack%2Fkarbor~731974","triplet_id":"openstack%2Fkarbor~master~I8544a3024a1769adc3fc534d93f5ecca8720dd3f","project":"openstack/karbor","branch":"master","topic":"reno-openstackdocstheme","hashtags":[],"change_id":"I8544a3024a1769adc3fc534d93f5ecca8720dd3f","subject":"Switch to newer openstackdocstheme and reno versions","status":"ABANDONED","created":"2020-05-30 16:28:35.000000000","updated":"2021-05-18 07:35:06.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"edbda55f1a7fb570db2c83f1353f7a9a8081b5ba","_number":731974,"virtual_id_number":731974,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":19346,"name":"yinwei","email":"yinweiishere@icloud.com","username":"yinwei"},{"_account_id":20883,"name":"Yuval Brik","email":"yuval@brik.org.il","username":"yuvalbr"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":2023,"name":"Saggi Mizrahi","email":"ficoos+osreview@gmail.com","username":"ficoos"},{"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-09-16 07:14:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":13070,"name":"Eran Gampel","email":"eran@gampel.net","username":"Gampel"},{"_account_id":17151,"name":"chenying","email":"chenyingko@gmail.com","username":"chenying"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19346,"name":"yinwei","email":"yinweiishere@icloud.com","username":"yinwei"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20883,"name":"Yuval Brik","email":"yuval@brik.org.il","username":"yuvalbr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":2023,"name":"Saggi Mizrahi","email":"ficoos+osreview@gmail.com","username":"ficoos"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"value":1,"date":"2020-09-16 05:51:22.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13070,"name":"Eran Gampel","email":"eran@gampel.net","username":"Gampel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17151,"name":"chenying","email":"chenyingko@gmail.com","username":"chenying"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":19346,"name":"yinwei","email":"yinweiishere@icloud.com","username":"yinwei"},{"_account_id":20883,"name":"Yuval Brik","email":"yuval@brik.org.il","username":"yuvalbr"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":2023,"name":"Saggi Mizrahi","email":"ficoos+osreview@gmail.com","username":"ficoos"},{"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":13070,"name":"Eran Gampel","email":"eran@gampel.net","username":"Gampel"},{"_account_id":17151,"name":"chenying","email":"chenyingko@gmail.com","username":"chenying"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2023,"name":"Saggi Mizrahi","email":"ficoos+osreview@gmail.com","username":"ficoos"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":13070,"name":"Eran Gampel","email":"eran@gampel.net","username":"Gampel"},{"_account_id":17151,"name":"chenying","email":"chenyingko@gmail.com","username":"chenying"},{"_account_id":19346,"name":"yinwei","email":"yinweiishere@icloud.com","username":"yinwei"},{"_account_id":20883,"name":"Yuval Brik","email":"yuval@brik.org.il","username":"yuvalbr"},{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-02 09:45:22.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-07 12:25:29.000000000","updated_by":{"_account_id":2023,"name":"Saggi Mizrahi","email":"ficoos+osreview@gmail.com","username":"ficoos"},"reviewer":{"_account_id":2023,"name":"Saggi Mizrahi","email":"ficoos+osreview@gmail.com","username":"ficoos"},"state":"REVIEWER"},{"updated":"2020-07-07 12:25:29.000000000","updated_by":{"_account_id":13070,"name":"Eran Gampel","email":"eran@gampel.net","username":"Gampel"},"reviewer":{"_account_id":13070,"name":"Eran Gampel","email":"eran@gampel.net","username":"Gampel"},"state":"REVIEWER"},{"updated":"2020-07-07 12:25:29.000000000","updated_by":{"_account_id":17151,"name":"chenying","email":"chenyingko@gmail.com","username":"chenying"},"reviewer":{"_account_id":17151,"name":"chenying","email":"chenyingko@gmail.com","username":"chenying"},"state":"REVIEWER"},{"updated":"2020-07-07 12:25:29.000000000","updated_by":{"_account_id":19346,"name":"yinwei","email":"yinweiishere@icloud.com","username":"yinwei"},"reviewer":{"_account_id":19346,"name":"yinwei","email":"yinweiishere@icloud.com","username":"yinwei"},"state":"REVIEWER"},{"updated":"2020-07-07 12:25:29.000000000","updated_by":{"_account_id":20883,"name":"Yuval Brik","email":"yuval@brik.org.il","username":"yuvalbr"},"reviewer":{"_account_id":20883,"name":"Yuval Brik","email":"yuval@brik.org.il","username":"yuvalbr"},"state":"REVIEWER"},{"updated":"2020-07-07 12:25:29.000000000","updated_by":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"reviewer":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"state":"REVIEWER"},{"updated":"2020-07-07 12:25:29.000000000","updated_by":{"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},"reviewer":{"_account_id":29272,"name":"liushuai","email":"liushuai@cmss.chinamobile.com","username":"liushuai"},"state":"REVIEWER"},{"updated":"2020-09-16 07:14:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dfa9c80e18f3a16cac990d019e9808ddb32d2c52","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-30 16:28:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b1d4fb475a2784300a7835ee9908f87c5aa6eb69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-30 17:16:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d85b37ae3c1d477397ce6aa9ba6d18f9 : SUCCESS in 3m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8059cc50ed204518b70abf97376d0c59 : SUCCESS in 4m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd96f8ceb0bb4119b151522c3328a009 : SUCCESS in 4m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c092b673575f4264a6124c54f3e21f31 : SUCCESS in 2m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea5cdd3ba5cb4894be9d3211b70b98a9 : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03734e439fc549f5bf2d1ea65457c474 : SUCCESS in 4m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b6d72e39861f40d7bb1f56074fa80344 : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f2c3f70c0064cb9afd5c2dcfc287795 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/787c76c144144a789b766ca253a5880c : SUCCESS in 3m 54s\n- karbor-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/ce1359e3e82447f79d25aad3ef26b605 : FAILURE in 46m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ff60cb2343004e77aa90b9156fbe591fcf860a56","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-02 09:45:22.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e031e2afdf5c756a9304fd7cbe23738e745c640f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-07-07 12:24:52.000000000","message":"Patch Set 1:\n\nSorry, added wrong group","accounts_in_message":[],"_revision_number":1},{"id":"71ff46873b8fb5ae52ec1990bc33cfd9235e4805","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2020-07-22 02:46:33.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"cd54bbccc9a78c2ea901d6b69c56af5443d44228","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 03:26:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/66916b78fed549faaea092eddc39fec6 : SUCCESS in 4m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13ad02bec74e49159d4a291cc24cfc40 : FAILURE in 4m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1792a5289e464b9c93ee1fcacee560fc : FAILURE in 4m 46s\n- requirements-check https://zuul.opendev.org/t/openstack/build/bae66b694a924191a13c5a38d32e757c : SUCCESS in 2m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fec3817353cd4b65b9f9d9a617b17fb6 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13764047a6af461c9f56b3dbed708240 : FAILURE in 5m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/18bf48118aad4e29a754ba551b7bf601 : FAILURE in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/673572875ad645bd898c63ac2c728473 : FAILURE in 4m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e7e125e83924a6fab4969dfed4777a9 : SUCCESS in 5m 27s\n- karbor-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/715c3483082047f2b1acfbe0f50629d5 : FAILURE in 29m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4ae319f2245d70e2fccecb4247b30756c0a3a4ab","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2020-09-16 05:51:22.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"fd69aeb50b2e00e9b599769e47e84c11b1976c12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 07:14:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2dd2a9cf308646a59e50992865de7f6f : SUCCESS in 3m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0dd540fcfc6c4a20886ab08ef50d4f86 : SUCCESS in 4m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d2d3dfd5e45470e884971d8b81a5882 : SUCCESS in 4m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7cd9a9cc152b495ab26e96028498cea6 : SUCCESS in 1m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18b0716859f44cfa8fcff45bc59e6931 : SUCCESS in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b1fa209251a486283db2cc8790230a7 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99b6597e26ed4beb80273747503ef1ae : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aec192cc544849be89d8f44271d5e17b : SUCCESS in 4m 38s\n- karbor-fullstack https://zuul.opendev.org/t/openstack/build/f07be35ad7494e3b9374615ac2a81864 : FAILURE in 1h 21m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"edbda55f1a7fb570db2c83f1353f7a9a8081b5ba","tag":"autogenerated:gerrit:abandon","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-05-18 07:35:06.000000000","message":"Abandoned\n\nI will not push this further.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"866cf4292144e18ebc9d8a4dc54868f2249a002c","revisions":{"596da2ea5404954b0c34aa3c2c4525f853abfb1e":{"kind":"REWORK","_number":1,"created":"2020-05-30 16:28:35.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/74/731974/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor","ref":"refs/changes/74/731974/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor refs/changes/74/731974/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor refs/changes/74/731974/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor refs/changes/74/731974/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor refs/changes/74/731974/1"}}},"commit":{"parents":[{"commit":"ee26961cfcf8e0abb221b7cdea2a8c79e0e28782","subject":"Cleanup py27 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/ee26961cfcf8e0abb221b7cdea2a8c79e0e28782"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-30 16:26:44.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-30 16:27:33.000000000","tz":120},"subject":"Switch to newer openstackdocstheme and reno versions","message":"Switch to newer openstackdocstheme and reno versions\n\nSwitch to openstackdocstheme 2.2.1 and reno 3.1.0 versions. Using\nthese versions will allow especially:\n* Linking from HTML to PDF document\n* Allow parallel building of documents\n* Fix some rendering problems\n\nUpdate Sphinx version as well.\n\nDisable openstackdocs_auto_name to use \u0027project\u0027 variable as name.\n\nChange pygments_style to \u0027native\u0027 since old theme version always used\n\u0027native\u0027 and the theme now respects the setting and using \u0027sphinx\u0027 can\nlead to some strange rendering.\n\nopenstackdocstheme renames some variables, so follow the renames\nbefore the next release removes them. A couple of variables are also\nnot needed anymore, remove them.\n\nSee also\nhttp://lists.openstack.org/pipermail/openstack-discuss/2020-May/014971.html\n\nChange-Id: I8544a3024a1769adc3fc534d93f5ecca8720dd3f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/596da2ea5404954b0c34aa3c2c4525f853abfb1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/596da2ea5404954b0c34aa3c2c4525f853abfb1e"}]},"branch":"refs/heads/master"},"99f7164d251fc63119c51bd13f2fe6350398079d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-07-22 02:46:33.000000000","uploader":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"ref":"refs/changes/74/731974/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor","ref":"refs/changes/74/731974/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor refs/changes/74/731974/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor refs/changes/74/731974/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor refs/changes/74/731974/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor refs/changes/74/731974/2"}}},"commit":{"parents":[{"commit":"0764f5dab12034efdc41d35ccd627f2a6618c459","subject":"Update hacking for Python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/0764f5dab12034efdc41d35ccd627f2a6618c459"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-30 16:26:44.000000000","tz":120},"committer":{"name":"Jiao Pengju","email":"jiaopengju_yewu@cmss.chinamobile.com","date":"2020-07-22 02:46:33.000000000","tz":0},"subject":"Switch to newer openstackdocstheme and reno versions","message":"Switch to newer openstackdocstheme and reno versions\n\nSwitch to openstackdocstheme 2.2.1 and reno 3.1.0 versions. Using\nthese versions will allow especially:\n* Linking from HTML to PDF document\n* Allow parallel building of documents\n* Fix some rendering problems\n\nUpdate Sphinx version as well.\n\nDisable openstackdocs_auto_name to use \u0027project\u0027 variable as name.\n\nChange pygments_style to \u0027native\u0027 since old theme version always used\n\u0027native\u0027 and the theme now respects the setting and using \u0027sphinx\u0027 can\nlead to some strange rendering.\n\nopenstackdocstheme renames some variables, so follow the renames\nbefore the next release removes them. A couple of variables are also\nnot needed anymore, remove them.\n\nSee also\nhttp://lists.openstack.org/pipermail/openstack-discuss/2020-May/014971.html\n\nChange-Id: I8544a3024a1769adc3fc534d93f5ecca8720dd3f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/99f7164d251fc63119c51bd13f2fe6350398079d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/99f7164d251fc63119c51bd13f2fe6350398079d"}]},"branch":"refs/heads/master"},"866cf4292144e18ebc9d8a4dc54868f2249a002c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-09-16 05:51:22.000000000","uploader":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"ref":"refs/changes/74/731974/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor","ref":"refs/changes/74/731974/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor refs/changes/74/731974/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor refs/changes/74/731974/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor refs/changes/74/731974/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor refs/changes/74/731974/3"}}},"commit":{"parents":[{"commit":"fcca3935af26a8b1b78eabfdd345840a3a2dd8c0","subject":"Merge \"Switch legacy Zuul jobs to native Zuul v3 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/fcca3935af26a8b1b78eabfdd345840a3a2dd8c0"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-30 16:26:44.000000000","tz":120},"committer":{"name":"Jiao Pengju","email":"jiaopengju_yewu@cmss.chinamobile.com","date":"2020-09-16 05:51:22.000000000","tz":0},"subject":"Switch to newer openstackdocstheme and reno versions","message":"Switch to newer openstackdocstheme and reno versions\n\nSwitch to openstackdocstheme 2.2.1 and reno 3.1.0 versions. Using\nthese versions will allow especially:\n* Linking from HTML to PDF document\n* Allow parallel building of documents\n* Fix some rendering problems\n\nUpdate Sphinx version as well.\n\nDisable openstackdocs_auto_name to use \u0027project\u0027 variable as name.\n\nChange pygments_style to \u0027native\u0027 since old theme version always used\n\u0027native\u0027 and the theme now respects the setting and using \u0027sphinx\u0027 can\nlead to some strange rendering.\n\nopenstackdocstheme renames some variables, so follow the renames\nbefore the next release removes them. A couple of variables are also\nnot needed anymore, remove them.\n\nSee also\nhttp://lists.openstack.org/pipermail/openstack-discuss/2020-May/014971.html\n\nChange-Id: I8544a3024a1769adc3fc534d93f5ecca8720dd3f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/866cf4292144e18ebc9d8a4dc54868f2249a002c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/866cf4292144e18ebc9d8a4dc54868f2249a002c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
