)]}'
{"id":"openstack%2Fkarbor~734884","triplet_id":"openstack%2Fkarbor~master~I1130da54ea08e4e22fa45070c0341b439d1b2561","project":"openstack/karbor","branch":"master","topic":"gitignore-stestr","hashtags":[],"change_id":"I1130da54ea08e4e22fa45070c0341b439d1b2561","subject":"ignore `.stestr/` dir","status":"ABANDONED","created":"2020-06-10 16:17:29.000000000","updated":"2021-06-09 22:50:21.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3d8035ddaeea29969af810af8b801c94af30e290","_number":734884,"virtual_id_number":734884,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-06-10 17:20:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-10 17:20:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"df41d4fbaf2f4d3b1c6e61c8f1126054b10bc500","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-10 16:17:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c0c4f8907f6473f746c4e8217c14a965029eae64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 17:20:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/759ce934d4044218b96f7a9cd2c3d137 : SUCCESS in 8m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/73736cfd25444dd988ec1f3589e9f681 : SUCCESS in 9m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c303e0f1170843c8a1fcec30005407d0 : SUCCESS in 8m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6354d65b95054917a3fc1c392a244d42 : SUCCESS in 7m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7a1d8ce99c9d4fbcaac073f74b50a3b7 : SUCCESS in 9m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed70f7771a2c413da768187592c27e95 : SUCCESS in 9m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9d38a231f46491cb739014166536da4 : SUCCESS in 8m 03s\n- karbor-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/b91243387c26412e853fa9c2e488a851 : FAILURE in 54m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3d8035ddaeea29969af810af8b801c94af30e290","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-06-09 22:50:21.000000000","message":"Abandoned\n\nkarbor is retired now http://lists.openstack.org/pipermail/openstack-discuss/2020-November/018643.html","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"00501b36e3e037f2d68ed27a75254bfe5c5c65f4","revisions":{"00501b36e3e037f2d68ed27a75254bfe5c5c65f4":{"kind":"REWORK","_number":1,"created":"2020-06-10 16:17:29.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/84/734884/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor","ref":"refs/changes/84/734884/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor refs/changes/84/734884/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor refs/changes/84/734884/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor refs/changes/84/734884/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor refs/changes/84/734884/1"}}},"commit":{"parents":[{"commit":"ee26961cfcf8e0abb221b7cdea2a8c79e0e28782","subject":"Cleanup py27 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/ee26961cfcf8e0abb221b7cdea2a8c79e0e28782"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-10 16:15:59.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-10 16:17:13.000000000","tz":120},"subject":"ignore `.stestr/` dir","message":"ignore `.stestr/` dir\n\nThis is test artifact that don\u0027t need to be versioned and could be ignored.\n\nChange-Id: I1130da54ea08e4e22fa45070c0341b439d1b2561\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/00501b36e3e037f2d68ed27a75254bfe5c5c65f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/00501b36e3e037f2d68ed27a75254bfe5c5c65f4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
