)]}'
{"id":"openstack%2Fkarbor~739160","triplet_id":"openstack%2Fkarbor~master~I4b73cee94eacc77aa866ab984f137120a4250f8e","project":"openstack/karbor","branch":"master","topic":"cleanup","hashtags":[],"change_id":"I4b73cee94eacc77aa866ab984f137120a4250f8e","subject":"Remove obsolete files","status":"ABANDONED","created":"2020-07-03 03:50:21.000000000","updated":"2021-06-09 22:50:14.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f96464fa36fae80f351257a76cd3f981b6717c9a","_number":739160,"virtual_id_number":739160,"owner":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-07-03 04:17:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-03 04:17:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f8a62b86fafd96df63e013946dbdaf5cc89b9341","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2020-07-03 03:50:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c2220a99e224df2ec2d67aeb4420384448b4f389","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-03 04:17:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b8b72b9d0cd4aeb9da83f4c4205fc47 : SUCCESS in 5m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9001466b8d524c75849e43e37e937ceb : SUCCESS in 4m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1360c333bfc8423fbda946df76778c54 : SUCCESS in 2m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9f354f861774c00aecc67131944529b : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a804af3bf61f463d924852b1ebb73946 : SUCCESS in 5m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cc0d029dc3c14a57ab788728583a2de0 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd7a0cd96f094e34ae30f99aebfc3513 : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38055b4b684542369d9e900c2ecfecf2 : SUCCESS in 7m 04s\n- karbor-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/5c750fe238c0427fb5c4f174764d97b7 : FAILURE in 25m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f96464fa36fae80f351257a76cd3f981b6717c9a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-06-09 22:50:14.000000000","message":"Abandoned\n\nkarbor is retired now http://lists.openstack.org/pipermail/openstack-discuss/2020-November/018643.html","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"92bb4088cf1188fbf9ed08380ec5ce88875af45c","revisions":{"92bb4088cf1188fbf9ed08380ec5ce88875af45c":{"kind":"REWORK","_number":1,"created":"2020-07-03 03:50:21.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/60/739160/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor","ref":"refs/changes/60/739160/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor refs/changes/60/739160/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor refs/changes/60/739160/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor refs/changes/60/739160/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor refs/changes/60/739160/1"}}},"commit":{"parents":[{"commit":"0764f5dab12034efdc41d35ccd627f2a6618c459","subject":"Update hacking for Python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/0764f5dab12034efdc41d35ccd627f2a6618c459"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-07-03 03:44:09.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-07-03 03:50:12.000000000","tz":480},"subject":"Remove obsolete files","message":"Remove obsolete files\n\nThe files babel.cfg is not used anymore, remove it.\n\nThe mox3 repo is being retired. Remove mox3 from lower-constraints.txt.\n\nTo see: [1] I0f69963aabe00b5ebfcc3072c3ba9fd033d73920; [2] If897d98c15261f85eca5ddf074ac6cd2182ac1ad\n\nChange-Id: I4b73cee94eacc77aa866ab984f137120a4250f8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/92bb4088cf1188fbf9ed08380ec5ce88875af45c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/92bb4088cf1188fbf9ed08380ec5ce88875af45c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
