)]}'
{"id":"openstack%2Fkarbor~740571","triplet_id":"openstack%2Fkarbor~master~I632452a559ad25085ddd7384c5eddbb6e422d630","project":"openstack/karbor","branch":"master","hashtags":[],"change_id":"I632452a559ad25085ddd7384c5eddbb6e422d630","subject":"Replace assertItemsEqual with assertCountEqual","status":"ABANDONED","created":"2020-07-12 03:25:17.000000000","updated":"2021-06-09 22:50:05.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a2ea54df4b58b8657a02f7e66b876b07648cc423","_number":740571,"virtual_id_number":740571,"owner":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-07-12 04:01:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-12 04:01:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"10f61585ee76cf250059fd60a9b8312ea130e452","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2020-07-12 03:25:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"68cf0d285a214b8d5a5627b696591d2894053df8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-12 04:01:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a90d7a8e271e4d2ca7fbfc490e20643c : FAILURE in 3m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a5d3403fa174cf4be0cf4dff9eadc2a : FAILURE in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8c42dceced24965b94fee1e0b60ce68 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd9da158352b4401a740f6d2511bf7f0 : FAILURE in 4m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d1f9ad2f5c9f44c995abe922891e04a7 : FAILURE in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3894d7223dc4a989e34b97e27d8eb0e : FAILURE in 4m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f9d38ac7a41425aa3349fc5733163bc : SUCCESS in 5m 00s\n- karbor-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/9cb4ffa28509472ab74747c0413a2329 : FAILURE in 29m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a2ea54df4b58b8657a02f7e66b876b07648cc423","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-06-09 22:50:05.000000000","message":"Abandoned\n\nkarbor is retired now http://lists.openstack.org/pipermail/openstack-discuss/2020-November/018643.html","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"68b151a8a726c39083da005e20e9b07439980deb","revisions":{"68b151a8a726c39083da005e20e9b07439980deb":{"kind":"REWORK","_number":1,"created":"2020-07-12 03:25:17.000000000","uploader":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"ref":"refs/changes/71/740571/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor","ref":"refs/changes/71/740571/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor refs/changes/71/740571/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor refs/changes/71/740571/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor refs/changes/71/740571/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor refs/changes/71/740571/1"}}},"commit":{"parents":[{"commit":"0764f5dab12034efdc41d35ccd627f2a6618c459","subject":"Update hacking for Python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/0764f5dab12034efdc41d35ccd627f2a6618c459"}]}],"author":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-07-12 03:24:54.000000000","tz":480},"committer":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-07-12 03:24:54.000000000","tz":480},"subject":"Replace assertItemsEqual with assertCountEqual","message":"Replace assertItemsEqual with assertCountEqual\n\nassertItemsEqual was removed from Python\u0027s unittest.TestCase in\nPython 3.3 [1][2]. We have been able to use them since then, because\ntesttools required unittest2, which still included it. With testtools\nremoving Python 2.7 support [3][4], we will lose support for\nassertItemsEqual, so we should switch to use assertCountEqual.\n\n[1] - https://bugs.python.org/issue17866\n[2] - https://hg.python.org/cpython/rev/d9921cb6e3cd\n[3] - testing-cabal/testtools#286\n[4] - testing-cabal/testtools#277\n\nChange-Id: I632452a559ad25085ddd7384c5eddbb6e422d630\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/68b151a8a726c39083da005e20e9b07439980deb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/68b151a8a726c39083da005e20e9b07439980deb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
